)]}'
{"id":"openstack%2Fos-brick~937697","triplet_id":"openstack%2Fos-brick~master~I9870590fdc62c48a3de4baae38dadf8ba1090769","project":"openstack/os-brick","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-08-20 01:14:32.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9870590fdc62c48a3de4baae38dadf8ba1090769","subject":"Use oslo.utils to strip scope from an IPv6 address","status":"MERGED","created":"2024-12-13 14:57:21.000000000","updated":"2025-08-20 01:15:24.000000000","submitted":"2025-08-20 01:14:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"937697","meta_rev_id":"c470e8bf623e1775937035fda34a813312d3217b","_number":937697,"virtual_id_number":937697,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-20 01:14:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"all":[{"value":2,"date":"2025-08-19 14:45:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":2,"date":"2025-08-19 14:45:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"all":[{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":1,"date":"2025-08-19 14:45:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-13 16:47:31.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2024-12-13 18:06:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-12-13 18:21:05.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2024-12-13 19:37:55.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-12-13 22:19:46.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-12-18 12:09:09.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"},{"updated":"2024-12-19 08:00:09.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2024-12-26 15:28:14.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-08-19 14:45:41.000000000","updated_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"},{"updated":"2025-08-19 14:45:52.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2025-08-19 18:12:03.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"CC"}],"messages":[{"id":"178ce8bffb489d5e969252543aa85b42ad130c9b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-12-13 14:57:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7cdcdd8bd42ff5bd0d6f458ee0dab590fa9a2c2a","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-12-13 16:47:31.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/3a957a18151e4fe2942c55f0650adbd9\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/81d00d84019c4439857b8e3b654f9ff2 : FAILURE in 1h 40m 15s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/dc3bb5c99c3744ffa6cea7e727aec850 : FAILURE in 1h 49m 29s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/6d877017036a4b44b468a7c2dd8e57a5 : FAILURE in 1h 47m 40s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/1e70c40fae6c4713a632b566364b5b2a : FAILURE in 1h 49m 41s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/428220ee434b4b6bb3d346a7a27f0b83 : RETRY_LIMIT in 25s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/304a2d85cbeb49e685523e88341ff54c : RETRY_LIMIT in 25s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/76b27cc9955c4de09d2734ee6effa7e5 : RETRY_LIMIT in 25s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a168085a3d4846c9985561f5883a383b : RETRY_LIMIT in 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"af326373733e32de3c42d0f1cdc46fb79b9fbcf2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-13 18:06:47.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8c650211ec184af8b250be5ae694d9de\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/44e4ddc303354e80994431bc6cedd1e4 : SUCCESS in 2m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/89d2584accf9431f871cebb9325b84a2 : FAILURE in 1h 46m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2efeae03ded4abaad46622303d55496 : FAILURE in 5m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b9445de6d0794746af6527c5e2270859 : FAILURE in 3m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fdf085d549724e0498360b7cdc28d455 : FAILURE in 6m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50ef58593b154c2792f88ec144a5e77c : SUCCESS in 7m 06s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/8e5badb4dc704dfea5ea30da184d9b2d : FAILURE in 4m 18s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/6e227fbd39de4e2db0684b588a4da638 : TIMED_OUT in 3h 03m 46s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d0ea5192bd524d91a6b4f2ae5f9c2120 : FAILURE in 2h 43m 32s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/e7f94dc6ab9942edbe1840f5e5f72f4f : FAILURE in 1h 19m 15s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/6a297725e44c4177a8e721111fb3275a : SUCCESS in 5m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1aa74bd3079faf8017e6e4a33384c184a01fbe04","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-12-13 18:21:05.000000000","message":"Patch Set 1:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/111/.","accounts_in_message":[],"_revision_number":1},{"id":"d366ab9a6839e2be2a2c5d50745ae915b84b1ca3","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-12-13 19:37:55.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/5db12b07c98f4511a272c3de6d0d18e1 : FAILURE in 1h 14m 02s","accounts_in_message":[],"_revision_number":1},{"id":"1d81db5a440ec3624922896aba36bfc3add65e38","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-12-13 22:14:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/115/.","accounts_in_message":[],"_revision_number":1},{"id":"e66376b7206d1df6fdf042249eee94bf1623c23b","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-12-13 22:19:46.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_os-brick https://elab-os-logsrv.delllabs.net/97/937697/1/check/DellEMC_PowerFlex_os-brick/fa86e7c/DellEMC_PowerFlex_os-brick/172 : FAILURE in 2h 53m 00s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"397c3f35b46c299ff9c2129ec8158031a5914642","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-12-18 12:09:09.000000000","message":"Patch Set 1:\n\nBuild failed. To recheck use \u0027run-NEC V Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/97/937697/1/20241218104505/NEC-V-ISCSI : SUCCESS in 4h 06m 13s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/97/937697/1/20241218104505/NEC-V-FC : FAILURE in 10h 24m 02s","accounts_in_message":[],"_revision_number":1},{"id":"c814a03dfb5615d3ef6dbff23b87e78b0fece97d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-12-19 00:26:54.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"0e7ba387a0a9591ba8cc31ee8af9f2bb706eeb25","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-12-19 01:40:15.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/5b1b85a9bd4b4534b7c0806d1eff2898 : SUCCESS in 1h 11m 36s","accounts_in_message":[],"_revision_number":2},{"id":"8c70d6c332248741c5ddc9f25706013cfec441b5","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-12-19 02:23:06.000000000","message":"Patch Set 2:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/adac394adb684b50941d5c9da01cdbb9\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/43788e19140f4858a6fa1b9a4b861fc6 : SUCCESS in 41m 26s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/720ade165bb74d8f911ca057a6cb210f : SUCCESS in 38m 25s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/6fa56dc75f60422e8dc1d4d163b35b48 : FAILURE in 1h 55m 42s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d8b95a0d925545f493b27f613362fb60 : SUCCESS in 42m 31s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/568c6e5e1f524356a9caaa9008c71528 : RETRY_LIMIT in 24s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/873c94ed43e440d0832ef58b3c451815 : RETRY_LIMIT in 24s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/cfe3c8487de445deb7de7361ade8f14c : RETRY_LIMIT in 25s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/cba215e581d04ca3a7e4b98dd7f69698 : RETRY_LIMIT in 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"69f7f396c2a7d92b83bb77561a69e91eac4bab7e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-19 02:31:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/74664adf50744ae9857966a9cc253077\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/eef469875ee7499a977e41c81b63f520 : SUCCESS in 2m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5a8f2f62aa2d46a99317c9441e49b9e7 : SUCCESS in 1h 57m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0c15d03d3014fb7b3c54b4c8e96fde7 : SUCCESS in 3m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7e1e8db354db425792570925d8ea2ea1 : SUCCESS in 3m 51s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/25f70072fb914ceb95da1708ed038b64 : SUCCESS in 3m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02487a4ad9c74630af868c612a8d899a : SUCCESS in 5m 54s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/a3c6422ec79b4c08b1f60ad349bee284 : SUCCESS in 3m 41s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/3dce36a581ce43e9ae2bf8cb284dcae4 : SUCCESS in 1h 09m 08s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/615c1c7602784c76bb0e6dadc5464df6 : SUCCESS in 1h 55m 01s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/0fe42feda8384837b336509e7687a344 : SUCCESS in 1h 35m 49s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/5028f811984d49e4aff07253d8902dd4 : SUCCESS in 3m 15s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0c5be5ad6bbeb2bba71c493155a3088c21118333","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-12-19 08:00:09.000000000","message":"Patch Set 2:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/97/937697/2/fc : SUCCESS in 1h 01m 14s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/97/937697/2/iscsi : SUCCESS in 1h 12m 29s","accounts_in_message":[],"_revision_number":2},{"id":"37aebca01f8287fdbf1426b560f115d4e3d385ff","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-12-19 15:33:09.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_os-brick https://elab-os-logsrv.delllabs.net/97/937697/2/check/DellEMC_PowerFlex_os-brick/fb707aa/DellEMC_PowerFlex_os-brick/175 : SUCCESS in 2h 11m 08s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"8772b259c644f7307d597e83cdb738e03380c1e7","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-12-24 02:27:33.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/97/937697/2/20241224073026/NEC-V-ISCSI : SUCCESS in 3h 52m 08s\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/97/937697/2/20241224073026/NEC-V-FC : SUCCESS in 3h 57m 05s","accounts_in_message":[],"_revision_number":2},{"id":"04b53c721306e4e9134971cafe81b79d5abbec8c","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2024-12-26 15:28:14.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/97/937697/2/20241226151020/NEC-ISCSI : SUCCESS in 5h 23m 33s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/97/937697/2/20241226151020/NEC-FC : SUCCESS in 9h 17m 50s","accounts_in_message":[],"_revision_number":2},{"id":"f7d75145f77c962ada8c90f778acd135e13ef1a6","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2025-08-19 14:45:41.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"de82af292dfcca28fd9f8f31b6cf4523f30c06c5","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2025-08-19 14:45:52.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"acc93b3a1429948308886e850d87062ee1bcb0b3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 14:46:28.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6853c7fe1cbff1697d38c218fb7202fd48191d60","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 18:03:44.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1376e6af105b4cbb920189d4ac9b2292\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/cc8543aca39c4dd1a646f6dbfb7bb995 : SUCCESS in 3m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bd75843725204eecb421fadf308c4e7c : FAILURE in 50m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0948a46a93a0490292b2732571e88750 : SUCCESS in 5m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/241d2a4699994a22bb40e7ae7bd3fc3d : SUCCESS in 4m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fd1aafe317bf47ffaf921d111f7afad5 : SUCCESS in 4m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb8607e3d6da4d1c91c1f82bbc3d9652 : SUCCESS in 3m 56s\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d8c8c15d351c45c0bdd8864c2b6503b0 : SUCCESS in 2h 12m 21s","accounts_in_message":[],"_revision_number":2},{"id":"501b8871626d0f4315ab31de0380d13429d2e94d","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-08-19 18:12:03.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"284cf5b5d5c30b30b864ea9df663af4aa0289bec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 21:02:41.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/291ca989a7b7458aa3cf89f700c33056\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/c9d0ba63a3bf48dca053133f928b0323 : SUCCESS in 3m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0657dfd487d84cb4a25583a1fd23f28a : FAILURE in 51m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/097171929bd2474f80b83f2092cdc408 : SUCCESS in 2m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3261ef0be5da4b18995d9ac397f43b36 : SUCCESS in 3m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ad76be0ef60245ad8f9456fed38bb130 : SUCCESS in 3m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/791e8e73cca2418fb774a7673c68b49e : SUCCESS in 6m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1030cb931a5344499aba89ed89cebab4 : SUCCESS in 3m 55s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/9d9888e1085a4743875e4fa489bc6757 : SUCCESS in 5m 16s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/7ea9388977224bc8b3fc94660e2162ad : SUCCESS in 2h 13m 02s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/616e6ba2312d44f6b5ffee841608453f : SUCCESS in 1h 36m 18s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/48940d0c093d4766aaae56b175da9740 : SUCCESS in 56m 35s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/f4d6be2cb42c435db87204d095e75408 : SUCCESS in 2m 50s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"53b78d6a2b94eb8f88df6bdbd7d0aa57b35f9d27","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-08-19 21:14:54.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"013235b4647ec853a7f27e63bcf7c451f808f08d","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-08-19 21:30:47.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/e72a5adb8ed9430497c6a0dec3661566 : SUCCESS in 43m 31s","accounts_in_message":[],"_revision_number":2},{"id":"b758ba40d3bfeadb9b800cc5bbf21a2e46f2cadb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 23:20:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa1d7914d6414a2ab6c79ea05b4462dc\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/a1b45bab76c04fa18e26e53c7b74e7f0 : SUCCESS in 2m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d19303418a184052a5fc39fd320e43c7 : SUCCESS in 1h 34m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/809a9b2e298c4eb2b1df8ab593202ec5 : SUCCESS in 2m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/87fb52faa1794a3fa46145b8e814e1e6 : SUCCESS in 5m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c793031bd84840c3981ce3ce05fb272d : SUCCESS in 2m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2fbfc71063bb498392d4bb215848abf1 : SUCCESS in 3m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18ce9664b93a49d0a7dffd5e33c0f218 : SUCCESS in 5m 55s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/d7337661614c41ff8b44a2de1b02a8ae : SUCCESS in 4m 24s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/149d6fee3c0f4a8eb4daf5795b3385d9 : SUCCESS in 1h 09m 30s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/2d536fd4a1b44442ad1dbd02cf0d8ad4 : SUCCESS in 1h 55m 10s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/118c3662bd3248b7b16ce8663e757960 : POST_FAILURE in 1h 30m 32s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/1247de9b249e4109be96aa24d237ca77 : SUCCESS in 5m 11s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"16d32f8bd63b792c1359a3792d816bac4a3dfcdc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 23:21:16.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4401be6220fd82ce09e590bf21dc9f2ea712707a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-20 01:14:32.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/911617ba83b0499f98b131bf53220cf4\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/ae5898ae7f44416d9ca8253425199a79 : SUCCESS in 2m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/43eb2161c2324232ba7e84a0b2c4eefc : SUCCESS in 58m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ac4a11940f44546a31d9127223b3d46 : SUCCESS in 2m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a3c82052f61f4e8f8be74d8423a0a708 : SUCCESS in 5m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3d8fe458fc4b4dc984743797d1aa2e8c : SUCCESS in 5m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a2f4d2c5d824f849fef5ca6a83f7da1 : SUCCESS in 8m 45s\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/73436afe27af4eb39788eb85d36fb4a4 : SUCCESS in 1h 51m 45s","accounts_in_message":[],"_revision_number":2},{"id":"ebcbee8e754463960f1244403e77b1e0934bfce6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-20 01:14:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"c470e8bf623e1775937035fda34a813312d3217b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-20 01:15:24.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/19f8fe06c9ef4a0cb4d78f1b0f3ec076\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1e00b7410784c4daa61be0e9288325c : SUCCESS in 43s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5832f18d25b44de0663a2de7a83d5fc1f7c76158","revisions":{"5f89dd4b5abae5ba9d3f2dac1d085658eba811eb":{"kind":"REWORK","_number":1,"created":"2024-12-13 14:57:21.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/97/937697/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/97/937697/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/97/937697/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/97/937697/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/97/937697/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/97/937697/1"}}},"commit":{"parents":[{"commit":"73687e1290eb8eb036250ad4605df5857363c2d4","subject":"Merge \"Bump hacking to 7\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/73687e1290eb8eb036250ad4605df5857363c2d4"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-12-13 14:55:22.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-12-13 14:56:57.000000000","tz":540},"subject":"Use oslo.utils to strip scope from an IPv6 address","message":"Use oslo.utils to strip scope from an IPv6 address\n\noslo.utils provides the utility method to remove scope from an IPv6\naddress since 7.3.0[1].\n\n[1] 076aaac87f6478a95bfb156bd674dfd0558f6782\n\nChange-Id: I9870590fdc62c48a3de4baae38dadf8ba1090769\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/5f89dd4b5abae5ba9d3f2dac1d085658eba811eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/5f89dd4b5abae5ba9d3f2dac1d085658eba811eb"}]},"branch":"refs/heads/master"},"5832f18d25b44de0663a2de7a83d5fc1f7c76158":{"kind":"REWORK","_number":2,"created":"2024-12-19 00:26:54.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/97/937697/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/97/937697/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/97/937697/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/97/937697/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/97/937697/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/97/937697/2"}}},"commit":{"parents":[{"commit":"73687e1290eb8eb036250ad4605df5857363c2d4","subject":"Merge \"Bump hacking to 7\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/73687e1290eb8eb036250ad4605df5857363c2d4"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-12-13 14:55:22.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-12-19 00:26:54.000000000","tz":0},"subject":"Use oslo.utils to strip scope from an IPv6 address","message":"Use oslo.utils to strip scope from an IPv6 address\n\noslo.utils provides the utility method to remove scope from an IPv6\naddress since 7.3.0[1].\n\n[1] 076aaac87f6478a95bfb156bd674dfd0558f6782\n\nChange-Id: I9870590fdc62c48a3de4baae38dadf8ba1090769\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/5832f18d25b44de0663a2de7a83d5fc1f7c76158"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/5832f18d25b44de0663a2de7a83d5fc1f7c76158"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
