)]}'
{"id":"openstack%2Fos-brick~938095","triplet_id":"openstack%2Fos-brick~master~Iaa8c3ebfbb408048625b5d2e639c5948c725ca55","project":"openstack/os-brick","branch":"master","attention_set":{},"removed_from_attention_set":{"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2024-12-26 09:05:17.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Iaa8c3ebfbb408048625b5d2e639c5948c725ca55","subject":"Wrap librbd calls in native threads","status":"ABANDONED","created":"2024-12-19 23:33:37.000000000","updated":"2024-12-26 09:05:17.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"aa1c96b2dfebb644879adbfceb5f8405383c8716","_number":938095,"virtual_id_number":938095,"owner":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-12-24 14:31:56.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-20 00:45:35.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-12-20 01:16:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-12-20 01:22:02.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-12-20 01:27:26.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2024-12-20 10:11:57.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2024-12-24 15:00:07.000000000","updated_by":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"reviewer":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"state":"CC"},{"updated":"2024-12-24 22:41:04.000000000","updated_by":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"reviewer":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"state":"CC"}],"messages":[{"id":"f26ca39ca3e57f00081ec5110f7c83a86a26bb76","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-12-19 23:33:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bd70d1a9e0b1069684cb012de3b9bb18e4f1114c","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-12-20 00:45:35.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/671c2cb843554951980b6ac79e4439ac : SUCCESS in 1h 10m 00s","accounts_in_message":[],"_revision_number":1},{"id":"cd2eacefe9fd7b7ed1d3d189529da82fcbc0500e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-20 01:16:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cb933030a9114dd497103bd6705b0f24\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/5eabb14b5d074d38a80a9a3e499beec7 : SUCCESS in 2m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7dd6efb41def442191bb5d64b8af6449 : SUCCESS in 1h 36m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45df358ccd21464a969642c189e63650 : SUCCESS in 3m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/463f7d11f4df46d88ff57b36f60bdce9 : SUCCESS in 3m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7b3e3df0c4394450924fd53b0303de3d : SUCCESS in 3m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4cfde3d91876490fa752667fcb30e720 : SUCCESS in 4m 41s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/6987ea7b9a854b93bc3c5f4b9615ba15 : SUCCESS in 4m 21s (non-voting)\n- os-brick-src-devstack-plugin-ceph-v https://zuul.opendev.org/t/openstack/build/8fe2be1138394bddaad75ec629a7725e : FAILURE in 1h 39m 39s\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/209e4ae0f01d43a986adfb95cee8c3b4 : FAILURE in 1h 27m 40s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/64265ac53f654d63bbdbf1ddee15cb56 : SUCCESS in 1h 29m 37s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/18eb44dbc84d4b5492ed7f2df4f0660b : SUCCESS in 55m 38s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/8de5dacfb0054e2d98ed973437c3a655 : SUCCESS in 3m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1fa60dc2f4319d60341a874b3240b62bed441cdf","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-12-20 01:22:02.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_os-brick https://elab-os-logsrv.delllabs.net/95/938095/1/check/DellEMC_PowerFlex_os-brick/8a3a071/DellEMC_PowerFlex_os-brick/176 : SUCCESS in 1h 48m 09s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"05e05fd6bf46a3d582d966f8b954384740e02fa2","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-12-20 01:27:26.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/4762ec89c9314c80b56f68588e6a83d3\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/4e64d6fc748847058d16164f076c9f34 : SUCCESS in 42m 11s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/4d0480b8e2954927a87814dcb82acb2b : FAILURE in 46m 03s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/16f5584d131b46acbee838470b146f68 : FAILURE in 1h 53m 19s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f85d2c49284c420d8b71d4fee60d1e0f : FAILURE in 40m 14s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e65c40b4db86428e9e488a33d393e8e7 : RETRY_LIMIT in 24s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c4cd171b00e94e66a05cee4f9bedb37c : RETRY_LIMIT in 24s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/2e501a0c19a8417cbb3bb91b6a69361f : RETRY_LIMIT in 25s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d305438b05ff42878fb4294e62e1793c : RETRY_LIMIT in 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8df6cc55a87d6a7796e103b2e6314db48179e4c2","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-12-20 10:11:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/157/.","accounts_in_message":[],"_revision_number":1},{"id":"4089e1f07d32002b1f9b9615ebc0bc87958115a2","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-12-20 13:36:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/146/.","accounts_in_message":[],"_revision_number":1},{"id":"e0fd476402fd7950075229f0e178e2c4bebd4ae1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-12-24 12:37:30.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"17ed9e6dc7fd4c5c628f8ca2becfd4c9e7803d75","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-12-24 13:36:16.000000000","message":"Patch Set 2:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/166/.","accounts_in_message":[],"_revision_number":2},{"id":"2259ec51a608fffce5b0633f4e08ac39dec7ddc9","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-12-24 13:53:35.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/8d18307fc7fe4362a5e1e20459cd5fd5 : SUCCESS in 1h 14m 13s","accounts_in_message":[],"_revision_number":2},{"id":"5f083a47a69323fd0fd3ad9a29020ea9ea1474c8","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2024-12-24 13:58:03.000000000","message":"Patch Set 2:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/155/.","accounts_in_message":[],"_revision_number":2},{"id":"6026c8c5effe78efcf393b3e142169b35383df14","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-12-24 14:10:45.000000000","message":"Patch Set 2:\n\nBuild succeeded.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/55f7395881724ec2aeaa1eebf8f924fe\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/2ae49c503c2f480c9160255dcaa83c02 : SUCCESS in 44m 53s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/5f26cb659ffe40f0b79e22681cf539cf : SUCCESS in 35m 59s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/48720eebdd154aa0a5cd2a2e6f064fc5 : FAILURE in 1h 32m 45s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/e0ff5c90315e4043961d041947f7a019 : SUCCESS in 45m 59s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/477a0c500f6a449dbbec6ff13fc871c0 : RETRY_LIMIT in 25s (non-voting)\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/557b4663be1f4155a0860c33ed2921f1 : RETRY_LIMIT in 24s (non-voting)\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/2295ea384db94f9592798f7ee62bb0d0 : RETRY_LIMIT in 24s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/5b025963481149f891b91cecb653f001 : RETRY_LIMIT in 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"07936d391e22ceeef018e7221302a8f5e00ec21a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-24 14:31:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e6dc3af2bf3f4883a5e21dc8a8d1ad8c\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/a9b6cf7dbcf344be8f3574df603e5b88 : SUCCESS in 2m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5d75d74adb394fb88e7ad39aec65e0c8 : SUCCESS in 1h 49m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6488fe81e8894b29945681c626f569fb : SUCCESS in 4m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4a61f89f723e4d158c97aaf35a160dfb : SUCCESS in 4m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f2b7ad390cdc485c8436699e67eb04a6 : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05f3fdc9aec74122b53e08910c0eeeb4 : SUCCESS in 6m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/468ebb04e85740ceabe511938d33aee2 : SUCCESS in 4m 41s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/57a97435bef547649488b03815c6995d : SUCCESS in 4m 21s (non-voting)\n- os-brick-src-devstack-plugin-ceph-v https://zuul.opendev.org/t/openstack/build/8c47f950134e4bcb88ce1c6f26cdca25 : SUCCESS in 1h 50m 46s\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/7256063348e74105a63d14cd6096e904 : SUCCESS in 1h 33m 26s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/d4aae07a3ee64d5b9030254cdc8bcc5a : SUCCESS in 1h 40m 23s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/27ee081325a94f1a9220aeaea7304379 : SUCCESS in 1h 10m 26s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/411735286db841a4a1336a6360b62a69 : SUCCESS in 4m 54s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8e7e103ef5a828112dbc501a9eae15ba9445b81f","author":{"_account_id":31980,"name":"Hitachi VSP CI","email":"knomuracinder@gmail.com","username":"hitachi-ci"},"date":"2024-12-24 15:00:07.000000000","message":"Patch Set 2:\n\n- hitachi_vsp_fc http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/95/938095/2/fc : SUCCESS in 0h 47m 27s\n- hitachi_vsp_iscsi http://ec2-18-177-230-241.ap-northeast-1.compute.amazonaws.com/refs/changes/95/938095/2/iscsi : SUCCESS in 0h 54m 24s","accounts_in_message":[],"_revision_number":2},{"id":"f3c07dfbc3661b2eb02d8906ddf8213d5cb97611","tag":"autogenerated:NEC_V_Cinder_CI","author":{"_account_id":34155,"name":"NEC V Cinder CI","email":"nec-v-cinder-istorage@istorage.jp.nec.com","username":"nec--v-cinder-ci"},"date":"2024-12-24 22:41:04.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- NEC-V-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/95/938095/2/20241225034037/NEC-V-FC : SUCCESS in 3h 59m 33s\n- NEC-V-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/95/938095/2/20241225034037/NEC-V-ISCSI : SUCCESS in 4h 00m 24s","accounts_in_message":[],"_revision_number":2},{"id":"aa1c96b2dfebb644879adbfceb5f8405383c8716","tag":"autogenerated:gerrit:abandon","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-12-26 09:05:17.000000000","message":"Abandoned\n\nFollowing reasons makes this patch not suitable:\n1. running rados.Rados in a native thread causes the main thread to hang and doesn\u0027t proceed\n2. the RBDVolumeIOWrapper contains \"property\" methods which causes unexpected behavior when wrapped with Proxy methods\n\nWITHOUT PATCH\n\n(Pdb) volume_file\n\u003cos_brick.initiator.linuxrbd.RBDVolumeIOWrapper object at 0x76e3d67bfd00\u003e\n\n\n(Pdb) volume_file.rbd_image\n\u003cos_brick.initiator.linuxrbd.RBDVolume object at 0x76e3d68786d0\u003e\n(Pdb) volume_file._rbd_volume.image\n\u003cos_brick.initiator.linuxrbd.RBDVolume object at 0x76e3d68786d0\u003e\n\n\nWITH PATCH\n\n(Pdb) volume_file\n\u003cos_brick.initiator.linuxrbd.RBDVolumeIOWrapper object at 0x7a5d3bb59510\u003e\n\n\u003cos_brick.initiator.linuxrbd.RBDImageMetadata object at 0x7a5d3bb582e0\u003e\n(Pdb) volume_file._rbd_volume.image\n\u003cos_brick.initiator.linuxrbd.RBDVolume object at 0x7a5d3bc55450\u003e\n(Pdb) volume_file.rbd_image\n\u003cfunction Proxy.__getattr__.\u003clocals\u003e.doit at 0x7a5d3bb7fa30\u003e","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f4b90415b39487d155444dd5bab13e48476b6dd7","revisions":{"8ae0ab643181804cae865574e36190baaeebb07b":{"kind":"REWORK","_number":1,"created":"2024-12-19 23:33:37.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/95/938095/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/95/938095/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/95/938095/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/95/938095/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/95/938095/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/95/938095/1"}}},"commit":{"parents":[{"commit":"1487dc93b60c33a1170632d484dafccc2bea4854","subject":"Merge \"Fix multipathd dependency in NVMe-oF connections\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/1487dc93b60c33a1170632d484dafccc2bea4854"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-12-19 23:25:28.000000000","tz":330},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-12-19 23:25:28.000000000","tz":330},"subject":"Wrap librbd calls in native threads","message":"Wrap librbd calls in native threads\n\nCurrently there are 2 places where we don\u0027t run librbd calls in\nnative threads:\n\n1. creating a connection to Rados\n2. Initializing the RBD volume object from an existing RBD image\n\nThese calls when done with green thread might hang the process hence\nshould be wrapped in native threads.\n\nAlso the eventlet dependency is added again (removed in [1]) and\nthe version is kept same as Cinder requirements[2] even though we\nhave newer versions available[3].\n\n[1] https://review.opendev.org/c/openstack/os-brick/+/699987\n[2] https://github.com/openstack/cinder/blob/master/requirements.txt#L7\n[3] https://pypi.org/project/eventlet/#history\n\nChange-Id: Iaa8c3ebfbb408048625b5d2e639c5948c725ca55\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/8ae0ab643181804cae865574e36190baaeebb07b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/8ae0ab643181804cae865574e36190baaeebb07b"}]},"branch":"refs/heads/master"},"f4b90415b39487d155444dd5bab13e48476b6dd7":{"kind":"REWORK","_number":2,"created":"2024-12-24 12:37:30.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/95/938095/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/95/938095/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/95/938095/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/95/938095/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/95/938095/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/95/938095/2"}}},"commit":{"parents":[{"commit":"1487dc93b60c33a1170632d484dafccc2bea4854","subject":"Merge \"Fix multipathd dependency in NVMe-oF connections\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/1487dc93b60c33a1170632d484dafccc2bea4854"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-12-19 23:25:28.000000000","tz":330},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-12-24 12:37:27.000000000","tz":330},"subject":"Wrap librbd calls in native threads","message":"Wrap librbd calls in native threads\n\nCurrently when creating a connection to Rados, we don\u0027t wrap the\nlibrbd call in a native thread.\n\nThese type of calls when done with green thread might hang the process\nhence should be wrapped in native threads.\n\nAlso the eventlet dependency is added again (removed in [1]) and\nthe version is kept same as Cinder requirements[2] even though we\nhave newer versions available[3].\n\n[1] https://review.opendev.org/c/openstack/os-brick/+/699987\n[2] https://github.com/openstack/cinder/blob/master/requirements.txt#L7\n[3] https://pypi.org/project/eventlet/#history\n\nChange-Id: Iaa8c3ebfbb408048625b5d2e639c5948c725ca55\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/f4b90415b39487d155444dd5bab13e48476b6dd7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/f4b90415b39487d155444dd5bab13e48476b6dd7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
