)]}'
{"id":"openstack%2Fos-brick~950995","triplet_id":"openstack%2Fos-brick~stable%2F2025.1~Idd0577cf31f186a8cd12dd3d4054b0a1a91ec3d4","project":"openstack/os-brick","branch":"stable/2025.1","attention_set":{},"removed_from_attention_set":{"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2025-10-21 17:06:37.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Idd0577cf31f186a8cd12dd3d4054b0a1a91ec3d4","subject":"FC: Avoid calling ``lsscsi`` during disconnect","status":"MERGED","created":"2025-05-27 13:34:48.000000000","updated":"2025-10-21 17:07:24.000000000","submitted":"2025-10-21 17:06:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"950995","cherry_pick_of_change":943123,"cherry_pick_of_patch_set":4,"meta_rev_id":"d2d345061d20cde5ab2cbdc10b3b9a38757b815f","_number":950995,"virtual_id_number":950995,"owner":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-21 17:06:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2025-10-21 15:14:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2025-07-11 20:45:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2025-10-21 15:14:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}],"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-27 13:49:44.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2025-05-27 14:11:22.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2025-05-27 15:38:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-04 16:01:03.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"CC"},{"updated":"2025-07-11 20:45:04.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2025-07-12 00:53:19.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2025-10-21 15:14:24.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"}],"messages":[{"id":"2cae4100963cdc1dec61e5905edba4b8e1895ead","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2025-05-27 13:34:48.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"7593d4f0de938dd726049bff2dbb14a9f04cd5b5","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-05-27 13:49:44.000000000","message":"Patch Set 1:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/920/.","accounts_in_message":[],"_revision_number":1},{"id":"1757241d9e8d7e44e0c6e2cd559e974fa7f907ac","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-05-27 13:49:59.000000000","message":"Patch Set 1:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/794/.","accounts_in_message":[],"_revision_number":1},{"id":"dd0ba9a2988c81598989c082ddd107ee95c8244c","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-05-27 14:11:22.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/2220e0a1eff6443bb91898c1ca5c435b\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/3ac959b74b6e4279acf4cc0b3488d435 : FAILURE in 24m 43s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c7f266decd7140eeaffa870da84508a0 : FAILURE in 28m 01s\n- pure-devstack-cinder-tempest-nvme-roce-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/660f8d252f09426aa490c35af5df0c3b : FAILURE in 23m 40s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/579ba032e88048ea8eb9377050eab41c : FAILURE in 23m 10s\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/cb8e3f34e64a4484b16553bc994b7f22 : FAILURE in 34m 36s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/394c6d1c11a14c2f92d68bb298131f56 : FAILURE in 35m 42s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c2eeb2467ddb477fac00d9179cc9249c : FAILURE in 32m 43s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/b57d3d2ea2624a538ab5b58e572d899a : FAILURE in 30m 49s","accounts_in_message":[],"_revision_number":1},{"id":"7a6227a0d5b7ffadda6b39ef6f5af28164e2dcdf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-27 15:38:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e5e40e889f234ba3b47ce6439d514d2e\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fb57da7a31c045578e607a3ae917d573 : SUCCESS in 1h 37m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a7b450d64e743c78f9f055d5e8d87e1 : SUCCESS in 3m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/03eff6fba0df478ca7ea250f34fcbebf : SUCCESS in 3m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5099c84c896f48508b63abf5ac7b9ea6 : SUCCESS in 3m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c307510963e64488ab1bfb2fc6d91c3f : SUCCESS in 6m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/60893aa91d8d43109ac7e3872ecda8cf : SUCCESS in 4m 33s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/ce7bd1590a89486ba040d9835447fe3f : SUCCESS in 4m 24s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/bf4c5b68937348c0af7d6e1551a4e820 : SUCCESS in 1h 47m 23s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b192befe1b4943848bc1575e42b4f294 : FAILURE in 1h 30m 27s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/347dead4d2bb4ea9aee4642a70fcdb7c : SUCCESS in 1h 19m 07s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/b80f32622fba4f27b0f1a77764e38f2c : SUCCESS in 4m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b1fd687aacd4129a1898cb95a6d1bfc2845ac3a4","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2025-05-28 06:00:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9c8e06ce2f7834c818a3c8640cb36575ddd1eb21","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-28 07:59:48.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cd32a2fe4bb740fe8b0b5cc10a3f622d\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cea84c20e0114b029e9000b838e1bb64 : SUCCESS in 1h 17m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b68539b6d0f446d8b4ba86ceb79665bb : SUCCESS in 3m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b065d9af30d64be981d118e9df36ab8e : SUCCESS in 4m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7f641ffc52b74a329bfd6f70cfc1eeac : SUCCESS in 3m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ac5616177814170b403ed8924892329 : SUCCESS in 6m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8c5ce1a2898448da9625fc1247838dad : SUCCESS in 3m 21s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/5d9e75680517475f882e0f9c90baefb0 : SUCCESS in 4m 56s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/f52164116dcc4e55afcb70014f87ce6a : SUCCESS in 1h 26m 37s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/7125fbad08c3454992aca736dd8c3b6e : FAILURE in 1h 26m 32s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/ff57b9b60b364998a602285c6bfc71c4 : POST_FAILURE in 1h 57m 30s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/5aca0f65ded4434283e2616ca38a8ac1 : SUCCESS in 3m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"09d0bc1b84891bf18af089f7c115da5e2084d14f","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2025-06-03 05:26:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a0cb65b1ff225d11b13dc87c78cbbda371371e6a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-03 08:10:31.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6b4f40a74f624f0fab194db5bdf1dafc\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e5759b2ec8e44e3fbb8fbfe8127ae95a : SUCCESS in 1h 24m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23cef29349264079976d7beb7abca1e9 : SUCCESS in 3m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0a1e6ab272dc48ca8647f0f342a761ed : SUCCESS in 8m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7fc8f89f12c541f1b1e521e4907fe9ec : SUCCESS in 3m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6ca9195aaac4435bfaf37041a13a8af : SUCCESS in 6m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/15649b10646c48f5a378009c14c933be : SUCCESS in 4m 56s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/c5e2a93055bc4c84b9e202f49c9c3981 : SUCCESS in 3m 39s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/ef80c0aae7cf48d598ceadf2296c947e : SUCCESS in 1h 58m 49s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a9cade08264e4af18017535ce61ddee7 : FAILURE in 1h 54m 05s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/d14759b822844340916693be2ffa1c03 : POST_FAILURE in 2h 10m 58s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/14d6ff8606674a759b2604dc4db51a9c : SUCCESS in 3m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"346b4b0ec8fb5267331bd572d499570928b10142","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-06-04 16:01:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c38f7427c18292d4854630edfa9e75b8943f2110","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-07-11 17:47:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"81f3ac0f8be1e260724e3cca1fe2cf24e5437b3f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-11 20:08:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3285b62535b946399ffbbd1792fb63eb\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4ab92d05d3eb490aa78dc5abcf5cc208 : SUCCESS in 1h 39m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0da5520147f24bf78c648915636bc694 : SUCCESS in 3m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/29c8b01a97e44e07bb2e28248c6799b4 : SUCCESS in 4m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4f4fb72489a0494c9ffc422bb200fc30 : SUCCESS in 6m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7fe51e776d74e3a8cd268b8d6e61592 : SUCCESS in 6m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d0c60e45b6644a07918181a0faeb2c2e : SUCCESS in 3m 18s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/ee656da65acb41e791d9df757711e088 : SUCCESS in 5m 30s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/12fd641969094119a74c1144949ed54f : FAILURE in 2h 07m 18s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/119568bcddc04340b55713daa38e57f3 : SUCCESS in 1h 46m 54s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/9f0776e7534647c0bcda2d75583c1701 : POST_FAILURE in 1h 43m 21s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/f6a85b72111243b8969ec800f65e1e50 : SUCCESS in 3m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"609115604fdd5923c9fb85f116290f03c4bac69a","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-07-11 20:45:04.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"65619eabed453a1aaa70b9757663e715cdc1e467","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2025-07-12 00:53:19.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-tempest-plugin-ontap-iscsi cinder-tempest-plugin-ontap-iscsi : NODE_FAILURE in 0s\n- cinder-tempest-plugin-ontap-nfs cinder-tempest-plugin-ontap-nfs : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"be74a9e689198c066f27a4c78e7df9748dfe3cd8","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2025-10-21 15:14:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d0ec4f1f78a88092cfb508eeaa7177223740f142","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-21 15:14:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fa58da8eac16b898dbd6215cb7a12bedb923bebe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-21 17:06:37.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/891043d7eab5400eb6b1984ed7a87c61\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/136a50b211c943f19af0ee682284e9a5 : SUCCESS in 57m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0c9ada6854141ec85468b59a6897e46 : SUCCESS in 4m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0968ad2311d54f33b97b175dd3889a2e : SUCCESS in 3m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/21de26480b3947b88e5a58adc8816a80 : SUCCESS in 2m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9eb6a2479b149f3bcaf3f4584815125 : SUCCESS in 7m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0ea3c12133df4aeca33abc96249195fd : SUCCESS in 6m 59s\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/ee366f6220004abe8877b48dd3cc0ee0 : SUCCESS in 1h 47m 35s","accounts_in_message":[],"_revision_number":1},{"id":"2c20cd12740c200f8a06696c63b489fbeb8074dc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-21 17:06:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d2d345061d20cde5ab2cbdc10b3b9a38757b815f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-21 17:07:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/30169d66b80f4c82afa42129c2e84b29\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e40f280cc2984e26b9a0d55bca673620 : SUCCESS in 38s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/edc28ab072db44799d5927abd87fb9e6 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"69eb4e01ce34d8f32520b5aea681cac930ba7311","revisions":{"69eb4e01ce34d8f32520b5aea681cac930ba7311":{"kind":"REWORK","_number":1,"created":"2025-05-27 13:34:48.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/95/950995/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/95/950995/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/95/950995/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/95/950995/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/95/950995/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/95/950995/1"}}},"commit":{"parents":[{"commit":"6329bdf251e3517d4a73e0afb6d31a1c7e8307e8","subject":"Update TOX_CONSTRAINTS_FILE for stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/6329bdf251e3517d4a73e0afb6d31a1c7e8307e8"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2025-03-02 23:25:41.000000000","tz":330},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2025-05-27 13:34:48.000000000","tz":0},"subject":"FC: Avoid calling ``lsscsi`` during disconnect","message":"FC: Avoid calling ``lsscsi`` during disconnect\n\nWhen we disconnect a fibre channel volume, we get the\n/dev/sdX path for each device and remove them by executing\nthe ``tee -a /sys/block/sdX/device/delete`` command.\n\nCurrently, in the disconnect workflow, we call ``lsscsi``\ncommand to fetch the HCTL values but never use them.\nThis execution of ``lsscsi`` for each device can cause\nperformance issues if we have many devices connected to\nthe system.\n\nFollowing is the workflow:\n1. Get the volume path symlink in /dev/disk/by-path[1]\n/dev/disk/by-path/pci-0000:00:07.0-fc-0x5000d310047d1842-lun-1\n\n2. Translate it to /dev/SdX[2]\n\n3. Call lsscsi to fetch the HCTL values[3]\n\n4. Remove individual devices using the /dev/sdX path[4]\n\nAs we can clearly see that step 3 is unnecessary in this workflow.\n\nAlso the get_name_from_path method translates the /dev/disk/by-path\nsymlink to real path /dev/sdX so we don\u0027t need to call get_device_info\nand it\u0027s functionality is redundant here.\n\nMar 11 20:07:50 ubuntu2204.localdomain cinder-volume[252928]: DEBUG os_brick.initiator.connectors.fibre_channel [None req-9033642e-43ba-4076-a460-63e562b69b3c admin None] Real path /dev/sda for symlink /dev/disk/by-path/pci-0000:00:07.0-f\nc-0x5000d310047d1842-lun-1 {{(pid\u003d252928) disconnect_volume /opt/stack/data/venv/lib/python3.10/site-packages/os_brick/initiator/connectors/fibre_channel.py:385}}\n\n[1] https://opendev.org/openstack/os-brick/src/commit/6e83ac6eeee8f3a4b3265a4e927dca1bc190e088/os_brick/initiator/connectors/fibre_channel.py#L381\n[2] https://opendev.org/openstack/os-brick/src/commit/6e83ac6eeee8f3a4b3265a4e927dca1bc190e088/os_brick/initiator/connectors/fibre_channel.py#L384\n[3] https://opendev.org/openstack/os-brick/src/commit/6e83ac6eeee8f3a4b3265a4e927dca1bc190e088/os_brick/initiator/connectors/fibre_channel.py#L393\n[4] https://opendev.org/openstack/os-brick/src/branch/master/os_brick/initiator/connectors/fibre_channel.py#L438\n\nCloses-Bug: #2102053\n\nChange-Id: Idd0577cf31f186a8cd12dd3d4054b0a1a91ec3d4\n(cherry picked from commit d9392320d7d9494ca71cb808f52d7e8d98c09658)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/69eb4e01ce34d8f32520b5aea681cac930ba7311"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/69eb4e01ce34d8f32520b5aea681cac930ba7311"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
