)]}'
{"id":"openstack%2Fos-brick~954599","triplet_id":"openstack%2Fos-brick~stable%2F2024.1~Idfe59f024232c0ba1f4e41611c975eb5ce0d03bb","project":"openstack/os-brick","branch":"stable/2024.1","topic":"storpool-backports-2024.1","attention_set":{},"removed_from_attention_set":{"35429":{"account":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"last_update":"2025-10-22 15:27:40.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Idfe59f024232c0ba1f4e41611c975eb5ce0d03bb","subject":"StorPool: Add the iSCSI configuration API","status":"ABANDONED","created":"2025-07-10 11:02:39.000000000","updated":"2025-10-22 15:27:40.000000000","total_comment_count":1,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":true,"cherry_pick_of_change":939776,"cherry_pick_of_patch_set":4,"meta_rev_id":"19719ac901d84bea87951aec0183d571f9c7d169","_number":954599,"virtual_id_number":954599,"owner":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-07-10 15:29:21.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":-2,"date":"2025-10-22 15:27:32.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"}]},"pending_reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"}]},"reviewer_updates":[{"updated":"2025-07-10 11:57:55.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2025-07-10 15:29:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-10 16:01:06.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2025-10-22 15:27:32.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"}],"messages":[{"id":"6b01aea05854f6648490ca81fccfaa9c448c7115","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"date":"2025-07-10 11:02:39.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"3f424d2c651aa4780c5e74fb3e3d9d837e4cb4cc","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"date":"2025-07-10 11:05:32.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"b469af87fee993254cce93e725f6ff9ae4eea3dd","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"date":"2025-07-10 11:05:52.000000000","message":"Topic set to storpool-backports-2024.1","accounts_in_message":[],"_revision_number":1},{"id":"26ef01017c49b568313e0800b1e3746a62e52b31","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2025-07-10 11:57:55.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/6fefacfa35e14b6b978189cef7bdfed4\n\n- pure-devstack-cinder-tempest-iscsi-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/777d3737a0b54068b12a1630c4d756fb : RETRY_LIMIT Host unreachable in 8m 53s\n- pure-devstack-cinder-tempest-fc-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/4e1c4a86339a412781cc61c282a115e0 : NODE_FAILURE Node request 200-0000072931 failed in 0s\n- pure-devstack-cinder-tempest-nvme-tcp-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/4f1081a538fd419f8829c9399d1623d7 : NODE_FAILURE Node request 200-0000073471 failed in 0s","accounts_in_message":[],"_revision_number":1},{"id":"bfb40e5e43f93773b92b656ebb1b8a8281640028","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 15:29:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c26910a4b9a44c34bcfeb03ce4ea11f8\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e6871348cc9246a68c758e43764825a7 : SUCCESS in 1h 38m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08d8341062664bc192a909bca7754eb0 : SUCCESS in 3m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9f3ff8ab61c94f189a47713a2ed601de : FAILURE in 2m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/73f2bfe365c141d38e09225c7b4c248e : SUCCESS in 4m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb06826557434926802907e0d681975e : SUCCESS in 6m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fde03b06b4f14425a9d8f00ab8db2986 : SUCCESS in 3m 32s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/8ccb5514753b48ffb121b4782800980b : SUCCESS in 4m 07s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/2f60ef0a67ea455dbcc7a10227f555ac : SUCCESS in 56m 56s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/11314ec0e48a460fa917902ac3e75afd : SUCCESS in 1h 51m 32s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/a1848ad60a8744b59fa9b05b9ceba08e : SUCCESS in 1h 35m 23s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/e9f98318d9f94d4294d61f35c1de2916 : FAILURE in 2m 36s (non-voting)\n\nWarning:\n  Comments left for invalid file os_brick/initiator/connectors/iscsi.py","accounts_in_message":[],"_revision_number":1},{"id":"24a91975673b798d4c3d2a209eb9f0827d87b717","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-07-10 16:01:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1216/.","accounts_in_message":[],"_revision_number":1},{"id":"2aed314e43b04518359f481d1a40d3d93f9d7846","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-07-10 16:27:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1346/.","accounts_in_message":[],"_revision_number":1},{"id":"fefea8e1bacaf3a45c20e0ce75bda59d8ebd4992","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-10-22 15:27:32.000000000","message":"Patch Set 1: Code-Review-2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"19719ac901d84bea87951aec0183d571f9c7d169","tag":"autogenerated:gerrit:abandon","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-10-22 15:27:40.000000000","message":"Abandoned\n\nbranch being deleted","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"15ab15e2ef85e11100d51b2e4a061fd7e3dc988c","revisions":{"15ab15e2ef85e11100d51b2e4a061fd7e3dc988c":{"kind":"REWORK","_number":1,"created":"2025-07-10 11:02:39.000000000","uploader":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"ref":"refs/changes/99/954599/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/99/954599/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/99/954599/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/99/954599/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/99/954599/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/99/954599/1"}}},"commit":{"parents":[{"commit":"eb946fad82dc88aa8c33f85fefd661461cc445f0","subject":"StorPool: Do not use packages `storpool` and `storpool.spopenstack`","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/eb946fad82dc88aa8c33f85fefd661461cc445f0"}]}],"author":{"name":"Biser Milanov","email":"biser.milanov@storpool.com","date":"2025-01-22 09:45:05.000000000","tz":120},"committer":{"name":"Biser Milanov","email":"biser.milanov@storpool.com","date":"2025-07-10 11:02:39.000000000","tz":0},"subject":"StorPool: Add the iSCSI configuration API","message":"StorPool: Add the iSCSI configuration API\n\nChange-Id: Idfe59f024232c0ba1f4e41611c975eb5ce0d03bb\nSigned-off-by: Biser Milanov \u003cbiser.milanov@storpool.com\u003e\n(cherry picked from commit 5f5a8f7563e7ee1f6fc9d8b828a360d5c7c0c1b3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/15ab15e2ef85e11100d51b2e4a061fd7e3dc988c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/15ab15e2ef85e11100d51b2e4a061fd7e3dc988c"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Code-Review\u003dMIN"],"failing_atoms":["label:Code-Review\u003dMAX"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
