)]}'
{"id":"openstack%2Fos-brick~957599","triplet_id":"openstack%2Fos-brick~master~I19065ad683b9ff8174814d6e8c500178c885bee4","project":"openstack/os-brick","branch":"master","topic":"lightbits-pick-first-9e6e34a","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2026-01-14 21:33:22.000000000","reason":"\u003cGERRIT_ACCOUNT_5314\u003e replied on the change","reason_account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-02-04 16:32:11.000000000","reason":"Change was submitted"},"10058":{"account":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"last_update":"2026-02-04 15:03:29.000000000","reason":"\u003cGERRIT_ACCOUNT_10058\u003e replied on the change","reason_account":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"}},"33612":{"account":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"last_update":"2026-02-04 16:32:11.000000000","reason":"Change was submitted"},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2026-02-04 16:32:11.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I19065ad683b9ff8174814d6e8c500178c885bee4","subject":"LightOS: update discovery client configuration","status":"MERGED","created":"2025-08-17 08:32:43.000000000","updated":"2026-02-04 16:33:04.000000000","submitted":"2026-02-04 16:32:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":31,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"957599-lightbits-pick-first-9e6e34a","meta_rev_id":"6e7068f3f23f063e68a5e3c605b0d84a84fffc70","_number":957599,"virtual_id_number":957599,"owner":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-04 16:32:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2026-02-04 14:28:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2026-02-04 14:32:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":1,"date":"2026-02-04 15:03:29.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2026-02-04 14:32:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2026-02-04 14:28:29.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"CC":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-17 09:21:00.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-08-17 09:35:07.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2025-08-17 10:15:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-18 13:42:47.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2025-08-18 17:44:16.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2025-08-20 14:14:43.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2025-09-02 03:29:33.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-01-14 15:18:23.000000000","updated_by":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"reviewer":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"state":"REVIEWER"},{"updated":"2026-01-15 14:46:31.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-01-15 16:08:14.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2026-02-04 14:32:48.000000000","updated_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"}],"messages":[{"id":"667f76d0d442539a44ae27251fcefe228902ae16","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"date":"2025-08-17 08:32:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"53bb76d44e5a01701523eb36f88e7ffc00a73632","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-08-17 09:21:00.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/0a08c3728ee94236a903f37880568e07 : SUCCESS in 46m 04s","accounts_in_message":[],"_revision_number":1},{"id":"69816e1c038a5521500a2cb28c0093f196c1a88e","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-17 09:35:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1454/.","accounts_in_message":[],"_revision_number":1},{"id":"58cf6d01bd6fd0611fc76c9578d9b34669ddfa09","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-17 09:36:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1589/.","accounts_in_message":[],"_revision_number":1},{"id":"163f2d5423a4ad05100c483d922c5604d3dd3176","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-17 10:15:19.000000000","message":"Patch Set 1: Verified-1\n\n(5 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4581059f28ab4d54b1b95d6938077173\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7adcb84ef9c94415b2ca541fd94b0d83 : SUCCESS in 1h 39m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/431da01b74274a959c809518cfa0df90 : FAILURE in 1m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b90d2c351fc24227a50a1412e94b32b9 : SUCCESS in 4m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7110b80adf864cffae218b045616f357 : SUCCESS in 4m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b5e80e0f1ddf4fc1862639f342e8b9b7 : SUCCESS in 7m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/557a16df282c40eabbffdd70f0ffffde : SUCCESS in 3m 27s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/89261a0440e54369aeb17b867c65fe49 : SUCCESS in 6m 06s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/b80fab1b19a149538a85c8b457206e93 : FAILURE in 1h 09m 27s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/ea2f1262c879492dbf758d2f6e8bb8d3 : SUCCESS in 1h 36m 32s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/1674c11b25954b3ca06b2274a5c6f82c : SUCCESS in 49m 50s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/6982072ad075458ba26eca041b9519f8 : SUCCESS in 4m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"640632964f9c8ca02dd77fc80f0fa4a20e387007","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"date":"2025-08-18 08:17:43.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"39354be61025bb459900dae9a71475099a60db25","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-08-18 10:04:08.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/795ef9a250af43269a73cfdb24046e51 : FAILURE in 52m 11s","accounts_in_message":[],"_revision_number":2},{"id":"2fbd4da7ae3c70f6d534eb0a960d617f7499d27f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 10:33:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cbf4d8febf994955889d29035557e045\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/34d076fa71d54faa9729a9c7deb5e56f : SUCCESS in 1h 28m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7c54d8ad21947f0bd654fe458bece36 : SUCCESS in 2m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b5ff5f544905413ca2e6f2a2aad5628b : SUCCESS in 2m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/492c42f446424ca6bc403077371bc1f4 : SUCCESS in 2m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/833b2af973154f458b6edcc668d8d752 : SUCCESS in 4m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d9b7a59ec074aaf85a16b307e41680e : SUCCESS in 4m 47s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/f027be413bf043c7bfdd7f14026af6eb : SUCCESS in 4m 06s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/59e0cbd6c5bc49b598e57b3078e55b0a : SUCCESS in 1h 48m 29s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/f32da9f48c8c4be783c4d568dfb5eeae : SUCCESS in 1h 37m 05s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/0f973b66dc71495ba43f164a02d05d74 : POST_FAILURE in 1h 31m 40s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/5099d9bacd0d46d088fe27617233cfcf : SUCCESS in 2m 30s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"335205863d509502d06358f1000920c248c27dab","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-18 11:00:17.000000000","message":"Patch Set 2:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1458/.","accounts_in_message":[],"_revision_number":2},{"id":"f577ddaf7b26b17d18f6a2a8cdb5737321b12c3e","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-18 11:05:53.000000000","message":"Patch Set 2:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1593/.","accounts_in_message":[],"_revision_number":2},{"id":"58a00d05b33b33bd3398ef1837fe8782afeb1b30","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-18 13:42:47.000000000","message":"Patch Set 2:\n\n(5 comments)","accounts_in_message":[],"_revision_number":2},{"id":"93d804306bdf7a0a12e6c5cf46398d2fabde4ddb","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-18 13:44:28.000000000","message":"Patch Set 2:\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"40f7a6d972b04a69d2dfc21f24d3885922013199","author":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"date":"2025-08-18 14:21:01.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"06dcca37a5cb7e8bec84d2d6296f45189a99bc9f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"date":"2025-08-18 14:43:12.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"907ae39302ea596b9f7dae7c7ade5de1aa18c19d","author":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"date":"2025-08-18 14:45:45.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"054b6bc21306a299f0f0d0e540944c44d5eb4ba4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"date":"2025-08-18 14:58:57.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"6d554d3987102fe7d29ba4d510ea061acfc0b0ae","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-18 15:01:45.000000000","message":"Patch Set 4:\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"0944f4f094c05a9a02ac72726e237d85ddf80b25","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-18 15:05:13.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"5954ed20b8309e060934a8fce2c9b9ab474597e0","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-18 15:05:56.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"cc6f1b3a3fadecfcaf1ca094ffc72e4db704c928","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-18 15:06:21.000000000","message":"Patch Set 5:\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"765aba9181351f2362f4f747adc64f903568370a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"date":"2025-08-18 15:19:45.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"d06b2dcc4e1d257cd4966afcd9200be551323131","author":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"date":"2025-08-18 15:20:57.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"36416f66b81d28281098031178aa87872b1af788","author":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"date":"2025-08-18 15:26:38.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"1a793a6f64b523e430472044d7f05840ebfef6ec","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2025-08-18 17:44:16.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"fab35766dad86636edf279692e95fb8b8bf8ca45","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-08-18 18:46:25.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/7628903fd3f940ee8bc7b541f90d5afa : SUCCESS in 46m 48s","accounts_in_message":[],"_revision_number":6},{"id":"7f8d876aa6786060b4b73ce73a6d956e66c5e895","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-18 18:51:52.000000000","message":"Patch Set 6:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1600/.","accounts_in_message":[],"_revision_number":6},{"id":"bd718215b184acbf9946e4675a06a259fe5a35df","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-18 19:03:46.000000000","message":"Patch Set 6:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1465/.","accounts_in_message":[],"_revision_number":6},{"id":"59ba9bd2ba16080d0eb63a5c50f1a6486e3733ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 19:32:34.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d76da8ce06df4d46819eb8c0f5120f07\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ac5c7eadec5244d4ad93402d833593ad : SUCCESS in 1m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/94d2043dc0584841a53af3341bcfb891 : SUCCESS in 55m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f61af3d63b2b4952b22671b7b807e0a8 : SUCCESS in 2m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3b81110c6568422290d325db92d293b5 : SUCCESS in 2m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/71cb5e76e6a04b78a3f2b190f440ad7a : SUCCESS in 3m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a81f5e480821452e8d5d84e64908d8fe : SUCCESS in 5m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a458008c17b04115b61f7be00387a42f : SUCCESS in 6m 51s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/ce834d34380646a598aa72114e6bb2a5 : SUCCESS in 5m 11s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/968aa69c02b4464fa78a066d75227001 : SUCCESS in 1h 10m 00s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/3c51dac2aade4ab6a5a021e55edf27cb : SUCCESS in 1h 52m 15s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/3e9150c738b446ff9835b6b71652f7a1 : SUCCESS in 53m 08s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/8d51faaaafba4937a8f6ccf38faedc3c : SUCCESS in 3m 11s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"c69fba5bac212c0befb0a2d5423cfece29ac5e42","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"date":"2025-08-19 12:51:43.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"1ee91a7829f4a2b36b4c8ce5750be674549432a6","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-19 14:49:14.000000000","message":"Patch Set 7:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1481/.","accounts_in_message":[],"_revision_number":7},{"id":"10c0de4f2a152d47978d51b1bb33007d19861c89","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 14:49:51.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d9ebd73415174603888f41161e598962\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/2016096a59064cdab416a6bf8c25bb63 : SUCCESS in 2m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9cf560bc923141b2983abdb88c00fc7c : SUCCESS in 1h 50m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c850203a315545fb8c8bcfd2c0614eba : SUCCESS in 2m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6d8cbf62d3484c80803c6f8018b2f183 : SUCCESS in 2m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5aca93bccd034a95989ffc67db56de55 : SUCCESS in 2m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9c7d7bd4bae54e049303e2043d7b1249 : SUCCESS in 3m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e57e6bac75f4d6ea8642f6bbb824c19 : SUCCESS in 3m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3e023486f01a4b37a209649a24c0c6a6 : FAILURE in 5m 41s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/b5a55b931b5b46fdb0bd07f810e08558 : SUCCESS in 2m 37s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/713a60a308c54a74938f3dc909ee76ab : SUCCESS in 1h 42m 53s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/22b232d16e5f4849ae186cc279c0794a : SUCCESS in 1h 57m 26s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/eb34d81d36e349f58f57333b3fb4be9b : SUCCESS in 1h 49m 15s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/b51fa8b836764adba9182c2776394794 : SUCCESS in 2m 47s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"5f4691ef1313a06a37a9a993dc1b6733b57fcfe0","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-19 14:53:58.000000000","message":"Patch Set 7:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1616/.","accounts_in_message":[],"_revision_number":7},{"id":"8bde1f4a736110245d9920c4ee8e885fac6d8564","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-08-19 15:04:00.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/9dcba124526041459f8a7eb9f7da099e : SUCCESS in 48m 09s","accounts_in_message":[],"_revision_number":7},{"id":"239bb9b634fad3cbeb260a2b0d3a99b4a4a2709c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"date":"2025-08-20 01:28:42.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"25d0d1a5df8dc14acc6562a0350402efc1e949c5","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-08-20 02:11:34.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/319557766e564dc79c20bc75d56080eb : SUCCESS in 41m 02s","accounts_in_message":[],"_revision_number":8},{"id":"8809368b56bda334a4087cf67a66f82a3d0ee9fe","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-20 02:48:48.000000000","message":"Patch Set 8:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1490/.","accounts_in_message":[],"_revision_number":8},{"id":"3b3dd19ac36e671854c671856322e3fb1c43223a","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-20 02:54:49.000000000","message":"Patch Set 8:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1625/.","accounts_in_message":[],"_revision_number":8},{"id":"0429f7bdfe93d0fee0ba548c9c4ceb8de1073416","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-20 03:05:23.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/78fa9e90089b472e8b600a2f9770b51c\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/22a0a4f46a3c4d718ddaa1f6e2a14fde : SUCCESS in 1m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/196618fd9f324be5ad092720213d1be9 : SUCCESS in 1h 02m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f3ff074a2c441a6887517127c6dc149 : SUCCESS in 3m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/07facd98282b4a73aaf50ce240503048 : SUCCESS in 7m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7ca4ce833bbf4f728947a038f164d53d : SUCCESS in 3m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3871228920d34e0cb1623de7e882cd42 : SUCCESS in 5m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7970ccbac9e4041b38437fbd6382fa6 : SUCCESS in 6m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9d37f74b5bd64c2683f2982aaeabec32 : SUCCESS in 3m 44s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/a31baea6c1c647c2ac2746a8f3273ffb : SUCCESS in 4m 37s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/ecd4724a2bae422c8faed47a40381cb0 : SUCCESS in 1h 35m 48s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/91bfd22fe5e741f2a3bec143de20d2ad : SUCCESS in 1h 35m 18s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/83083a230ebe462c8eb6b5233655b70d : FAILURE in 1h 12m 51s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/3a150754e78047cc9783aa5b1c94852c : SUCCESS in 2m 51s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"3caa11d3b03896b919682461476402de871f866a","author":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"date":"2025-08-20 05:31:08.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"fdcf02966177766d844e6b0c1d533242c561fabb","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2025-08-20 14:14:43.000000000","message":"Patch Set 8: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"9b6ddd79ccd2aa438a4df1abb2b6dfa5fd5e6ff9","author":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"date":"2025-08-21 09:03:23.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"fdfe96bd857f14c0b899703cfbd532f97c822dd7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"date":"2025-08-21 09:09:22.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"661ba42408af1b2230e2f69ae5f2f3010c753313","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-21 11:31:35.000000000","message":"Patch Set 9: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/09d3f392c81e48eb926ce40ed04c2a1b\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/4adf273a253d4686be09520e4a8c32e8 : SUCCESS in 2m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8792ed82862a48a1acef894662375cdb : SUCCESS in 1h 48m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/655d76da7da14e47ba731a84f6c6dd04 : FAILURE in 3m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/88d0dce063b149649edc8cd5fac8e178 : SUCCESS in 2m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4e5bc2df128d4fe7b860b9189f545108 : SUCCESS in 4m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/59f858a1d6f9483aaea2ff07b6d51e6c : SUCCESS in 6m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77c1439b834d42868fd85c1341b7752f : SUCCESS in 5m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d7880eacaf814f25aa6cc3e4217d341e : SUCCESS in 6m 39s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/428ec8e553834f7f9df1a5d6a6e820cb : SUCCESS in 4m 07s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/182ea1129dfa4744b3c3f9c2d5131bf1 : SUCCESS in 1h 00m 08s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/178d287e972c4afeb91aabf53d6d99c4 : SUCCESS in 2h 21m 14s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/debba9c640b54e1990fca1856d1de5c4 : SUCCESS in 1h 37m 52s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/8f97a30456b44872810290576d28c700 : SUCCESS in 4m 40s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"c07c6d0b038af311ad4d92886c6632dafbe18692","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"date":"2025-08-21 12:05:48.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"6346259e203eeed48d8def00573f5a0c09ea4ea2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-21 14:01:38.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/64ae628a9fd14254a36aebb3c46a68a9\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/5ba74c700e084e1c8218925dfecc7774 : SUCCESS in 2m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0a920be157c142cd9c2a8572dca41dd7 : SUCCESS in 1h 52m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ecc5b56377f4cdfb66644606014158c : SUCCESS in 5m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/79e49f1e23d44bef843b9befc4d0021b : SUCCESS in 4m 12s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b0cc925093c24088ab615a9da9db3eb6 : SUCCESS in 4m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/effdbd201e4848acab60597175c6966f : SUCCESS in 6m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91ec0f3c614a44a28006788cad43f059 : SUCCESS in 9m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cae8dd95fd784b948ca13068adef6341 : SUCCESS in 3m 34s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/df610fa8eca34b39bc57983fded7e989 : SUCCESS in 5m 56s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/13573881c0e040d0aa738327c3df344c : SUCCESS in 1h 38m 00s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/2753fc34244c4fdb8964cba4d9ab7451 : SUCCESS in 1h 44m 22s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/5a493d8e1d8b4aad82589a0f255f6604 : SUCCESS in 1h 31m 35s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/b8366f5948df485f8631a0ac679f8b8f : SUCCESS in 5m 29s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"37f87f923db068aeb527bf6a8b4ef36fb33a14a9","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-08-21 16:38:57.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/3bd872f76bc74c61bd42169405384dfd : SUCCESS in 44m 27s","accounts_in_message":[],"_revision_number":10},{"id":"c32a73f444dd95c356b66eb2da8e4e311eedde78","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-21 20:25:42.000000000","message":"Patch Set 10:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1519/.","accounts_in_message":[],"_revision_number":10},{"id":"69ce2c44a9cc1bd492c729e567d50bef643e1c67","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-21 21:20:19.000000000","message":"Patch Set 10:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1654/.","accounts_in_message":[],"_revision_number":10},{"id":"87b9e57321ed97576a823bb010501bb09d41de0b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"date":"2025-08-25 09:28:16.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"8a9630e14d70d667b657cc1275d721ab18876417","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2025-08-25 09:30:10.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/7db5ad9c146546bda3f903409dbb3526 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":11},{"id":"bf7faa444a797e4772156422d09fd13975a9d3d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-25 11:27:18.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d8ca956345f84faa9a702ba889e6ea00\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c68f094fd44742b89c335f9dac913974 : SUCCESS in 2m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9757bb4687da402886a42bc4ef1d6ca7 : SUCCESS in 1h 47m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/489b93c60a544a6a83068b9fe3bd7ecd : SUCCESS in 3m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1d482712169f422996ca495056127c0d : SUCCESS in 2m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4f54abe4aba94aa6975e2d2379452562 : SUCCESS in 3m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7b8b35241b224e328d2ef7a79674d03a : SUCCESS in 7m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/542db5824d544bb38355159b26bd298f : SUCCESS in 7m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e85107c441fb4813949e28a71cd7a6e2 : SUCCESS in 3m 50s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/b96d85d04a3e43259eec32ad8b619c53 : SUCCESS in 5m 33s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/9d7922e70e6e40a99eb48ab34d99f873 : SUCCESS in 1h 54m 20s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/3ad46b85e499418aa1fdb1bbc52b3c24 : SUCCESS in 1h 48m 41s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/8f9258484df84cc18c487d5a1df2f228 : SUCCESS in 1h 26m 38s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/727853e99ffb445fb3dd9a765e8450fe : SUCCESS in 4m 08s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"ad807758c9ab4ced43e578a21440facfb588795d","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-25 12:46:50.000000000","message":"Patch Set 11:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/1549/.","accounts_in_message":[],"_revision_number":11},{"id":"61ea6cb4b414ca2bab12aece0e34faf2075fe103","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2025-08-25 13:05:05.000000000","message":"Patch Set 11:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/1683/.","accounts_in_message":[],"_revision_number":11},{"id":"487cac3fb0e5e7045a5ecae6ce7f2f1f82da12a2","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2025-09-02 03:29:33.000000000","message":"Patch Set 11:\n\nBuild failed. To recheck use \u0027run-NEC Cinder CI\u0027. For 3rd party ci contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/99/957599/11/20250902020716/NEC-FC : FAILURE in 10h 14m 18s\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/99/957599/11/20250902020716/NEC-ISCSI : FAILURE in 10h 22m 14s","accounts_in_message":[],"_revision_number":11},{"id":"d6cc6e2c45da1e200f2ed43f27bbc1c448a272ed","author":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"date":"2025-10-12 06:36:22.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"475b7451a7b74b3b0031f2cfdfa2260bd464a644","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-01-14 21:33:22.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"6b9b20a7c4a72c6412f953431f9dd9adb200aebf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"date":"2026-01-15 13:44:54.000000000","message":"Uploaded patch set 12: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":12},{"id":"2a5af07580a5a08fccbceb4bad1b784f1e5c42a7","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-01-15 14:46:31.000000000","message":"Patch Set 12:\n\nBuild Failed \n\n* quobyteci:  os-brick_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I19065ad683b9ff8174814d6e8c500178c885bee4 .","accounts_in_message":[],"_revision_number":12},{"id":"e263dedbd03f508553e90ac028bcf0135bcb7b40","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-15 16:01:14.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a3bb48d4b4c8458da06d4835edbecd77\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/34c7d660e1464c13847648ce8325d70f : SUCCESS in 2m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9b7a5a638d1f47e795d3f51b2de65d64 : SUCCESS in 1h 31m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29363bfcac9a4a0db8408994f608c884 : SUCCESS in 3m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d360712d6f1d4fa2948c0faba96c5f0e : SUCCESS in 3m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6a20eab0dfb14e8a98bca3894ea7204c : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04a0ab03819e454c94aa4c256067e5be : SUCCESS in 7m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ddcd6ef273e0474d9f8c5c89c95514d1 : SUCCESS in 2m 46s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/3dafc9659eb84c688378de4651c7dbec : SUCCESS in 2m 16s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/7c7cdf9883344ce391c2f48e08ca7ea4 : SUCCESS in 2h 11m 02s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b76277aa5f894677aa9db2b8272f4ecb : SUCCESS in 1h 59m 21s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/a1853f58427349c5bf86cb0266ba51a0 : SUCCESS in 2h 02m 42s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/11b1637023884636b251652075c92e5a : SUCCESS in 3m 37s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"e5bcce746c50a03683e1fbcc407a4733611448fb","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-01-15 16:08:14.000000000","message":"Patch Set 12:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/15_01_2026_16_01_24\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":12},{"id":"6cb6522ea5ac6356bdcfd79ef456b64916821a56","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-01-16 08:08:46.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/26efd735735245779e7538017b8aad29 : FAILURE in 1h 03m 59s","accounts_in_message":[],"_revision_number":12},{"id":"fc5d266445f05e3c7fb73cfcf5a2dedfea4b6ec6","author":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"date":"2026-01-21 14:38:39.000000000","message":"Patch Set 12: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"8cd9a306a86b023541a1a050bdc5d57edd06e5a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"date":"2026-01-22 08:25:10.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":13},{"id":"eb6dee3297582d01bcdbbae0ab9abfad7c155204","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-01-22 09:27:02.000000000","message":"Patch Set 13:\n\nBuild Successful \n\n* quobyteci:  os-brick_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I19065ad683b9ff8174814d6e8c500178c885bee4 .","accounts_in_message":[],"_revision_number":13},{"id":"3c35924a094b235715784032141cdcb5138b20a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-22 10:33:08.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b66f7f90b8564d58bffbac9af76b21d7\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/bc0b99e69fb5451e91f60072e77e54f1 : SUCCESS in 5m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cbbbfbc9abe9429194222b13ebdb049c : SUCCESS in 1h 43m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2311bac3f5084fa4bec463218cd2d469 : SUCCESS in 3m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d5a14747f963413fb6c8e41f8b497137 : SUCCESS in 3m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/70b3c7a9029c42d485882aaf5e6d2bb6 : SUCCESS in 4m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c676e676fb1466d9ee78bcb03c52b6d : SUCCESS in 7m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/10fc387c5ac148d3a6c1c799a543271c : SUCCESS in 4m 04s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/fac421a7ef954ad9b3404dd881707cf8 : SUCCESS in 3m 45s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/538952204c76447b823901a594415621 : SUCCESS in 1h 44m 19s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/6083dc1993b14fd99dd27cc6df32aa3f : SUCCESS in 2h 01m 54s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/16e7206d9fed440a896da3a8b24f488f : SUCCESS in 1h 24m 56s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/621ce39675ca4cf59ba8423997f00267 : SUCCESS in 2m 46s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"d73fa09daa177d8d44bd0f6d679f9dc1722c6176","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-01-22 10:39:55.000000000","message":"Patch Set 13:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/22_01_2026_10_33_19\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":13},{"id":"b09664d543684afb47124ff89c4ac86c0b88b919","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-01-22 16:16:07.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/65fa7a24baae458690f7b535e8bb6dbc : FAILURE in 43m 03s","accounts_in_message":[],"_revision_number":13},{"id":"2bffb64df75181f74cbf03ecd2e3ea5c1c92bf21","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-02-04 14:28:29.000000000","message":"Patch Set 13: Code-Review+2 Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"b511c0ebb00e013f9b19708d25d67ea3a54d39a0","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2026-02-04 14:32:48.000000000","message":"Patch Set 13: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":13},{"id":"ac2324ff1a163f3c34064f68b7cf8ba62454c0ce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-04 14:33:11.000000000","message":"Patch Set 13: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":13},{"id":"f49b5a846d440178b2f5976037189ddd26e7d8f4","author":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"date":"2026-02-04 15:03:29.000000000","message":"Patch Set 13: Code-Review+1","accounts_in_message":[],"_revision_number":13},{"id":"77cac256aaa2d1210c76314759e3b6f8d82172fe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-04 16:32:11.000000000","message":"Patch Set 13: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/efd3de8d9bdf4c2dac29c6597fbb09cc\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/4eea6a8a88de4829a94faaa8356239b8 : SUCCESS in 2m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/18cbbc728629480b90778cb9c2b279b1 : SUCCESS in 1h 17m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/289e9f1cf3384c33b91219354707bfed : SUCCESS in 3m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b9096d583aa54b809c2a3495d54e8b77 : SUCCESS in 3m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8ad71832fb724f49bd63f689ed68ce73 : SUCCESS in 3m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba29e7c5dd2f444ba78b05578ec3903c : SUCCESS in 8m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/39cc10db53074bc6adf14fb2f37bdc5d : SUCCESS in 3m 35s\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/96cbc6e759b24eaf9b8bbbc90783535c : SUCCESS in 1h 57m 35s","accounts_in_message":[],"_revision_number":13},{"id":"c76d18b205263b7f1ccd9ec75542dc3d415da370","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-04 16:32:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":13},{"id":"6e7068f3f23f063e68a5e3c605b0d84a84fffc70","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-04 16:33:04.000000000","message":"Patch Set 13:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a63d926ab73144cd9e85e79a7f30ff3b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6cad213e3d8c4fa3af50b7e1c4601fe5 : SUCCESS in 41s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f5bc7b48b8184bd19be04533e4d596f8 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":13}],"current_revision_number":13,"current_revision":"e7d2cd0ad3a9d37132103cacd1fb1a5d5acec52a","revisions":{"f6fee87a5d9f409ae49ac21c2afb4b9529f17812":{"kind":"REWORK","_number":1,"created":"2025-08-17 08:32:43.000000000","uploader":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"ref":"refs/changes/99/957599/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/99/957599/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/99/957599/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/99/957599/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/99/957599/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/99/957599/1"}}},"commit":{"parents":[{"commit":"0a5a387523a449060df088182f2242765b6f2436","subject":"iSCSI: Add debug log for WWN and multipath","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/0a5a387523a449060df088182f2242765b6f2436"}]}],"author":{"name":"yuval","email":"yuval@lightbitslabs.com","date":"2025-08-14 09:24:30.000000000","tz":180},"committer":{"name":"yuval","email":"yuval@lightbitslabs.com","date":"2025-08-17 08:32:24.000000000","tz":180},"subject":"LightOS: make discovery client configuration configurable","message":"LightOS: make discovery client configuration configurable\n\nThe LightOS connector uses hardcoded values for discovery client\nconnection parameters, preventing customization for different\ndeployment environments and containers with custom networks.\n\nAdd three configurable options to os_brick group:\n- discovery_client_address (default: localhost)\n- discovery_client_port (default: 6060)\n- discovery_client_dir_path (default: /etc/discovery-client/discovery.d/)\n\nThis enables flexible deployments and container networking while\nmaintaining backward compatibility with existing configurations.\n\nChange-Id: I19065ad683b9ff8174814d6e8c500178c885bee4\nSigned-off-by: yuval \u003cyuval@lightbitslabs.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/f6fee87a5d9f409ae49ac21c2afb4b9529f17812"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/f6fee87a5d9f409ae49ac21c2afb4b9529f17812"}]},"branch":"refs/heads/master"},"b27a8e2b1731c9862cf05999d605e506940829e4":{"kind":"REWORK","_number":2,"created":"2025-08-18 08:17:43.000000000","uploader":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"ref":"refs/changes/99/957599/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/99/957599/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/99/957599/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/99/957599/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/99/957599/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/99/957599/2"}}},"commit":{"parents":[{"commit":"0a5a387523a449060df088182f2242765b6f2436","subject":"iSCSI: Add debug log for WWN and multipath","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/0a5a387523a449060df088182f2242765b6f2436"}]}],"author":{"name":"yuval","email":"yuval@lightbitslabs.com","date":"2025-08-14 09:24:30.000000000","tz":180},"committer":{"name":"yuval","email":"yuval@lightbitslabs.com","date":"2025-08-18 08:17:29.000000000","tz":180},"subject":"LightOS: make discovery client configuration configurable","message":"LightOS: make discovery client configuration configurable\n\nThe LightOS connector uses hardcoded values for discovery client\nconnection parameters, preventing customization for different\ndeployment environments and containers with custom networks.\n\nAdd three configurable options to os_brick group:\n- discovery_client_address (default: localhost)\n- discovery_client_port (default: 6060)\n- discovery_client_dir_path (default: /etc/discovery-client/discovery.d/)\n\nThis enables flexible deployments and container networking while\nmaintaining backward compatibility with existing configurations.\n\nChange-Id: I19065ad683b9ff8174814d6e8c500178c885bee4\nSigned-off-by: yuval \u003cyuval@lightbitslabs.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/b27a8e2b1731c9862cf05999d605e506940829e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/b27a8e2b1731c9862cf05999d605e506940829e4"}]},"branch":"refs/heads/master"},"31fa82bd9b81a7ea19e8f54c4552747c1fa17f41":{"kind":"REWORK","_number":3,"created":"2025-08-18 14:43:12.000000000","uploader":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"ref":"refs/changes/99/957599/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/99/957599/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/99/957599/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/99/957599/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/99/957599/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/99/957599/3"}}},"commit":{"parents":[{"commit":"0a5a387523a449060df088182f2242765b6f2436","subject":"iSCSI: Add debug log for WWN and multipath","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/0a5a387523a449060df088182f2242765b6f2436"}]}],"author":{"name":"yuval","email":"yuval@lightbitslabs.com","date":"2025-08-14 09:24:30.000000000","tz":180},"committer":{"name":"yuval","email":"yuval@lightbitslabs.com","date":"2025-08-18 14:43:04.000000000","tz":180},"subject":"LightOS: make discovery client configuration configurable","message":"LightOS: make discovery client configuration configurable\n\nThe LightOS connector uses hardcoded values for discovery client\nconnection parameters, preventing customization for different\ndeployment environments and containers with custom networks.\n\nAdd three configurable options to os_brick group:\n- discovery_client_address (default: localhost)\n- discovery_client_port (default: 6060)\n- discovery_client_dir_path (default: /etc/discovery-client/discovery.d/)\n\nThis enables flexible deployments and container networking while\nmaintaining backward compatibility with existing configurations.\n\nChange-Id: I19065ad683b9ff8174814d6e8c500178c885bee4\nSigned-off-by: yuval \u003cyuval@lightbitslabs.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/31fa82bd9b81a7ea19e8f54c4552747c1fa17f41"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/31fa82bd9b81a7ea19e8f54c4552747c1fa17f41"}]},"branch":"refs/heads/master"},"25b46f16cb6f222fbf137400fa65c8af1c85239e":{"kind":"REWORK","_number":4,"created":"2025-08-18 14:58:57.000000000","uploader":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"ref":"refs/changes/99/957599/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/99/957599/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/99/957599/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/99/957599/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/99/957599/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/99/957599/4"}}},"commit":{"parents":[{"commit":"0a5a387523a449060df088182f2242765b6f2436","subject":"iSCSI: Add debug log for WWN and multipath","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/0a5a387523a449060df088182f2242765b6f2436"}]}],"author":{"name":"yuval","email":"yuval@lightbitslabs.com","date":"2025-08-14 09:24:30.000000000","tz":180},"committer":{"name":"yuval","email":"yuval@lightbitslabs.com","date":"2025-08-18 14:58:48.000000000","tz":180},"subject":"LightOS: make discovery client configuration configurable","message":"LightOS: make discovery client configuration configurable\n\nThe LightOS connector uses hardcoded values for discovery client\nconnection parameters, preventing customization for different\ndeployment environments and containers with custom networks.\n\nAdd three configurable options to os_brick group:\n- discovery_client_address (default: localhost)\n- discovery_client_port (default: 6060)\n- discovery_client_dir_path (default: /etc/discovery-client/discovery.d/)\n\nThis enables flexible deployments and container networking while\nmaintaining backward compatibility with existing configurations.\n\nChange-Id: I19065ad683b9ff8174814d6e8c500178c885bee4\nSigned-off-by: yuval \u003cyuval@lightbitslabs.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/25b46f16cb6f222fbf137400fa65c8af1c85239e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/25b46f16cb6f222fbf137400fa65c8af1c85239e"}]},"branch":"refs/heads/master"},"ecabb0d28fe20276b43772f932e8674a0ec56227":{"kind":"REWORK","_number":5,"created":"2025-08-18 15:05:13.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/99/957599/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/99/957599/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/99/957599/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/99/957599/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/99/957599/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/99/957599/5"}}},"commit":{"parents":[{"commit":"0a5a387523a449060df088182f2242765b6f2436","subject":"iSCSI: Add debug log for WWN and multipath","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/0a5a387523a449060df088182f2242765b6f2436"}]}],"author":{"name":"yuval","email":"yuval@lightbitslabs.com","date":"2025-08-14 09:24:30.000000000","tz":180},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-18 15:05:05.000000000","tz":540},"subject":"LightOS: make discovery client configuration configurable","message":"LightOS: make discovery client configuration configurable\n\nThe LightOS connector uses hardcoded values for discovery client\nconnection parameters, preventing customization for different\ndeployment environments and containers with custom networks.\n\nAdd three configurable options to os_brick group:\n- discovery_client_address (default: localhost)\n- discovery_client_port (default: 6060)\n- discovery_client_dir_path (default: /etc/discovery-client/discovery.d/)\n\nThis enables flexible deployments and container networking while\nmaintaining backward compatibility with existing configurations.\n\nChange-Id: I19065ad683b9ff8174814d6e8c500178c885bee4\nSigned-off-by: yuval \u003cyuval@lightbitslabs.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/ecabb0d28fe20276b43772f932e8674a0ec56227"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/ecabb0d28fe20276b43772f932e8674a0ec56227"}]},"branch":"refs/heads/master"},"30b4f60ec7bb7ec42540f87a3e646bc669e9bb81":{"kind":"REWORK","_number":6,"created":"2025-08-18 15:19:45.000000000","uploader":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"ref":"refs/changes/99/957599/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/99/957599/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/99/957599/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/99/957599/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/99/957599/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/99/957599/6"}}},"commit":{"parents":[{"commit":"0a5a387523a449060df088182f2242765b6f2436","subject":"iSCSI: Add debug log for WWN and multipath","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/0a5a387523a449060df088182f2242765b6f2436"}]}],"author":{"name":"yuval","email":"yuval@lightbitslabs.com","date":"2025-08-14 09:24:30.000000000","tz":180},"committer":{"name":"yuval","email":"yuval@lightbitslabs.com","date":"2025-08-18 15:19:24.000000000","tz":180},"subject":"LightOS: make discovery client configuration configurable","message":"LightOS: make discovery client configuration configurable\n\nThe LightOS connector uses hardcoded values for discovery client\nconnection parameters, preventing customization for different\ndeployment environments and containers with custom networks.\n\nAdd three configurable options to os_brick group:\n- discovery_client_address (default: localhost)\n- discovery_client_port (default: 6060)\n- discovery_client_dir_path (default: /etc/discovery-client/discovery.d/)\n\nThis enables flexible deployments and container networking while\nmaintaining backward compatibility with existing configurations.\n\nChange-Id: I19065ad683b9ff8174814d6e8c500178c885bee4\nSigned-off-by: yuval \u003cyuval@lightbitslabs.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/30b4f60ec7bb7ec42540f87a3e646bc669e9bb81"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/30b4f60ec7bb7ec42540f87a3e646bc669e9bb81"}]},"branch":"refs/heads/master"},"004488ebe4e96528bb72e7af0cae50819c282a67":{"kind":"REWORK","_number":7,"created":"2025-08-19 12:51:43.000000000","uploader":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"ref":"refs/changes/99/957599/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/99/957599/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/99/957599/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/99/957599/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/99/957599/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/99/957599/7"}}},"commit":{"parents":[{"commit":"0a5a387523a449060df088182f2242765b6f2436","subject":"iSCSI: Add debug log for WWN and multipath","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/0a5a387523a449060df088182f2242765b6f2436"}]}],"author":{"name":"yuval","email":"yuval@lightbitslabs.com","date":"2025-08-14 09:24:30.000000000","tz":180},"committer":{"name":"yuval","email":"yuval@lightbitslabs.com","date":"2025-08-19 12:51:33.000000000","tz":180},"subject":"LightOS: make discovery client configuration configurable","message":"LightOS: make discovery client configuration configurable\n\nThe LightOS connector uses hardcoded values for discovery client\nconnection parameters, preventing customization for different\ndeployment environments and containers with custom networks.\n\nAdd three configurable options to os_brick group:\n- discovery_client_address (default: localhost)\n- discovery_client_port (default: 6060)\n- discovery_client_dir_path (default: /etc/discovery-client/discovery.d/)\n\nThis enables flexible deployments and container networking while\nmaintaining backward compatibility with existing configurations.\n\nChange-Id: I19065ad683b9ff8174814d6e8c500178c885bee4\nSigned-off-by: yuval \u003cyuval@lightbitslabs.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/004488ebe4e96528bb72e7af0cae50819c282a67"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/004488ebe4e96528bb72e7af0cae50819c282a67"}]},"branch":"refs/heads/master"},"ff57dd5049d8b8629bb0eb564e4151a3ca5520b6":{"kind":"REWORK","_number":8,"created":"2025-08-20 01:28:42.000000000","uploader":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"ref":"refs/changes/99/957599/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/99/957599/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/99/957599/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/99/957599/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/99/957599/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/99/957599/8"}}},"commit":{"parents":[{"commit":"0a5a387523a449060df088182f2242765b6f2436","subject":"iSCSI: Add debug log for WWN and multipath","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/0a5a387523a449060df088182f2242765b6f2436"}]}],"author":{"name":"yuval","email":"yuval@lightbitslabs.com","date":"2025-08-14 09:24:30.000000000","tz":180},"committer":{"name":"yuval","email":"yuval@lightbitslabs.com","date":"2025-08-20 01:28:33.000000000","tz":180},"subject":"LightOS: make discovery client configuration configurable","message":"LightOS: make discovery client configuration configurable\n\nThe LightOS connector uses hardcoded values for discovery client\nconnection parameters, preventing customization for different\ndeployment environments and containers with custom networks.\n\nAdd three configurable options to os_brick group:\n- discovery_client_address (default: localhost)\n- discovery_client_port (default: 6060)\n- discovery_client_dir_path (default: /etc/discovery-client/discovery.d/)\n\nThis enables flexible deployments and container networking while\nmaintaining backward compatibility with existing configurations.\n\nChange-Id: I19065ad683b9ff8174814d6e8c500178c885bee4\nSigned-off-by: yuval \u003cyuval@lightbitslabs.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/ff57dd5049d8b8629bb0eb564e4151a3ca5520b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/ff57dd5049d8b8629bb0eb564e4151a3ca5520b6"}]},"branch":"refs/heads/master"},"b2ab1929c35271b7cc7d5cccf7f18c543d09928d":{"kind":"REWORK","_number":9,"created":"2025-08-21 09:09:22.000000000","uploader":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"ref":"refs/changes/99/957599/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/99/957599/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/99/957599/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/99/957599/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/99/957599/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/99/957599/9"}}},"commit":{"parents":[{"commit":"0a5a387523a449060df088182f2242765b6f2436","subject":"iSCSI: Add debug log for WWN and multipath","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/0a5a387523a449060df088182f2242765b6f2436"}]}],"author":{"name":"yuval","email":"yuval@lightbitslabs.com","date":"2025-08-14 09:24:30.000000000","tz":180},"committer":{"name":"yuval","email":"yuval@lightbitslabs.com","date":"2025-08-21 09:09:13.000000000","tz":180},"subject":"LightOS: make discovery client configuration configurable","message":"LightOS: make discovery client configuration configurable\n\nThe LightOS connector uses hardcoded values for discovery client\nconnection parameters, preventing customization for different\ndeployment environments and containers with custom networks.\n\nAdd three configurable options to os_brick group:\n- discovery_client_address (default: localhost)\n- discovery_client_port (default: 6060)\n- discovery_client_dir_path (default: /etc/discovery-client/discovery.d/)\n\nThis enables flexible deployments and container networking while\nmaintaining backward compatibility with existing configurations.\n\nChange-Id: I19065ad683b9ff8174814d6e8c500178c885bee4\nSigned-off-by: yuval \u003cyuval@lightbitslabs.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/b2ab1929c35271b7cc7d5cccf7f18c543d09928d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/b2ab1929c35271b7cc7d5cccf7f18c543d09928d"}]},"branch":"refs/heads/master"},"e3f5a4ad6e25249f5d50ca301cbb38e6ce9add6d":{"kind":"REWORK","_number":10,"created":"2025-08-21 12:05:48.000000000","uploader":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"ref":"refs/changes/99/957599/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/99/957599/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/99/957599/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/99/957599/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/99/957599/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/99/957599/10"}}},"commit":{"parents":[{"commit":"0a5a387523a449060df088182f2242765b6f2436","subject":"iSCSI: Add debug log for WWN and multipath","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/0a5a387523a449060df088182f2242765b6f2436"}]}],"author":{"name":"yuval","email":"yuval@lightbitslabs.com","date":"2025-08-14 09:24:30.000000000","tz":180},"committer":{"name":"yuval","email":"yuval@lightbitslabs.com","date":"2025-08-21 12:05:39.000000000","tz":180},"subject":"LightOS: make discovery client configuration configurable","message":"LightOS: make discovery client configuration configurable\n\nThe LightOS connector uses hardcoded values for discovery client\nconnection parameters, preventing customization for different\ndeployment environments and containers with custom networks.\n\nAdd three configurable options to os_brick group:\n- discovery_client_address (default: localhost)\n- discovery_client_port (default: 6060)\n- discovery_client_dir_path (default: /etc/discovery-client/discovery.d/)\n\nThis enables flexible deployments and container networking while\nmaintaining backward compatibility with existing configurations.\n\nChange-Id: I19065ad683b9ff8174814d6e8c500178c885bee4\nSigned-off-by: yuval \u003cyuval@lightbitslabs.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/e3f5a4ad6e25249f5d50ca301cbb38e6ce9add6d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/e3f5a4ad6e25249f5d50ca301cbb38e6ce9add6d"}]},"branch":"refs/heads/master"},"223f434a7feb8edd34fd35537235008732d7eca4":{"kind":"REWORK","_number":11,"created":"2025-08-25 09:28:16.000000000","uploader":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"ref":"refs/changes/99/957599/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/99/957599/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/99/957599/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/99/957599/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/99/957599/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/99/957599/11"}}},"commit":{"parents":[{"commit":"0a5a387523a449060df088182f2242765b6f2436","subject":"iSCSI: Add debug log for WWN and multipath","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/0a5a387523a449060df088182f2242765b6f2436"}]}],"author":{"name":"yuval","email":"yuval@lightbitslabs.com","date":"2025-08-14 09:24:30.000000000","tz":180},"committer":{"name":"yuval","email":"yuval@lightbitslabs.com","date":"2025-08-25 09:28:06.000000000","tz":180},"subject":"LightOS: make discovery client configuration configurable","message":"LightOS: make discovery client configuration configurable\n\nThe LightOS connector uses hardcoded values for discovery client\nconnection parameters, preventing customization for different\ndeployment environments and containers with custom networks.\n\nAdd three configurable options to os_brick group:\n- discovery_client_address (default: localhost)\n- discovery_client_port (default: 6060)\n- discovery_client_dir_path (default: /etc/discovery-client/discovery.d/)\n\nThis enables flexible deployments and container networking while\nmaintaining backward compatibility with existing configurations.\n\nChange-Id: I19065ad683b9ff8174814d6e8c500178c885bee4\nSigned-off-by: yuval \u003cyuval@lightbitslabs.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/223f434a7feb8edd34fd35537235008732d7eca4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/223f434a7feb8edd34fd35537235008732d7eca4"}]},"branch":"refs/heads/master"},"a93df41abd00d32c69e46fd0482f9c289c7775a1":{"kind":"NO_CODE_CHANGE","_number":12,"created":"2026-01-15 13:44:54.000000000","uploader":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"ref":"refs/changes/99/957599/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/99/957599/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/99/957599/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/99/957599/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/99/957599/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/99/957599/12"}}},"commit":{"parents":[{"commit":"0a5a387523a449060df088182f2242765b6f2436","subject":"iSCSI: Add debug log for WWN and multipath","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/0a5a387523a449060df088182f2242765b6f2436"}]}],"author":{"name":"yuval","email":"yuval@lightbitslabs.com","date":"2025-08-14 09:24:30.000000000","tz":180},"committer":{"name":"yuval","email":"yuval@lightbitslabs.com","date":"2026-01-15 13:38:44.000000000","tz":120},"subject":"LightOS: update discovery client configuration","message":"LightOS: update discovery client configuration\n\nThe LightOS connector uses hardcoded values for discovery client\nconnection parameters, preventing customization for different\ndeployment environments and containers with custom networks.\n\nExample of when this is needed:\nWe have 1 Cinder instance in a Docker container with network 10.10.10.1,\nand a discovery-service located at 10.10.10.2.\nIn the same cluster, we have a Nova instance in a container with\nnetwork 20.20.20.1, and a discovery-service at 20.20.20.2.\nI need to be able to set the network address for each instance.\n\nDefault values are set for backward compatibility.\n\nChange-Id: I19065ad683b9ff8174814d6e8c500178c885bee4\nSigned-off-by: yuval \u003cyuval@lightbitslabs.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/a93df41abd00d32c69e46fd0482f9c289c7775a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/a93df41abd00d32c69e46fd0482f9c289c7775a1"}]},"branch":"refs/heads/master"},"e7d2cd0ad3a9d37132103cacd1fb1a5d5acec52a":{"kind":"REWORK","_number":13,"created":"2026-01-22 08:25:10.000000000","uploader":{"_account_id":33612,"name":"yuval","email":"yuval@lightbitslabs.com","username":"yuval"},"ref":"refs/changes/99/957599/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/99/957599/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/99/957599/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/99/957599/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/99/957599/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/99/957599/13"}}},"commit":{"parents":[{"commit":"0a5a387523a449060df088182f2242765b6f2436","subject":"iSCSI: Add debug log for WWN and multipath","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/0a5a387523a449060df088182f2242765b6f2436"}]}],"author":{"name":"yuval","email":"yuval@lightbitslabs.com","date":"2025-08-14 09:24:30.000000000","tz":180},"committer":{"name":"yuval","email":"yuval@lightbitslabs.com","date":"2026-01-22 08:24:57.000000000","tz":120},"subject":"LightOS: update discovery client configuration","message":"LightOS: update discovery client configuration\n\nThe LightOS connector uses hardcoded values for discovery client\nconnection parameters, preventing customization for different\ndeployment environments and containers with custom networks.\n\nExample of when this is needed:\nWe have 1 Cinder instance in a Docker container with network 10.10.10.1,\nand a discovery-service located at 10.10.10.2.\nIn the same cluster, we have a Nova instance in a container with\nnetwork 20.20.20.1, and a discovery-service at 20.20.20.2.\nI need to be able to set the network address for each instance.\n\nDefault values are set for backward compatibility.\n\nChange-Id: I19065ad683b9ff8174814d6e8c500178c885bee4\nSigned-off-by: yuval \u003cyuval@lightbitslabs.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/e7d2cd0ad3a9d37132103cacd1fb1a5d5acec52a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/e7d2cd0ad3a9d37132103cacd1fb1a5d5acec52a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
