)]}'
{"id":"openstack%2Fos-brick~975358","triplet_id":"openstack%2Fos-brick~stable%2F2025.2~Iefe9dc7a411af3e01e0333e7bfe0c2dbd926d8ae","project":"openstack/os-brick","branch":"stable/2025.2","topic":"bug/2138280","attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2026-02-12 04:26:13.000000000","reason":"Vote got outdated and was removed: Code-Review-1"},"9236":{"account":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"last_update":"2026-03-06 04:36:26.000000000","reason":"\u003cGERRIT_ACCOUNT_36725\u003e replied on the change","reason_account":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"}},"36725":{"account":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"last_update":"2026-09-08 02:07:06.000000000","reason":"\u003cGERRIT_ACCOUNT_34860\u003e replied on the change","reason_account":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"}},"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2026-03-06 04:36:26.000000000","reason":"\u003cGERRIT_ACCOUNT_36725\u003e replied on the change","reason_account":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"}},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2026-02-06 04:47:09.000000000","reason":"\u003cGERRIT_ACCOUNT_36725\u003e replied on the change","reason_account":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Iefe9dc7a411af3e01e0333e7bfe0c2dbd926d8ae","subject":"Dell PowerFlex: Fixed volume detach issue","status":"NEW","created":"2026-02-01 08:23:37.000000000","updated":"2026-09-23 10:06:29.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":13,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":974811,"cherry_pick_of_patch_set":4,"meta_rev_id":"5b2b38327f856941ccf9ebc35438b5a95ecf2bb1","_number":975358,"virtual_id_number":975358,"owner":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-03-12 00:47:36.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2026-03-16 18:17:56.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":1,"date":"2026-09-08 02:07:06.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"}],"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-01 09:23:07.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-02-01 10:24:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-01 10:31:12.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2026-02-03 15:31:33.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"CC"},{"updated":"2026-02-06 04:47:09.000000000","updated_by":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2026-02-06 04:47:09.000000000","updated_by":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2026-03-06 04:36:26.000000000","updated_by":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"},{"updated":"2026-03-06 04:36:26.000000000","updated_by":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2026-03-06 16:22:07.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2026-03-16 18:17:56.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2026-09-08 02:07:06.000000000","updated_by":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"reviewer":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"state":"REVIEWER"},{"updated":"2026-09-23 10:06:29.000000000","updated_by":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"reviewer":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"state":"CC"}],"messages":[{"id":"c2697b56903f9fd0dd965abe283722879421ac51","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2026-02-01 08:23:37.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"6c7c612d5952df6e89d1fd77578cd891dd595aaf","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-02-01 09:23:07.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* quobyteci:  os-brick_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/Iefe9dc7a411af3e01e0333e7bfe0c2dbd926d8ae .","accounts_in_message":[],"_revision_number":1},{"id":"06df4f75ebe55883bddbf720f6856a7bdd7aa28a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-01 10:24:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be4b99c0a5314ba9a9e1136e689e84f0\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7a9ee89d90174055b98e502fe6a52271 : SUCCESS in 1h 51m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f996f5236fe14924a4c4d09d5551f0ec : SUCCESS in 3m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5871e31304b745b79e02e3cade060cf1 : SUCCESS in 2m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/38021a5cda594f5f9922ce94ca473cdb : SUCCESS in 3m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/721bf12610de4215bc53f6fea8077b8f : SUCCESS in 6m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2dc97eb6e987455d92f0339b6535166b : SUCCESS in 7m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d7232c2b8df4438d9f5c6e78591ca7b7 : SUCCESS in 3m 15s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/9c848bccfbdc4930910417d5831f2818 : SUCCESS in 3m 32s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/a8f7086ba8a14585a3cbb5b001e6c671 : FAILURE in 1h 10m 53s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/3cedd335ded04f97835a003b29e557f1 : SUCCESS in 1h 59m 35s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/41e5f8d9060d4d078e531b8c83b30198 : SUCCESS in 1h 06m 27s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/69b163fe1d164d999509f73ee456af12 : SUCCESS in 2m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a800befe62fa8345f018eaa5ea9091dec6b0eb40","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-02-01 10:31:12.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/01_02_2026_10_24_25\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":1},{"id":"bfe4303f3564c88c471438dc3c98b72570884149","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2026-02-02 08:42:00.000000000","message":"Topic cherrypick-bug/2138280-o2kz6ltntcf removed","accounts_in_message":[],"_revision_number":1},{"id":"a0990e769fe5d414faaba720005e35d2e9c90a3b","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2026-02-02 08:42:09.000000000","message":"Topic set to bug/2138280","accounts_in_message":[],"_revision_number":1},{"id":"0e1ff0fb6937dda2d82b0eb6ad3b78a312d01689","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2026-02-03 15:31:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"303e3b394308724e455e11f2de428a2980c01452","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2026-02-03 15:33:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"da114074d6be7d44de975c0f4bdcc69824b19870","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2026-02-06 04:47:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e495d6cd0b17d0f316d8be4a42b2c27037ecd534","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-02-12 03:05:36.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"65d0e79335bc2a24045d15e785a092b068ba56ca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2026-02-12 04:26:13.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"6fe60d595181fb74fa39c4fbc1332513406289d2","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2026-02-12 04:26:23.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7ac9ad9df02f25d777282c53d2f5e7326b13ad11","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-02-12 05:00:31.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* quobyteci:  os-brick_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Iefe9dc7a411af3e01e0333e7bfe0c2dbd926d8ae .","accounts_in_message":[],"_revision_number":2},{"id":"759deb786df5e87b43a5859bd6c3ce9a6a32d408","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-12 08:01:19.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/036e5b515ef7484bbb321b5784d0eccc\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/35f17604dac6480e8c89894611438d46 : SUCCESS in 1h 37m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7f9cc530cee47aaa4353ecd4e200bf2 : FAILURE in 3m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ee13be82138d4463847c32cf21275b83 : SUCCESS in 2m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3edbbe2254054ae3b3a166874dc6a24a : FAILURE in 3m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1ee770b986464fcdb335c2165490aded : FAILURE in 7m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc3b5a84feac425b9deb382cceb27b56 : SUCCESS in 3m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/37e7108f1ac74fbaaa583891df1eb2b2 : SUCCESS in 4m 35s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/0135643f10674b8ba3cf71bfcb68d22b : FAILURE in 3m 42s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/81509a77aa3447a0b7d7cc8e653b62bb : SUCCESS in 1h 52m 25s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/1a0f6aba4649446bab48c8cee9ca9af3 : SUCCESS in 1h 26m 30s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/d89e77e331fa4ff591c7cdba8670ebb6 : TIMED_OUT in 3h 34m 31s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/d8a84b384b4d4226970a884c6fa1407f : SUCCESS in 2m 02s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ef38bd73636070efaf6737baae92048108f695ec","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2026-02-13 05:54:29.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ccae7865012d09200edfa273312326c57165fde8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 08:01:33.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0dfc1ec59dcc4b4f8225cba76d121c90\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f70527e1dc854e85b3027119c81c4a26 : SUCCESS in 1h 48m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2944aabf8064424ebfe54bbe2dc34da7 : FAILURE in 3m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3637c56c672a448783b64952c94881cf : SUCCESS in 3m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/04f7f953fd3e4e93b95712df08316514 : FAILURE in 2m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1266fdf9d39c4b07915ccd3ffad105c3 : FAILURE in 6m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/130831fbab6c48e3b6abc7729332f922 : SUCCESS in 5m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e8f5091afd014542ac376903909c96af : SUCCESS in 3m 14s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/16ced847bde94a2282b6e8c0a95d8362 : FAILURE in 3m 41s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/9fb5806e4d5346ecaece8f599403cccc : SUCCESS in 1h 50m 05s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/71f481278c6748ca87d1f596db4ad926 : SUCCESS in 1h 46m 45s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/0c1838c14ed04827bb13ac1914c61195 : SUCCESS in 2h 00m 21s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/82e4a88a46ec4aa5bda57ad8c9b409b1 : SUCCESS in 4m 03s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a70a9118f383cb8583e1ff5498e60b9fdb966002","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2026-02-13 20:53:26.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3ce39e28999047af549acd1e1dd19f1cef59e738","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-13 23:14:26.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c5a0f73afbb3486a9f87060ed90862b0\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/05836f27b2394717a81e5f88949979cc : SUCCESS in 1h 48m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83ffa3a6f89d45e8b1eaf231972a9f3c : FAILURE in 3m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e9bf573f0c3747cd96395a29c91c2acf : SUCCESS in 3m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f3186cbb34a54b7181685cde4f394cd1 : FAILURE in 3m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/60a94654add54ff78fd297fd66292e9e : FAILURE in 5m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/686d202e11ac41ffa6c29e9009e65d5b : SUCCESS in 7m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/07bee0d3fb784b2cbc8012ac66f012c0 : SUCCESS in 3m 51s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/dba7667bfbfc4125ae123d57185a3398 : FAILURE in 2m 44s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/935a83afc5444767a53f85c79c9ab261 : SUCCESS in 1h 44m 13s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/a10bbdd816f84c3ab9e7f98e3fa7aa4b : SUCCESS in 1h 54m 12s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/c7a1f3738317442a8d7ff97e12cf8aa7 : SUCCESS in 2h 15m 46s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/431d6d0c07f14bac943471012149a8ba : SUCCESS in 3m 27s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b2df999e76a94888470b12af62c02011e7a3c311","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2026-03-06 04:36:26.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"be80023cca48d30e32fa0d4bf37209404899f601","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2026-03-06 06:36:02.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"1b21f1ef6256c3baa7557b9756287fba952d155c","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-03-06 07:11:38.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* quobyteci:  os-brick_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/Iefe9dc7a411af3e01e0333e7bfe0c2dbd926d8ae .","accounts_in_message":[],"_revision_number":3},{"id":"d5137eb18ddeb0794f567934bb6093a34b83f675","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-06 08:40:38.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b189242cde604857b9ef25981862ada8\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/454681591376467280016c82803b2464 : SUCCESS in 1h 40m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0daba3d9d784325b8f578ebd7f83ea7 : FAILURE in 3m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d11e2191611e4e0b95dc16b8ac6cf4f4 : FAILURE in 3m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3d1705017c234079921f1cfea598a485 : FAILURE in 4m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c8e87d95add14bb6a8f3dfc9cea5af6c : FAILURE in 5m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2cd6a7e63d6b4e80812777c9d779d481 : FAILURE in 3m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f821fba44768459a825d86dcfd1909a7 : SUCCESS in 3m 45s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/6c981b0fbaa240b692c44402cdab7ead : FAILURE in 3m 12s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/189b78631f74440a9b06fd0c876ac4f1 : SUCCESS in 1h 59m 19s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/4bd377d13b38445fa86506004a4b92ad : SUCCESS in 2h 03m 12s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/e2fa759381754e1a93d798be74d3145e : SUCCESS in 57m 43s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/4f0b5f0801484440bfa39b36b8757fa3 : FAILURE in 4m 24s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"746fd2765413eaacecd0b0e2ebc54372cb43459e","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-06 16:22:07.000000000","message":"Patch Set 3:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3188/.","accounts_in_message":[],"_revision_number":3},{"id":"52d0ff7e4a833a00e858687162551da348f7495d","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-06 17:01:45.000000000","message":"Patch Set 3:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3361/.","accounts_in_message":[],"_revision_number":3},{"id":"ebb155195cf3d64d6432eafc664c0c14e642ffcd","author":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"date":"2026-03-11 22:46:44.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"590ac0cea959dade99f82065b517a6b51a5a0595","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-12 00:47:36.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d95375fa12f40d18e50fd3469f379d9\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/be22b01f43d8475ca36f5b1438f16179 : SUCCESS in 1h 13m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e87972af87a4629a380d5edf555300d : SUCCESS in 4m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/582b78427e9041cf8add15c334c83219 : SUCCESS in 5m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ac5bc231a519412f8274190afc5ab798 : SUCCESS in 2m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ebdac73e7e5f48539379219c6d17fe86 : SUCCESS in 6m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fdbf11d7b69f47a98fefd4ae0e5857e6 : SUCCESS in 7m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f3ab58338cb544b28e5b4ee524f0a6e0 : SUCCESS in 3m 57s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/8a238f785c6e4ef6a169d148c0d7a840 : SUCCESS in 3m 56s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/9e8ba47b812049cf902e6f8013956c94 : SUCCESS in 1h 51m 09s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/2b1564b905034eebb26838ae2390c292 : SUCCESS in 1h 54m 47s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/04820739e8064c2db288872a4d9e9c9b : SUCCESS in 1h 34m 17s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/eed6acbe841f4aa5b542b78ce82561a5 : SUCCESS in 3m 30s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8cb17a51ea062c0f085f4e33fbbd7e2fc84cb08f","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2026-03-16 18:17:56.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"48f40d29c8a5b735bacba9b36fe10e36f315fe73","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2026-09-08 02:07:06.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"5b2b38327f856941ccf9ebc35438b5a95ecf2bb1","author":{"_account_id":31779,"name":"Jean Pierre Roquesalane","display_name":"happystacker","email":"jeanpierre.roquesalane@dell.com","username":"happystacker"},"date":"2026-09-23 10:06:29.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"a3e1a82839c3aabe9921580a0a060b9dc519fcd8","revisions":{"c22244f7a86e547c03a94b584de3c4310515d00d":{"kind":"REWORK","_number":1,"created":"2026-02-01 08:23:37.000000000","uploader":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"ref":"refs/changes/58/975358/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/58/975358/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/58/975358/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/58/975358/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/58/975358/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/58/975358/1"}}},"commit":{"parents":[{"commit":"0aa8ad5f0aa41ee3103d52a3bc38967dfadcb0cb","subject":"Update TOX_CONSTRAINTS_FILE for stable/2025.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/0aa8ad5f0aa41ee3103d52a3bc38967dfadcb0cb"}]}],"author":{"name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","date":"2026-01-27 08:37:03.000000000","tz":0},"committer":{"name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","date":"2026-02-01 08:23:37.000000000","tz":0},"subject":"Dell PowerFlex: Fixed volume detach issue","message":"Dell PowerFlex: Fixed volume detach issue\n\nProblem:\nVolumes attached prior to the os-brick \u003c 6.13.0 and driver \u003c 3.6.0\nrelease does not store ``sdc_guid``. The detach workflow introduced\nin os-brick\u003d6.13.0 and driver\u003d3.6.0 assumes ``sdc_guid`` is present,\ncausing detach operations to error for legacy attachments.\n\nFix:\nAdds backward compatibility for detach operations on legacy volumes\nthat do not include an sdc_guid in the connector information. When\nthe sdc_guid field is missing, the detachment is now handled by\nos-brick to ensure successful unmap operations for legacy attachments.\n\nCloses-Bug: #2138280\nChange-Id: Iefe9dc7a411af3e01e0333e7bfe0c2dbd926d8ae\nSigned-off-by: Nilesh Thathagar \u003cnilesh.thathagar@dell.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/c22244f7a86e547c03a94b584de3c4310515d00d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/c22244f7a86e547c03a94b584de3c4310515d00d"}]},"branch":"refs/heads/stable/2025.2"},"b37a97b596d003046c472862bf0842d5eb1b1104":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-02-12 04:26:13.000000000","uploader":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"ref":"refs/changes/58/975358/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/58/975358/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/58/975358/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/58/975358/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/58/975358/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/58/975358/2"}}},"commit":{"parents":[{"commit":"0aa8ad5f0aa41ee3103d52a3bc38967dfadcb0cb","subject":"Update TOX_CONSTRAINTS_FILE for stable/2025.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/0aa8ad5f0aa41ee3103d52a3bc38967dfadcb0cb"}]}],"author":{"name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","date":"2026-01-27 08:37:03.000000000","tz":0},"committer":{"name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","date":"2026-02-12 04:26:13.000000000","tz":0},"subject":"Dell PowerFlex: Fixed volume detach issue","message":"Dell PowerFlex: Fixed volume detach issue\n\nProblem:\nVolumes attached prior to the os-brick \u003c 6.13.0 and driver \u003c 3.6.0\nrelease does not store ``sdc_guid``. The detach workflow introduced\nin os-brick\u003d6.13.0 and driver\u003d3.6.0 assumes ``sdc_guid`` is present,\ncausing detach operations to error for legacy attachments.\n\nFix:\nAdds backward compatibility for detach operations on legacy volumes\nthat do not include an sdc_guid in the connector information. When\nthe sdc_guid field is missing, the detachment is now handled by\nos-brick to ensure successful unmap operations for legacy attachments.\n\nCloses-Bug: #2138280\nChange-Id: Iefe9dc7a411af3e01e0333e7bfe0c2dbd926d8ae\nSigned-off-by: Nilesh Thathagar \u003cnilesh.thathagar@dell.com\u003e\n(cherry pick from commit 09bece4842d7206353a0231f7ff1b60aa1d433fb)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/b37a97b596d003046c472862bf0842d5eb1b1104"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/b37a97b596d003046c472862bf0842d5eb1b1104"}]},"branch":"refs/heads/stable/2025.2","description":"Edit commit message"},"a3e1a82839c3aabe9921580a0a060b9dc519fcd8":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-03-06 06:36:02.000000000","uploader":{"_account_id":36725,"name":"Nilesh Thathagar","display_name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","username":"NileshT"},"ref":"refs/changes/58/975358/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/58/975358/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/58/975358/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/58/975358/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/58/975358/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/58/975358/3"}}},"commit":{"parents":[{"commit":"0ce1e6177662e30eb1b04886d2224e169ebfe463","subject":"[stable-only] constrain setuptools","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/0ce1e6177662e30eb1b04886d2224e169ebfe463"}]}],"author":{"name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","date":"2026-01-27 08:37:03.000000000","tz":0},"committer":{"name":"Nilesh Thathagar","email":"nilesh.thathagar@dell.com","date":"2026-03-06 06:36:02.000000000","tz":0},"subject":"Dell PowerFlex: Fixed volume detach issue","message":"Dell PowerFlex: Fixed volume detach issue\n\nProblem:\nVolumes attached prior to the os-brick \u003c 6.13.0 and driver \u003c 3.6.0\nrelease does not store ``sdc_guid``. The detach workflow introduced\nin os-brick\u003d6.13.0 and driver\u003d3.6.0 assumes ``sdc_guid`` is present,\ncausing detach operations to error for legacy attachments.\n\nFix:\nAdds backward compatibility for detach operations on legacy volumes\nthat do not include an sdc_guid in the connector information. When\nthe sdc_guid field is missing, the detachment is now handled by\nos-brick to ensure successful unmap operations for legacy attachments.\n\nCloses-Bug: #2138280\nChange-Id: Iefe9dc7a411af3e01e0333e7bfe0c2dbd926d8ae\nSigned-off-by: Nilesh Thathagar \u003cnilesh.thathagar@dell.com\u003e\n(cherry pick from commit 09bece4842d7206353a0231f7ff1b60aa1d433fb)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/a3e1a82839c3aabe9921580a0a060b9dc519fcd8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/a3e1a82839c3aabe9921580a0a060b9dc519fcd8"}]},"branch":"refs/heads/stable/2025.2","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
