)]}'
{"id":"openstack%2Fos-brick~977795","triplet_id":"openstack%2Fos-brick~stable%2F2025.1~I1bc35bcbf603e6a3bcbef2167618b69610021bd4","project":"openstack/os-brick","branch":"stable/2025.1","topic":"fix-iscsi-session-device-scanning-for-multiple-sessions-per-portal-2025.01","attention_set":{},"removed_from_attention_set":{"10058":{"account":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"last_update":"2026-04-08 16:57:47.000000000","reason":"Change was submitted"},"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2026-04-08 16:57:47.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1bc35bcbf603e6a3bcbef2167618b69610021bd4","subject":"Fix iSCSI session device scanning for multiple sessions per portal","status":"MERGED","created":"2026-02-24 13:22:43.000000000","updated":"2026-04-08 16:59:04.000000000","submitted":"2026-04-08 16:57:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"977795-fix-iscsi-session-device-scanning-for-multiple-sessions-per-portal-2025.01","meta_rev_id":"8ba5701d75f54f9da68a38487ddbae20eef8f29a","_number":977795,"virtual_id_number":977795,"owner":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-08 16:57:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2026-04-08 14:43:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2026-04-07 16:41:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2026-04-08 14:43:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}],"reviewers":{"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-24 14:03:23.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-02-24 15:20:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-20 05:58:46.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2026-03-20 13:26:00.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2026-04-07 16:41:35.000000000","updated_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"},{"updated":"2026-04-08 14:43:06.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"}],"messages":[{"id":"d727d0be26a31f11006ccce374ed4ad1bb001486","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"date":"2026-02-24 13:22:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"08e6c213521b4bebc384fc4a0936e1ecf43a01a1","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-02-24 14:03:23.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* quobyteci:  os-brick_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I1bc35bcbf603e6a3bcbef2167618b69610021bd4 .","accounts_in_message":[],"_revision_number":1},{"id":"fb96fddfd7f8c2e6bdbf8df9012675a920d0b446","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-24 15:20:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a7c203bf8e54428980816d87f283a4ba\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/18b6e6536f9e422e91a5c966eff3173a : SUCCESS in 1h 15m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac1486ba33494621a7e38f241550a2c4 : FAILURE in 3m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3e3a6c8b950a433ea4db6cabfd8057a5 : FAILURE in 6m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fc84f80cacaa416a9f1d3b0c6c966725 : FAILURE in 2m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a232082bc2ef419388492a18313c8128 : FAILURE in 7m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/438b3a5d11f240a6829808901a00294c : SUCCESS in 2m 20s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/4798e064e48d4ad1b77e2d35cead7149 : FAILURE in 2m 36s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/ec501fff1279436cb3f1d5f5cf093333 : SUCCESS in 1h 32m 21s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/2aabf6a30a934987a66a03d195c61d93 : SUCCESS in 1h 49m 50s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/ae40168169af405e9f3ad21a74809964 : SUCCESS in 1h 55m 21s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/f8412c29a1904efa95b83a56aaa1f952 : SUCCESS in 3m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e590638d5dc29a0b17f59fc32711e45abdb21623","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"date":"2026-03-19 21:26:32.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"9fa829373f9a6d83311a563fb93248e38c6b1abf","author":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"date":"2026-03-19 21:26:55.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1d7840093eab0b3aa7305ff48f72545403f3e0d1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"date":"2026-03-19 21:29:09.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"397c4b1745cb2192fbbbac7b3ec501920cfb725b","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-03-19 22:04:28.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* quobyteci:  os-brick_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I1bc35bcbf603e6a3bcbef2167618b69610021bd4 .","accounts_in_message":[],"_revision_number":2},{"id":"474c5800bdd46cc2df248a4d53cdcab4b23ff3f0","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-03-19 22:08:25.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* quobyteci:  os-brick_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I1bc35bcbf603e6a3bcbef2167618b69610021bd4 .","accounts_in_message":[],"_revision_number":3},{"id":"bc92625ebcedeb388f436b76f7b652f1d902adcb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-19 23:24:50.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e2204934238046a18d98539ca474e2b7\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d59d9632a7464223ae89cf6991ec1510 : SUCCESS in 1h 53m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40068f697e994244b6c667abe75bfaad : SUCCESS in 5m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0af658a5a53c4f9b823ec8596f710dcf : SUCCESS in 4m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/586c814a01c04b1abfffd13386426a0a : SUCCESS in 6m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4083a5d427b140239e8c676dd2d57ad3 : SUCCESS in 7m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a95ddfbc65e6491a8d4f5d3ca5bf4724 : SUCCESS in 5m 17s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/589f9d2c69db45ee8392ccc4a045b790 : SUCCESS in 4m 15s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/4cb013679bfc419aaa6e79628d97762f : SUCCESS in 1h 46m 09s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/0cb3d6f5349e427a9bd6e3a75532049f : SUCCESS in 1h 50m 49s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/78a841fa48244a7f9e4eae46abde7476 : SUCCESS in 1h 28m 24s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/338ac3c6006249bca6db8d261e0e0366 : SUCCESS in 2m 30s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"20d11bcf8f58c802d0fc2b43707068eb3c99508b","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-20 05:58:46.000000000","message":"Patch Set 2:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3367/.","accounts_in_message":[],"_revision_number":2},{"id":"52184c34bea061898df10c2a42c8707f877d7b7b","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-20 06:06:15.000000000","message":"Patch Set 2:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3540/.","accounts_in_message":[],"_revision_number":2},{"id":"037d1e667af56e4e6bb91183cb2d2ba80452787b","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-20 06:47:25.000000000","message":"Patch Set 3:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3369/.","accounts_in_message":[],"_revision_number":3},{"id":"469fc453eca95545409a910c965931508d38985a","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-20 06:58:59.000000000","message":"Patch Set 3:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3542/.","accounts_in_message":[],"_revision_number":3},{"id":"6194e3a99518fe1bb36b47b38edf67e826097488","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2026-04-07 16:41:35.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"e8c5b835a49a8ac31a0dcd57f736217d348b80af","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-04-08 14:43:06.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1ddbec783b065796d18d957ec88b09a727c4343b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-08 14:43:29.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"a65c8a08c1e827916b1187ca12c3e1a0099e4fa0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-08 16:57:47.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b7caaf661e6d404aa5ea1eb210c59b03\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6bb847b1f8144b16a065b3b7c2478fd0 : SUCCESS in 1h 46m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ab614c40900475ab277600096921575 : SUCCESS in 3m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/98493fec7d8140418749cdeed5aa7a6d : SUCCESS in 3m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/58bf110ddff545c3810aa3c45ccebe4c : SUCCESS in 4m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9dca9d2dc195426bae939f0e58504e40 : SUCCESS in 6m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d1b7d69d0d3e41468fff3a63cc1009a3 : SUCCESS in 4m 35s\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/cccbe60f1e4d45f899b4445ad076504a : SUCCESS in 2h 12m 01s","accounts_in_message":[],"_revision_number":3},{"id":"98125f11f9c94074e4d1a6eea66b48a160454597","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-08 16:57:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"8ba5701d75f54f9da68a38487ddbae20eef8f29a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-08 16:59:04.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a20007ab0c0b42f3a2f1cfacdddca2d3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1145df0cdf064dad9aea0565fa68fdca : SUCCESS in 1m 02s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0cd6ffa900a048f88d17128b5331fb78 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"eb1854e64363ebf9f6e435b3702182c693b7bf5a","revisions":{"4d949b6c553e57016b265a66cb8ae098d4f9f0a6":{"kind":"REWORK","_number":1,"created":"2026-02-24 13:22:43.000000000","uploader":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"ref":"refs/changes/95/977795/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/95/977795/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/95/977795/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/95/977795/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/95/977795/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/95/977795/1"}}},"commit":{"parents":[{"commit":"69eb4e01ce34d8f32520b5aea681cac930ba7311","subject":"FC: Avoid calling ``lsscsi`` during disconnect","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/69eb4e01ce34d8f32520b5aea681cac930ba7311"}]}],"author":{"name":"Erlon R. Cruz","email":"erlon@canonical.com","date":"2025-07-18 13:21:22.000000000","tz":-180},"committer":{"name":"Erlon R. Cruz","email":"erlon@canonical.com","date":"2026-02-24 13:21:46.000000000","tz":-180},"subject":"Fix iSCSI session device scanning for multiple sessions per portal","message":"Fix iSCSI session device scanning for multiple sessions per portal\n\nThis commit fixes the iSCSI connector\u0027s _get_connection_devices\nmethod to properly handle multiple valid sessions per (portal,\niqn) pair. Previously, the method used a dictionary mapping\n(portal, iqn) to a single session ID, which caused device discovery\nto miss devices from additional sessions and consequently, not\ncleaning up the removed devices.\n\nThe patch fixes that by modifing _get_connection_devices to use\na list of (portal, iqn, session_id)  tuples instead of a dictionary,\nallowing multiple sessions per portal ensuring that all devices\nare discovered across all valid iSCSI sessions.\n\nCloses-bug: #2116553\nCloses-bug: #2092195\nChange-Id: I1bc35bcbf603e6a3bcbef2167618b69610021bd4\nSigned-off-by: Erlon R. Cruz \u003cerlon@canonical.com\u003e\n(cherry picked from commit eb51e427aaa3b1270f839955d369378b2b7e6f19)\n(cherry picked from commit bec5240ddf974e7f8db7852c1dbb13f7ce138fc9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/4d949b6c553e57016b265a66cb8ae098d4f9f0a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/4d949b6c553e57016b265a66cb8ae098d4f9f0a6"}]},"branch":"refs/heads/stable/2025.1"},"e3e0bd99f42f45f0b5c8034bded5f876f8dd0306":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-03-19 21:26:32.000000000","uploader":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"ref":"refs/changes/95/977795/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/95/977795/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/95/977795/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/95/977795/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/95/977795/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/95/977795/2"}}},"commit":{"parents":[{"commit":"69eb4e01ce34d8f32520b5aea681cac930ba7311","subject":"FC: Avoid calling ``lsscsi`` during disconnect","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/69eb4e01ce34d8f32520b5aea681cac930ba7311"}]}],"author":{"name":"Erlon R. Cruz","email":"erlon@canonical.com","date":"2025-07-18 13:21:22.000000000","tz":-180},"committer":{"name":"Erlon R. Cruz","email":"erlon@canonical.com","date":"2026-03-19 21:26:32.000000000","tz":0},"subject":"Fix iSCSI session device scanning for multiple sessions per portal","message":"Fix iSCSI session device scanning for multiple sessions per portal\n\nThis commit fixes the iSCSI connector\u0027s _get_connection_devices\nmethod to properly handle multiple valid sessions per (portal,\niqn) pair. Previously, the method used a dictionary mapping\n(portal, iqn) to a single session ID, which caused device discovery\nto miss devices from additional sessions and consequently, not\ncleaning up the removed devices.\n\nThe patch fixes that by modifing _get_connection_devices to use\na list of (portal, iqn, session_id)  tuples instead of a dictionary,\nallowing multiple sessions per portal ensuring that all devices\nare discovered across all valid iSCSI sessions.\n\nCloses-bug: #2116553\nCloses-bug: #2092195\nDepends-on: Iec9ab866103d96d2c512c5d0fa69dcbeec2dcad2\nChange-Id: I1bc35bcbf603e6a3bcbef2167618b69610021bd4\nSigned-off-by: Erlon R. Cruz \u003cerlon@canonical.com\u003e\n(cherry picked from commit eb51e427aaa3b1270f839955d369378b2b7e6f19)\n(cherry picked from commit bec5240ddf974e7f8db7852c1dbb13f7ce138fc9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/e3e0bd99f42f45f0b5c8034bded5f876f8dd0306"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/e3e0bd99f42f45f0b5c8034bded5f876f8dd0306"}]},"branch":"refs/heads/stable/2025.1","description":"Edit commit message"},"eb1854e64363ebf9f6e435b3702182c693b7bf5a":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-03-19 21:29:09.000000000","uploader":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"ref":"refs/changes/95/977795/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/95/977795/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/95/977795/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/95/977795/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/95/977795/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/95/977795/3"}}},"commit":{"parents":[{"commit":"cf343922fadb8a947cf4ab4ac30d6e2601a970f0","subject":"[stable-only] constrain setuptools","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/cf343922fadb8a947cf4ab4ac30d6e2601a970f0"}]}],"author":{"name":"Erlon R. Cruz","email":"erlon@canonical.com","date":"2025-07-18 13:21:22.000000000","tz":-180},"committer":{"name":"Erlon R. Cruz","email":"erlon@canonical.com","date":"2026-03-19 21:29:09.000000000","tz":0},"subject":"Fix iSCSI session device scanning for multiple sessions per portal","message":"Fix iSCSI session device scanning for multiple sessions per portal\n\nThis commit fixes the iSCSI connector\u0027s _get_connection_devices\nmethod to properly handle multiple valid sessions per (portal,\niqn) pair. Previously, the method used a dictionary mapping\n(portal, iqn) to a single session ID, which caused device discovery\nto miss devices from additional sessions and consequently, not\ncleaning up the removed devices.\n\nThe patch fixes that by modifing _get_connection_devices to use\na list of (portal, iqn, session_id)  tuples instead of a dictionary,\nallowing multiple sessions per portal ensuring that all devices\nare discovered across all valid iSCSI sessions.\n\nCloses-bug: #2116553\nCloses-bug: #2092195\nDepends-on: Iec9ab866103d96d2c512c5d0fa69dcbeec2dcad2\nChange-Id: I1bc35bcbf603e6a3bcbef2167618b69610021bd4\nSigned-off-by: Erlon R. Cruz \u003cerlon@canonical.com\u003e\n(cherry picked from commit eb51e427aaa3b1270f839955d369378b2b7e6f19)\n(cherry picked from commit bec5240ddf974e7f8db7852c1dbb13f7ce138fc9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/eb1854e64363ebf9f6e435b3702182c693b7bf5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/eb1854e64363ebf9f6e435b3702182c693b7bf5a"}]},"branch":"refs/heads/stable/2025.1","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
