)]}'
{"id":"openstack%2Fos-brick~978786","triplet_id":"openstack%2Fos-brick~stable%2F2025.1~Iec9ab866103d96d2c512c5d0fa69dcbeec2dcad2","project":"openstack/os-brick","branch":"stable/2025.1","topic":"pkg_res","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2026-03-11 19:22:17.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iec9ab866103d96d2c512c5d0fa69dcbeec2dcad2","subject":"[stable-only] constrain setuptools","status":"MERGED","created":"2026-03-03 16:34:07.000000000","updated":"2026-03-11 19:23:45.000000000","submitted":"2026-03-11 19:22:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"978786-pkg_res","meta_rev_id":"26d4f611b70905b4f8c46f7dc3b0704464e3678f","_number":978786,"virtual_id_number":978786,"owner":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-11 19:22:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"all":[{"value":2,"date":"2026-03-05 23:35:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":1,"date":"2026-03-11 14:23:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-03-05 14:08:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"all":[{"value":1,"date":"2026-03-05 23:35:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-03 17:01:06.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2026-03-03 17:14:10.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-03-03 18:41:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-05 14:08:35.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2026-03-05 23:35:45.000000000","updated_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"},{"updated":"2026-03-11 14:23:54.000000000","updated_by":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"reviewer":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"state":"REVIEWER"}],"messages":[{"id":"88778f429f769c7b8913eb417f0b71dbaa606893","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-03-03 16:34:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9f404c3db8ef7027680f54df61cfb9367297b3bf","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-03 17:01:06.000000000","message":"Patch Set 1:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3304/.","accounts_in_message":[],"_revision_number":1},{"id":"116e74ffebb584bcc99ef55594b401f92681497c","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-03 17:01:10.000000000","message":"Patch Set 1:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3131/.","accounts_in_message":[],"_revision_number":1},{"id":"f9a164bb255e5af3f9e4179836de5274f1972188","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-03-03 17:14:10.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* quobyteci:  os-brick_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Iec9ab866103d96d2c512c5d0fa69dcbeec2dcad2 .","accounts_in_message":[],"_revision_number":1},{"id":"33d676168fcccb0a2f9bd1a5e196eb3236c0eb56","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-03 18:41:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/216c6c573c1a4969950fbbf87e65c5ec\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/52ea3fd3c90746518bacc3a77734c474 : SUCCESS in 1h 42m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f639fcdbfbe4217b7d796a95d864ef7 : SUCCESS in 2m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dd11899038a3458f838e4647ab97801f : SUCCESS in 7m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4ff3cc42c9064622bfdb7cfbe5680b5d : SUCCESS in 5m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9441a835c9694ce09ac4d59d25736d43 : FAILURE in 5m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a9f27bf9d8fc4979bb96d5c084c3a0a1 : SUCCESS in 3m 35s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/97a7d9ab3a0a42eb97079cfa90662e2a : SUCCESS in 5m 48s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/9b38c9f4e77744d9b5afd5122f34f030 : SUCCESS in 1h 59m 21s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/397ccfa150f94e6cbe413325505f52e7 : SUCCESS in 1h 27m 15s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/02ca7982df7242ff88bebd3f32b6c27d : SUCCESS in 2h 00m 54s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/19f719d6951b4d05a50fa015817362c5 : SUCCESS in 3m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"62578e2ee971c465d28b9e48aac530fbba0577e0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-03-04 18:09:11.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"b0921a32185a55cb6e9a911b92bc16392e3119b9","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-04 19:41:53.000000000","message":"Patch Set 2:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3324/.","accounts_in_message":[],"_revision_number":2},{"id":"bd484134401ce529200db77fe92c7a2a1d5473d5","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-04 19:46:59.000000000","message":"Patch Set 2:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3151/.","accounts_in_message":[],"_revision_number":2},{"id":"2042dce7be835460241fc67927826fcf7521d0a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-04 20:14:49.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b97a230a0a1f4466b38043899b7321f6\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b7f2854eb15b4ccfb2ec3d23b00d32ff : SUCCESS in 1h 52m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a138703b60640488fb942e3333b6d3e : SUCCESS in 2m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/67287e226ae748019eb6314c70fedc68 : SUCCESS in 3m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/15c5309f0f614851a0e34a9c7806cbe8 : SUCCESS in 4m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b414d15dfcab40749c9fe707a6d32990 : FAILURE in 10m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/be9a689d96444a5c9f9ad67e606bf6c0 : SUCCESS in 4m 02s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/4fea6f33034c42ea8f01a15e42ae8cce : SUCCESS in 3m 38s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/001c6a2315594d629c1f54e0283caf63 : SUCCESS in 57m 43s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/90201eda1bd14ce89e71068e58c9850a : SUCCESS in 1h 52m 40s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/dd009f0616d84a3bb80636a98040ead7 : POST_FAILURE in 1h 44m 38s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/9e4d0d9b68e14359ab0abfefb01a9f02 : SUCCESS in 2m 58s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"56e58e4fc6da058bcf88a151c7fc15e22e26d632","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-03-04 22:50:54.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0ee38f8525ecc78e14119c119636dee7320546a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-03-05 00:59:15.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"f4e48c5fa303110d92d3a26598bf0931c5764b02","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-03-05 01:09:53.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"ea265f1e6f04b10e37cb0422320167bfe0ec14b0","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-03-05 01:25:36.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* quobyteci:  os-brick_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/Iec9ab866103d96d2c512c5d0fa69dcbeec2dcad2 .","accounts_in_message":[],"_revision_number":3},{"id":"64458ff5c2164d6423a4c4872dca1e1dd953e906","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-05 01:29:48.000000000","message":"Patch Set 3:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3327/.","accounts_in_message":[],"_revision_number":3},{"id":"fd6f55cf48350356f8e42af883f4ffa03ce59e39","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-05 01:30:11.000000000","message":"Patch Set 3:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3154/.","accounts_in_message":[],"_revision_number":3},{"id":"47b1f50a1cd24e305a12558742031cecfff8c202","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-05 03:07:59.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/911edf5cab1e48d3bbec56457f13fbc3\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2df19616b8a34cc6802e429392fa7540 : SUCCESS in 1h 34m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef9693293f974913b8fc757a8d0afa50 : SUCCESS in 4m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fa7e2bc2aa2342ca8af5eb132cc7fd57 : SUCCESS in 3m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6853d4c9660a49378ef4aa5b6f99b8be : SUCCESS in 5m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9cbd7543f61842d7b503846bf5114f7b : SUCCESS in 5m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0dce0cbca2ee459f808e2a4e98631873 : SUCCESS in 3m 43s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/f50e94e3e5524b4aa0881a9ba08582fa : SUCCESS in 3m 14s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/2b1a77e968bf4caebf3ac7c1991e0aa1 : SUCCESS in 1h 47m 28s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/0fbc26eb74094deead6bdadbf94718b5 : SUCCESS in 1h 51m 18s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/6269505344bf4da7bc83719279e58d0a : SUCCESS in 1h 59m 23s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/20efd7da70d042f88b893b9283b82234 : SUCCESS in 2m 24s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"63326de76d55259bcf66413851029ca00c1e8e77","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2026-03-05 14:08:35.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1caed2ad321e69cdc9a78af038880905e6c00fee","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2026-03-05 23:35:45.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"d5c7e29d12a68bb4ac717789a92099659f71ab8d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-05 23:36:12.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"eae2d2d02aa662d1a52bb198f98069c0b6b9bf97","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-06 01:23:20.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bd625520a7a742dfabbe20e33bd515aa\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/30b05ce923474013b00bf5c565528a17 : SUCCESS in 1h 33m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9ac45efa90f4aea8ae733258acb5b3c : FAILURE in 3m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cbbb7283bd44406ca0ab1b0cdb46c069 : SUCCESS in 5m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9f2d464579e74fdeb4017ba0c95719da : FAILURE in 4m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e8f3aa86590410e9ed922f5dcdc32e5 : FAILURE in 7m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/785a2196741f43d6b3a9968a3c3a1a06 : SUCCESS in 3m 41s\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/b5e08ad4938b47e2ac5df4202142208b : SUCCESS in 1h 40m 41s","accounts_in_message":[],"_revision_number":3},{"id":"08cd837a836e5f80d409d29ecaa333c82bd49f32","author":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"},"date":"2026-03-11 14:23:54.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7fc4e503a6f75e87ae08dae336167a1ab05c0e93","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-11 16:54:07.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5afc4770484044989c8499fe55ffc747\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/806f217b0eec4d2e92545d9aad888904 : SUCCESS in 1h 51m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0bb5d474030445389947fe2c9cf227af : SUCCESS in 3m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0dbd03704a69449292bad5846bcae8a1 : SUCCESS in 4m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4380737fb5bd479992d17c45e411d2b1 : SUCCESS in 3m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d0830bbba4b4960a9cab1af7d8c827f : SUCCESS in 7m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/473bc55539cc443fbf565e942a7559f0 : SUCCESS in 3m 52s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/544fefde1a164eecabc15822e6556379 : SUCCESS in 5m 43s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/1537742a62ad464da17a1c9f5f93007b : SUCCESS in 2h 00m 56s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/28d05c443d4f49b5b55b4af7ff9753d2 : SUCCESS in 2h 27m 38s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/09b6997fcc6847c19e7183a77be0019c : SUCCESS in 2h 14m 44s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/e64386254da747a1a3223cb44e4d4720 : SUCCESS in 5m 57s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"785da3d898dfa84abe81a449ea28dee8d9352e36","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-11 16:54:46.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"0da51c46e6f078427d8bd544844a9f346239cf36","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-11 19:22:16.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/46f37e90636640fbae752ab9df2d45b5\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1ae10e359ef547a6a3496e7e499e9809 : SUCCESS in 2h 16m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4f837fe0c444d8c86cea167032ee1f9 : SUCCESS in 3m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/35fe52093c8c407caaa9556b368e6cdc : SUCCESS in 4m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/dc9f3243c4ae410a96f44bb4a87ace2b : SUCCESS in 4m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9b2585c9f9f449fb8830fbf50085a1b : SUCCESS in 7m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3544237fe1ec43ae8eb5b23f25115e28 : SUCCESS in 3m 39s\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/46fe1ba65c6247f498f8999d60fb18eb : SUCCESS in 1h 36m 04s","accounts_in_message":[],"_revision_number":3},{"id":"0beb87ad0144259cc84773673017346c326ad9f7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-11 19:22:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"26d4f611b70905b4f8c46f7dc3b0704464e3678f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-11 19:23:45.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b7cbe4b6f773417693f96ebb9a29f652\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea063419a2984c0699509ba7f9dcc656 : SUCCESS in 46s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7d9abf492755477aa5f76b26066e5d52 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"cf343922fadb8a947cf4ab4ac30d6e2601a970f0","revisions":{"9ba11c7bd5d78798c51e1570097548e09e406bd8":{"kind":"REWORK","_number":1,"created":"2026-03-03 16:34:07.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/86/978786/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/86/978786/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/86/978786/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/86/978786/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/86/978786/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/86/978786/1"}}},"commit":{"parents":[{"commit":"69eb4e01ce34d8f32520b5aea681cac930ba7311","subject":"FC: Avoid calling ``lsscsi`` during disconnect","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/69eb4e01ce34d8f32520b5aea681cac930ba7311"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2026-02-27 14:53:34.000000000","tz":-300},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2026-03-03 16:28:24.000000000","tz":-300},"subject":"[stable-only] constrain setuptools","message":"[stable-only] constrain setuptools\n\nBecause the windows support in current stable branches depends on\nthe no-longer-maintained os-win library, we need to constrain\nsetuptools to a version that contains pkg_resources.\n\nThis also allows us to continue to use flake8-logging-format in\nthis branch.\n\nChange-Id: Iec9ab866103d96d2c512c5d0fa69dcbeec2dcad2\nSigned-off-by: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\n(cherry picked from commit 0ce1e6177662e30eb1b04886d2224e169ebfe463)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/9ba11c7bd5d78798c51e1570097548e09e406bd8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/9ba11c7bd5d78798c51e1570097548e09e406bd8"}]},"branch":"refs/heads/stable/2025.1"},"3d96b7ac159767afa77409932abea1711be0bffe":{"kind":"REWORK","_number":2,"created":"2026-03-04 18:09:11.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/86/978786/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/86/978786/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/86/978786/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/86/978786/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/86/978786/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/86/978786/2"}}},"commit":{"parents":[{"commit":"69eb4e01ce34d8f32520b5aea681cac930ba7311","subject":"FC: Avoid calling ``lsscsi`` during disconnect","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/69eb4e01ce34d8f32520b5aea681cac930ba7311"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2026-02-27 14:53:34.000000000","tz":-300},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2026-03-04 18:08:25.000000000","tz":-300},"subject":"[stable-only] constrain setuptools","message":"[stable-only] constrain setuptools\n\nBecause the windows support in current stable branches depends on\nthe no-longer-maintained os-win library, we need to constrain\nsetuptools to a version that contains pkg_resources.\n\nThis also allows us to continue to use flake8-logging-format in\nthis branch.\n\nChange-Id: Iec9ab866103d96d2c512c5d0fa69dcbeec2dcad2\nSigned-off-by: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\n(cherry picked from commit 0ce1e6177662e30eb1b04886d2224e169ebfe463)\nChanges:\n  - also constrain setuptools version in the docs env\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/3d96b7ac159767afa77409932abea1711be0bffe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/3d96b7ac159767afa77409932abea1711be0bffe"}]},"branch":"refs/heads/stable/2025.1"},"cf343922fadb8a947cf4ab4ac30d6e2601a970f0":{"kind":"REWORK","_number":3,"created":"2026-03-05 00:59:15.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/86/978786/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/86/978786/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/86/978786/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/86/978786/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/86/978786/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/86/978786/3"}}},"commit":{"parents":[{"commit":"69eb4e01ce34d8f32520b5aea681cac930ba7311","subject":"FC: Avoid calling ``lsscsi`` during disconnect","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/69eb4e01ce34d8f32520b5aea681cac930ba7311"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2026-02-27 14:53:34.000000000","tz":-300},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2026-03-05 00:56:30.000000000","tz":-300},"subject":"[stable-only] constrain setuptools","message":"[stable-only] constrain setuptools\n\nBecause the windows support in current stable branches depends on\nthe no-longer-maintained os-win library, we need to constrain\nsetuptools to a version that contains pkg_resources.\n\nThis also allows us to continue to use flake8-logging-format in\nthis branch.\n\nChange-Id: Iec9ab866103d96d2c512c5d0fa69dcbeec2dcad2\nSigned-off-by: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\n(cherry picked from commit 0ce1e6177662e30eb1b04886d2224e169ebfe463)\nChanges:\n  - also constrain setuptools version in the docs env\n  - also constrain oslo.utils version in the docs env\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/cf343922fadb8a947cf4ab4ac30d6e2601a970f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/cf343922fadb8a947cf4ab4ac30d6e2601a970f0"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":10058,"name":"Erlon R. Cruz","email":"erlon.rodrigues.cruz@canonical.com","username":"sombrafam"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
