)]}'
{"id":"openstack%2Fos-brick~978787","triplet_id":"openstack%2Fos-brick~stable%2F2024.2~Iec9ab866103d96d2c512c5d0fa69dcbeec2dcad2","project":"openstack/os-brick","branch":"stable/2024.2","topic":"pkg_res","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2026-03-12 16:36:16.000000000","reason":"\u003cGERRIT_ACCOUNT_5314\u003e replied on the change","reason_account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2026-03-12 20:40:15.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iec9ab866103d96d2c512c5d0fa69dcbeec2dcad2","subject":"[stable-only] constrain setuptools","status":"MERGED","created":"2026-03-03 16:36:10.000000000","updated":"2026-03-12 20:42:04.000000000","submitted":"2026-03-12 20:40:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"978787-pkg_res","meta_rev_id":"f8e973ecec5c20a6f924f0b29992f30fac2b9893","_number":978787,"virtual_id_number":978787,"owner":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-12 20:40:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"all":[{"value":2,"date":"2026-03-05 23:35:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-03-05 14:18:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"all":[{"value":1,"date":"2026-03-05 23:35:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"}],"REVIEWER":[{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-03 17:16:44.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-03-03 17:22:38.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2026-03-03 18:29:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-05 14:18:57.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2026-03-05 23:35:29.000000000","updated_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"}],"messages":[{"id":"e4c899e6d81c6182c3922b6458a9499e033ccb55","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-03-03 16:36:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e8886131443022a1479ccb85dd253850f90283aa","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-03-03 17:16:44.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* quobyteci:  os-brick_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Iec9ab866103d96d2c512c5d0fa69dcbeec2dcad2 .","accounts_in_message":[],"_revision_number":1},{"id":"07682db3450be6d91f057fbe2e1a571499245a20","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-03 17:22:38.000000000","message":"Patch Set 1:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3132/.","accounts_in_message":[],"_revision_number":1},{"id":"c96864b5ce67740e45dad3907c7954a8f474c7d9","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-03 17:22:54.000000000","message":"Patch Set 1:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3305/.","accounts_in_message":[],"_revision_number":1},{"id":"077ceab5bfb7e13c7e439805009abb9b90054c40","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-03 18:29:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0393f3a1091743a8a29970ed23d93eb6\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a2ab85268348497aa1234bdf1de450e9 : SUCCESS in 1h 42m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68d3d3fa178047968998916074d4e21e : SUCCESS in 2m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d3f12c4ebe594f998c7ca0ad6461a0f2 : SUCCESS in 4m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8cfa10b80de1413c86ee2b10ce3f0cac : SUCCESS in 4m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9ee6968dc2da46cea16a843d8beb7ec6 : SUCCESS in 4m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19a98af589c846cd94c9833a43e5553a : FAILURE in 6m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0fbd2b0fad4045e98c8d83fe328b03f9 : SUCCESS in 4m 52s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/83fd668c98d743aab72b392c38fd9142 : SUCCESS in 2m 22s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/60fa372ff906456fb98e29fcdde28b1f : SUCCESS in 1h 03m 20s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/2273a74e8b964011ab48f8a31747defb : SUCCESS in 1h 38m 04s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/6c34f36774714879b87f47cdc7a47ae8 : SUCCESS in 1h 44m 09s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/98b5d3159a7543bfaa0c71e1b6247705 : FAILURE in 3m 26s (non-voting)\n\nWarning:\n  Comments left for invalid file os_brick/initiator/connectors/iscsi.py","accounts_in_message":[],"_revision_number":1},{"id":"ab38cf63df4de1357719281a395c1307425b4c2f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-03-05 01:24:12.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"d02150d7fe1951091fb62f23679d83e57fcb9e72","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-05 01:49:45.000000000","message":"Patch Set 2:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3328/.","accounts_in_message":[],"_revision_number":2},{"id":"64623015b7488430d8de637f1d840859c06d5ddb","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-05 01:50:15.000000000","message":"Patch Set 2:\n\nBuild failed (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3155/.","accounts_in_message":[],"_revision_number":2},{"id":"f6d05925d6a78f4b7c5e8d6de6357a43ea2b3ef8","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-03-05 01:50:37.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* quobyteci:  os-brick_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/Iec9ab866103d96d2c512c5d0fa69dcbeec2dcad2 .","accounts_in_message":[],"_revision_number":2},{"id":"911f8bd99587d6ba76f66362f71ad4311361e854","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-05 03:29:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/95265b07a6c841239c58896d68c4b745\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7386e444484e4b4da28bdb77eff1cec1 : SUCCESS in 1h 42m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7aa17432f44940c3b49f72fa436fbc64 : SUCCESS in 2m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5e4622b4d743418f969ea904e152eb49 : SUCCESS in 4m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e0096771131d43e99e1a440eb87b2929 : SUCCESS in 3m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2224f5c732d647adaddf1479e772c9f7 : SUCCESS in 2m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bb540645b25468e8fa73e9886f898a5 : SUCCESS in 5m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c9835fc0562949899f192da03add8a7c : SUCCESS in 3m 42s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/f08cadcb8d9745d1afb43d109f9db83b : SUCCESS in 6m 17s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/8fd5df358e754764b681ce8c531f7340 : SUCCESS in 1h 43m 54s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/1043654c67ce48a292b88910861f1558 : SUCCESS in 1h 18m 28s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/2b45b76dc0cf449f9c8afa2f67c1ff46 : SUCCESS in 2h 04m 42s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/777e44fcfe8b40a7ab35a653550a305e : FAILURE in 3m 13s (non-voting)\n\nWarning:\n  Comments left for invalid file os_brick/initiator/connectors/iscsi.py","accounts_in_message":[],"_revision_number":2},{"id":"125132262e8c14b93fdd6d6946165b5699d08ace","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2026-03-05 14:18:57.000000000","message":"Patch Set 2: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"3adf9fd61437c967c2ea835007c55935fc80e394","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-03-05 15:44:28.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e64e4d665372eb6b35c0f553b127c7ca7df38dd3","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2026-03-05 23:35:29.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b3e71c9aa58a89f14a3e1e85843ae0ffd22a81e7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-05 23:35:51.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a757d50e9b497552f1912e63de32c5117a41403b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-06 01:37:06.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d436759c5567480cbccf2e435337a87b\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/91798ab561464cdeaacbae6d222ddba4 : SUCCESS in 49m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8142c60f286e44c6a2b0a0a3a819782d : FAILURE in 2m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c5bc925fef34491ea4c88bcec0d216de : SUCCESS in 3m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e3d4d28f9580465a8efd89e822b56d88 : FAILURE in 3m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/126fa5a515a64a19bef73c90331a253b : FAILURE in 3m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/48ba9644ed254d8eb4ef0bd3d69a8d6b : SUCCESS in 10m 17s\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/916ba107e10b47eebd3aa5c86d0d3402 : SUCCESS in 1h 39m 02s","accounts_in_message":[],"_revision_number":2},{"id":"6ee04995d35557a79cef204bc83c428e836e4b05","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-03-12 16:36:16.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"91b1da90f3d2fcb90f74663a145f8a1b883aaaa0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-12 18:52:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7567f9e0b1b842eea9a45093e0823dfb\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a8e9a573441b496299caa0d7b66f2485 : SUCCESS in 1h 31m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76fff6ec26de4b7980e9b5a8b2f04b42 : SUCCESS in 2m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/817801559bda40c593b2658f9f0ee90b : SUCCESS in 3m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/acfc40f427c540d3ae3cdbf84beb01de : SUCCESS in 4m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b09c973755cf489da096dbc9b6c42d19 : SUCCESS in 4m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29b7a3e97581449f8702e51ea87eabb8 : SUCCESS in 6m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f07b92b5bb0741c1a08f9b0d18ceb8bb : SUCCESS in 3m 54s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/3c95d5b72eaf4590a6bfafdce22b1a0e : SUCCESS in 4m 44s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/ac8b04aac9a04773bad6d7f5b52012fe : SUCCESS in 1h 46m 39s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/82bd0880d0f444d58c257304468c566d : SUCCESS in 1h 37m 05s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/7057b7124ce2493da5e37719e9b2fa66 : SUCCESS in 1h 57m 58s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/04145264b6194ad286ed6616204b04db : FAILURE in 3m 41s (non-voting)\n\nWarning:\n  Comments left for invalid file os_brick/initiator/connectors/iscsi.py","accounts_in_message":[],"_revision_number":2},{"id":"5abadd21237bf42d1d62d75da83398ff9b80477d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-12 18:52:38.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c4c51c8c850172df6317bb3ac82c1a45349e506f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-12 20:40:15.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3abb687c03644b99abffc0c8475c2c60\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a07070eadd09434787fed959d53e9f8d : SUCCESS in 1h 13m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c61d023cd914752b05b6c720baad8b1 : SUCCESS in 3m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/25841f91231c43cb8ff60b61401fc620 : SUCCESS in 5m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fe2336307d65486eb2758b80c2d75c9b : SUCCESS in 3m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63c3b2427e354b0c95ebb4248bfdc6cb : SUCCESS in 5m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/45fa59246e0c45c69f7b49b4b2ed0b97 : SUCCESS in 2m 34s\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/5f9fc6158a6e46418a9b7216a003a220 : SUCCESS in 1h 44m 08s","accounts_in_message":[],"_revision_number":2},{"id":"2ff5c1742c3ab38d28cd2b5767bb4e48352e0850","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-12 20:40:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"f8e973ecec5c20a6f924f0b29992f30fac2b9893","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-12 20:42:04.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d0c8e6b8d6e64ce6bac68eaecf455276\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9bb2dbac60f4c469f8843f51d12120d : SUCCESS in 1m 39s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cc1099dd8baf497d864486e0328e829e : SUCCESS in 56s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d9e4edce694db499f024cf45a6ac741229f88fc7","revisions":{"76b21d83ec1d5fc3551c2c8b5b6b62ea9d47cba8":{"kind":"REWORK","_number":1,"created":"2026-03-03 16:36:10.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/87/978787/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/87/978787/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/87/978787/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/87/978787/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/87/978787/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/87/978787/1"}}},"commit":{"parents":[{"commit":"557817b9b064a271eafd015e6c3a61f6bcbc44c0","subject":"FC: Fix multipath wait issue with slow connections","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/557817b9b064a271eafd015e6c3a61f6bcbc44c0"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2026-02-27 14:53:34.000000000","tz":-300},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2026-03-03 16:34:59.000000000","tz":-300},"subject":"[stable-only] constrain setuptools","message":"[stable-only] constrain setuptools\n\nBecause the windows support in current stable branches depends on\nthe no-longer-maintained os-win library, we need to constrain\nsetuptools to a version that contains pkg_resources.\n\nThis also allows us to continue to use flake8-logging-format in\nthis branch.\n\nChange-Id: Iec9ab866103d96d2c512c5d0fa69dcbeec2dcad2\nSigned-off-by: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\n(cherry picked from commit 0ce1e6177662e30eb1b04886d2224e169ebfe463)\n(cherry picked from commit 9ba11c7bd5d78798c51e1570097548e09e406bd8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/76b21d83ec1d5fc3551c2c8b5b6b62ea9d47cba8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/76b21d83ec1d5fc3551c2c8b5b6b62ea9d47cba8"}]},"branch":"refs/heads/stable/2024.2"},"d9e4edce694db499f024cf45a6ac741229f88fc7":{"kind":"REWORK","_number":2,"created":"2026-03-05 01:24:12.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/87/978787/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/87/978787/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/87/978787/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/87/978787/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/87/978787/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/87/978787/2"}}},"commit":{"parents":[{"commit":"557817b9b064a271eafd015e6c3a61f6bcbc44c0","subject":"FC: Fix multipath wait issue with slow connections","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/557817b9b064a271eafd015e6c3a61f6bcbc44c0"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2026-02-27 14:53:34.000000000","tz":-300},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2026-03-05 01:23:55.000000000","tz":-300},"subject":"[stable-only] constrain setuptools","message":"[stable-only] constrain setuptools\n\nBecause the windows support in current stable branches depends on\nthe no-longer-maintained os-win library, we need to constrain\nsetuptools to a version that contains pkg_resources.\n\nThis also allows us to continue to use flake8-logging-format in\nthis branch.\n\nChange-Id: Iec9ab866103d96d2c512c5d0fa69dcbeec2dcad2\nSigned-off-by: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\n(cherry picked from commit 0ce1e6177662e30eb1b04886d2224e169ebfe463)\nChanges:\n  - also constrain setuptools version in the docs env\n  - also constrain oslo.utils version in the docs env\n(cherry picked from commit cf343922fadb8a947cf4ab4ac30d6e2601a970f0)\nChanges:\n  - made basepython for docs py310 (default on ubuntu-jammy)\n  - oslo.utils constrained to dalmatian versions\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/d9e4edce694db499f024cf45a6ac741229f88fc7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/d9e4edce694db499f024cf45a6ac741229f88fc7"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
