)]}'
{"id":"openstack%2Fos-brick~978984","triplet_id":"openstack%2Fos-brick~master~Id856e2e5cab48984ee5a51c556151c368e270b24","project":"openstack/os-brick","branch":"master","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2026-03-05 16:30:23.000000000","reason":"\u003cGERRIT_ACCOUNT_5314\u003e replied on the change","reason_account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},"34598":{"account":{"_account_id":34598,"name":"Manojkatari","email":"mkatari@redhat.com","username":"mkatari"},"last_update":"2026-03-05 18:16:20.000000000","reason":"Change was submitted"},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2026-03-05 11:30:08.000000000","reason":"\u003cGERRIT_ACCOUNT_27615\u003e replied on the change","reason_account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}}},"hashtags":[],"change_id":"Id856e2e5cab48984ee5a51c556151c368e270b24","subject":"UT: Fix test_connect_multipath_volume_no_wwid mock call extraction","status":"MERGED","created":"2026-03-05 09:24:08.000000000","updated":"2026-03-05 18:17:17.000000000","submitted":"2026-03-05 18:16:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"978984","meta_rev_id":"e1d9c9080bc82ca7e34677789e9b4ade2d612a61","_number":978984,"virtual_id_number":978984,"owner":{"_account_id":34598,"name":"Manojkatari","email":"mkatari@redhat.com","username":"mkatari"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-05 18:16:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2026-03-05 16:30:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-03-05 11:30:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2026-03-05 16:30:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-05 09:28:44.000000000","updated_by":{"_account_id":34598,"name":"Manojkatari","email":"mkatari@redhat.com","username":"mkatari"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2026-03-05 09:31:04.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-03-05 09:50:37.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-03-05 11:31:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-05 11:37:47.000000000","updated_by":{"_account_id":34598,"name":"Manojkatari","email":"mkatari@redhat.com","username":"mkatari"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2026-03-05 13:43:37.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"}],"messages":[{"id":"6211dbb2d61eee2e420bca4715f64a9c99ca5f02","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34598,"name":"Manojkatari","email":"mkatari@redhat.com","username":"mkatari"},"date":"2026-03-05 09:24:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3cc416bad11fe8d43914aba9e9b8a479774987b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34598,"name":"Manojkatari","email":"mkatari@redhat.com","username":"mkatari"},"date":"2026-03-05 09:26:58.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"da94be314918c03ec64eb490759b2549a5a64b7e","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-03-05 09:31:04.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- cinder-storpool-tempest https://spfactory.storpool.com/zuul/t/local/build/e65d7de56ee4485c87be6235e47b86f6 : RETRY_LIMIT in 30s","accounts_in_message":[],"_revision_number":2},{"id":"17b834cb8c2bfd54a00dc6fc0c44de1a6b7eac82","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2026-03-05 09:38:02.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"fa0c660f98ff446cb6f580c39fca03e9d1d0e5d2","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-03-05 09:50:37.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* quobyteci:  os-brick_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/Id856e2e5cab48984ee5a51c556151c368e270b24 .","accounts_in_message":[],"_revision_number":1},{"id":"85142123cefe04b75a753e8b472c0c8f373b9168","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-03-05 09:53:20.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* quobyteci:  os-brick_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/Id856e2e5cab48984ee5a51c556151c368e270b24 .","accounts_in_message":[],"_revision_number":2},{"id":"a726b08771b74aef125fe236cb6ac0c3eb4643cc","author":{"_account_id":34598,"name":"Manojkatari","email":"mkatari@redhat.com","username":"mkatari"},"date":"2026-03-05 11:07:31.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"21f4150b3764c868316e92a3804789bdb2a1e829","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2026-03-05 11:30:08.000000000","message":"Patch Set 2: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"5eaed2d94bb0645810b8da058a7f10a4744a0fd2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-05 11:31:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6792d653cccc4f2993907168c6a04e14\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dc6f5b32b6dd471bbcd3c5d2eb4b20c6 : SUCCESS in 1h 51m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/033d5e2f9f6441ffa4dc6f1e0bc5d0df : SUCCESS in 3m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f29ee0f7cd1145fab763cf31ba5ad8cd : SUCCESS in 3m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8673640a6d424d12a78b7d65324c291a : SUCCESS in 3m 29s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b77463f619da4267bc1a6365ef68d7e7 : SUCCESS in 5m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea499c499c0d4db8827d5facadccd0d5 : SUCCESS in 7m 40s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/2562c0adf59a409babb38699b20c157e : SUCCESS in 4m 00s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/3740894f81fa4a68a0eaa05eb3c09e61 : SUCCESS in 1h 38m 23s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/659590a31a1c4256aa8a084c8ea19449 : SUCCESS in 1h 39m 59s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/65b3c5f152564155b25bc7089aedc318 : SUCCESS in 2h 02m 25s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/d18e8bffa3d146d78f7c33f690cded63 : SUCCESS in 3m 01s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8efea9e011b582dc72000cd53da608df93a32f6f","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-05 13:43:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3165/.","accounts_in_message":[],"_revision_number":1},{"id":"31be1e899657fa28ebfcf9d528e87d6783d89326","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-05 14:26:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3338/.","accounts_in_message":[],"_revision_number":1},{"id":"3181f2045fd0398460eac192019de38d964b7357","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-05 14:55:36.000000000","message":"Patch Set 2:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3166/.","accounts_in_message":[],"_revision_number":2},{"id":"a8285dcef58e72a8b63e1afacc4e548eb86dafc4","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-03-05 15:35:51.000000000","message":"Patch Set 2:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/3339/.","accounts_in_message":[],"_revision_number":2},{"id":"8052df442381421e74ace464efc0bc47e3edb059","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-03-05 16:30:23.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"fb17305bfaf8c90fcab04ee88dd532be2e3c446f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-05 16:31:00.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3aae084da9d4ac1895be7e4f164c9006d99c2694","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-05 18:16:20.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d3392413349437989b0f511572795d1\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d656c26795e04b8eb3fe8848e6b6bb44 : SUCCESS in 1h 00m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc0ae638e1b64cde9f91869880f6f00c : SUCCESS in 2m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f05d80ca611b45b1b72cd106b41cb150 : SUCCESS in 3m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/15bde203aca64591ac3f16dc55663535 : SUCCESS in 3m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9a35a5efe504135a85bbf40e47e2228 : SUCCESS in 6m 53s\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/8e657374e6504733802bb1dea67c8954 : SUCCESS in 1h 44m 39s","accounts_in_message":[],"_revision_number":2},{"id":"31589583ebef209e7b9b3fe2fb49b846d07fd236","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-05 18:16:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"e1d9c9080bc82ca7e34677789e9b4ade2d612a61","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-05 18:17:17.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/44a19405ad81480ba845947f88dd1810\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d95c41cbccc4c35af8b75b81919dbd9 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b184ddcf916c5228de423c772600c73087b272a7","revisions":{"f94d563aba25901b70ebbba7ef4cab891a33d1b1":{"kind":"REWORK","_number":1,"created":"2026-03-05 09:24:08.000000000","uploader":{"_account_id":34598,"name":"Manojkatari","email":"mkatari@redhat.com","username":"mkatari"},"ref":"refs/changes/84/978984/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/84/978984/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/84/978984/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/84/978984/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/84/978984/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/84/978984/1"}}},"commit":{"parents":[{"commit":"0cfb85ad7e9c3c39326507eb1bc8fb19f2f01acc","subject":"Update master for stable/2026.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/0cfb85ad7e9c3c39326507eb1bc8fb19f2f01acc"}]}],"author":{"name":"mkatari","email":"mkatari@redhat.com","date":"2026-03-05 09:10:53.000000000","tz":330},"committer":{"name":"mkatari","email":"mkatari@redhat.com","date":"2026-03-05 09:23:20.000000000","tz":330},"subject":"UT: Fix test_connect_multipath_volume_no_wwid mock call extraction","message":"UT: Fix test_connect_multipath_volume_no_wwid mock call extraction\n\nUse call_args_list to extract mpath values instead of _mock_mock_calls\nto avoid issues with internal mock method calls like __str__().\n\nChange-Id: Id856e2e5cab48984ee5a51c556151c368e270b24\nSigned-off-by: mkatari \u003cmkatari@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/f94d563aba25901b70ebbba7ef4cab891a33d1b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/f94d563aba25901b70ebbba7ef4cab891a33d1b1"}]},"branch":"refs/heads/master"},"b184ddcf916c5228de423c772600c73087b272a7":{"kind":"REWORK","_number":2,"created":"2026-03-05 09:26:58.000000000","uploader":{"_account_id":34598,"name":"Manojkatari","email":"mkatari@redhat.com","username":"mkatari"},"ref":"refs/changes/84/978984/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/84/978984/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/84/978984/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/84/978984/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/84/978984/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/84/978984/2"}}},"commit":{"parents":[{"commit":"0cfb85ad7e9c3c39326507eb1bc8fb19f2f01acc","subject":"Update master for stable/2026.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/0cfb85ad7e9c3c39326507eb1bc8fb19f2f01acc"}]}],"author":{"name":"mkatari","email":"mkatari@redhat.com","date":"2026-03-05 09:10:53.000000000","tz":330},"committer":{"name":"mkatari","email":"mkatari@redhat.com","date":"2026-03-05 09:26:30.000000000","tz":330},"subject":"UT: Fix test_connect_multipath_volume_no_wwid mock call extraction","message":"UT: Fix test_connect_multipath_volume_no_wwid mock call extraction\n\nUse call_args_list to extract mpath values instead of _mock_mock_calls\nto avoid issues with internal mock method calls like __str__().\n\nChange-Id: Id856e2e5cab48984ee5a51c556151c368e270b24\nSigned-off-by: mkatari \u003cmkatari@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/b184ddcf916c5228de423c772600c73087b272a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/b184ddcf916c5228de423c772600c73087b272a7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
