)]}'
{"id":"openstack%2Fos-brick~986945","triplet_id":"openstack%2Fos-brick~master~I3c93d3a7e8eedcffadc206d8c25fb1af1b8888ea","project":"openstack/os-brick","branch":"master","topic":"hibiscus-pti","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2026-05-06 19:36:25.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3c93d3a7e8eedcffadc206d8c25fb1af1b8888ea","subject":"Update python tested versions for hibiscus","status":"MERGED","created":"2026-05-01 18:31:27.000000000","updated":"2026-05-06 19:38:34.000000000","submitted":"2026-05-06 19:36:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"986945-hibiscus-pti","meta_rev_id":"a42b5008da3dcfb9cc3629e31733af7b66b28245","_number":986945,"virtual_id_number":986945,"owner":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-06 19:36:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"all":[{"value":2,"date":"2026-05-06 17:12:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":1,"date":"2026-05-01 20:47:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"all":[{"value":1,"date":"2026-05-06 17:12:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-01 19:32:58.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-05-01 20:47:21.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2026-05-01 20:48:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-01 20:48:22.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2026-05-02 07:09:00.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2026-05-06 17:12:45.000000000","updated_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"}],"messages":[{"id":"277b48f68601de54f82787d69f44774e7bfdcd26","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-05-01 18:31:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e453997429246964181f3a4552576678902c3f90","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-05-01 19:32:58.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* quobyteci:  os-brick_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I3c93d3a7e8eedcffadc206d8c25fb1af1b8888ea .","accounts_in_message":[],"_revision_number":1},{"id":"e33d3978f479be908911b74ac614e08c409f89d4","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-05-01 20:47:21.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"fa2c27b16eaae429d291da1d29f17c4d77fa3ab6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-01 20:48:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be66f66a966d4d61a2b43d8ee138994b\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c7612a1f4e394ad492f962f116d59e23 : SUCCESS in 1m 33s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a9d6122799c44eba83e2f96055907c69 : SUCCESS in 2m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a005f44b3e074c3f8c43e9c3b9ec066f : SUCCESS in 2h 09m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b126db86052543468e81ea04d067a501 : SUCCESS in 3m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f4a4da8db8f74b38a90210271184c9f4 : SUCCESS in 2m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/73a206400e6447a092195138e4c13395 : SUCCESS in 3m 45s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ff6dca7b1eb04000adaa47d44596b025 : SUCCESS in 4m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80cbbb3f13694ead9c98355f21e515a8 : SUCCESS in 6m 39s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/27d040e4e8f5450590f33e939aa09504 : SUCCESS in 2m 18s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/74f2e50ccc244d26896ebfd9c129a532 : SUCCESS in 1h 01m 15s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/354a610976bd46188538499a961129dd : SUCCESS in 1h 23m 18s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/e3ad7bfeef0d49d1a187d37f2ba16bf6 : SUCCESS in 2h 01m 11s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/fd89be116ea74d19bb30c4ffcd7c4457 : SUCCESS in 3m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6bee4c80f95329795b25334ea625465fc1fb9a98","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-05-01 20:57:46.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/01_05_2026_20_51_27\nLeave a comment with","accounts_in_message":[],"_revision_number":1},{"id":"ea79dc17fdba09dde8c1699bea544801bc82f554","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-05-02 07:09:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4057/.","accounts_in_message":[],"_revision_number":1},{"id":"fef180d8ed956197bfffa6f9043167297dbbb779","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-05-02 10:25:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/3881/.","accounts_in_message":[],"_revision_number":1},{"id":"c2707181caadeaff607cbcab0330aea98c1a49c8","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2026-05-06 17:12:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ac677f2e97dfa087222e5593ec00db3eff665219","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-06 17:14:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d6afd4d71316b0167268df6a9a667d20dcf92d4e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-06 19:36:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/14acb28388e542b8a6d6223766ffe8eb\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/72ce9a4080e548cfb519db3ccd7e53cb : SUCCESS in 2m 24s\n- requirements-check https://zuul.opendev.org/t/openstack/build/84c6b0e195714032a49027f639226000 : SUCCESS in 3m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6da2e71cd4ca48b884896ff2307584f6 : SUCCESS in 2h 11m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12cd499f5fa64a42849e476810ee5736 : SUCCESS in 3m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b2a6ba4edb7b4887bda11d4cf008947f : SUCCESS in 4m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/adf9708f06754e94af848d5cfcc85092 : SUCCESS in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c60e2e9b4eca4cc4b1b0c67a0c0418f9 : SUCCESS in 7m 45s\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/313cab3485fa4f70a62217a7794ca66e : SUCCESS in 2h 15m 17s","accounts_in_message":[],"_revision_number":1},{"id":"6ff9625d26c279589d6e316d3de5a5203ac4bd11","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-06 19:36:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a42b5008da3dcfb9cc3629e31733af7b66b28245","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-06 19:38:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5d7dc9011b944668927b739b82f3a33c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7cf89acd7154c1c933bdb3c7f26ad4b : SUCCESS in 51s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"379b7a5fe27ee22148fe48b68c454059485b8ec9","revisions":{"379b7a5fe27ee22148fe48b68c454059485b8ec9":{"kind":"REWORK","_number":1,"created":"2026-05-01 18:31:27.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/45/986945/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/45/986945/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/45/986945/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/45/986945/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/45/986945/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/45/986945/1"}}},"commit":{"parents":[{"commit":"3a7984bb12f9c3bd82333f09b4f705076e4c2b52","subject":"Migrate setup.cfg to pyproject.toml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/3a7984bb12f9c3bd82333f09b4f705076e4c2b52"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2026-05-01 18:30:43.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2026-05-01 18:30:43.000000000","tz":-240},"subject":"Update python tested versions for hibiscus","message":"Update python tested versions for hibiscus\n\n- remove Python 3.10 classifier, add 3.13\n- update requires-python to 3.11\n\nNo changes needed in .zuul.yaml\n\nref: https://governance.openstack.org/tc/reference/runtimes/2026.2.html\n\nChange-Id: I3c93d3a7e8eedcffadc206d8c25fb1af1b8888ea\nSigned-off-by: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/379b7a5fe27ee22148fe48b68c454059485b8ec9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/379b7a5fe27ee22148fe48b68c454059485b8ec9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
