)]}'
{"id":"openstack%2Fos-brick~994253","triplet_id":"openstack%2Fos-brick~master~Iec82b571594d1b453e3aeafaf84d954c14125af9","project":"openstack/os-brick","branch":"master","topic":"oslotest-autospec","attention_set":{"8213":{"account":{"_account_id":8213,"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","username":"claudiub"},"last_update":"2026-06-23 11:45:45.000000000","reason":"Someone else replied on the change"},"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2026-06-22 12:32:11.000000000","reason":"\u003cGERRIT_ACCOUNT_4523\u003e replied on the change","reason_account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Iec82b571594d1b453e3aeafaf84d954c14125af9","subject":"tests: Enforce autospec\u003dTrue on mock patches","status":"NEW","created":"2026-06-22 09:42:13.000000000","updated":"2026-07-13 11:25:24.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f4ce7d4fb6ac130a6d367b459a322c212a91723f","_number":994253,"virtual_id_number":994253,"owner":{"_account_id":8213,"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","username":"claudiub"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-06-23 13:55:02.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-22 09:45:12.000000000","updated_by":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"reviewer":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"state":"CC"},{"updated":"2026-06-22 10:23:55.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-06-22 10:39:47.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-06-22 11:27:24.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2026-06-22 11:36:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-22 12:32:11.000000000","updated_by":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"reviewer":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"state":"REVIEWER"},{"updated":"2026-06-23 14:00:33.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2026-06-29 08:44:14.000000000","updated_by":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"a9acece939b8bb8c956c1cb2b0558813e2f7cccd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8213,"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","username":"claudiub"},"date":"2026-06-22 09:42:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"61f32fed25b8d1c89fcc4c41a03a7c2269f147cd","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-06-22 09:45:12.000000000","message":"Patch Set 1:\n\nBuild failed (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4471/.","accounts_in_message":[],"_revision_number":1},{"id":"40b3e1d11601b4b87aae9caf301d4de03a8b878a","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-06-22 10:23:55.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- cinder-storpool-tempest build/a0e31a4e25e54087aec3751eeff771aa : FAILURE in 40m 25s","accounts_in_message":[],"_revision_number":1},{"id":"64973dca7e0ce8260b508f14431bbc1898a08dd2","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-06-22 10:39:47.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* quobyteci:  os-brick_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Iec82b571594d1b453e3aeafaf84d954c14125af9 .","accounts_in_message":[],"_revision_number":1},{"id":"275d19b0a6d9cb3d0ad9c7a8539f53e4c16e7d73","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-06-22 11:27:24.000000000","message":"Patch Set 1:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/d2ddbac0f6c94b9eaa4babe32d877258\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/5c902eb1da604eadbfc9e3d8b800513c : FAILURE in 40m 19s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a19e914f5a4d46b5a64080529be8f56d : FAILURE in 35m 33s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f070392e7d4f400a8df30345a1ffac8e : FAILURE in 1h 44m 40s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/68065cebc395406cb9a0e329d50e9ea9 : FAILURE in 32m 05s","accounts_in_message":[],"_revision_number":1},{"id":"26dbcc85dca13c62f2fca28b66c48d1c51023ef4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-22 11:36:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e13d9f1e5ef54e96ac1e652090af22e8\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/da5d0900003048ddaec8febafa0adf90 : SUCCESS in 1h 27m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ebac9246d3c84fbfba7a2b4b9599ec08 : SUCCESS in 3m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3c9c41ed89b94e41a8d02c804135f449 : FAILURE in 4m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8f9969039a14460abbf388a257b861ff : FAILURE in 4m 10s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c9e4d39ef4584b289a45a2677952dd6e : FAILURE in 5m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a755dae532d846d2bb9e1bac05aafd5e : SUCCESS in 8m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5d1c0ff2ca064c11bbb39bd9d37ddbfd : SUCCESS in 6m 39s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/d2fe5295eba54f03a4e6189406a9a4ca : FAILURE in 5m 08s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/85cebfc25f884cd6b7cecdc0980ff581 : SUCCESS in 1h 48m 50s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/63283fe7d44e43169b5b264caca3d1c6 : SUCCESS in 1h 40m 32s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/9ae16b3d66df43b9afae3acc2ec475e1 : SUCCESS in 1h 52m 43s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/f227d0250d1c4169adccae0478409ca1 : SUCCESS in 3m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e6b43a7ce7641573775fbe047291f26cb27ea2ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8213,"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","username":"claudiub"},"date":"2026-06-23 09:23:37.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"f7e8881064ea63595a8d0560af84b17fab2e2435","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8213,"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","username":"claudiub"},"date":"2026-06-23 10:03:29.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"fb1058abde2d90b4840f8162da316fd6009080b2","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-06-23 10:27:25.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* quobyteci:  os-brick_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Iec82b571594d1b453e3aeafaf84d954c14125af9 .","accounts_in_message":[],"_revision_number":2},{"id":"ba66b7fc71ee26f7660f808eb8c9946ce3f859a6","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-06-23 11:06:44.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* quobyteci:  os-brick_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Iec82b571594d1b453e3aeafaf84d954c14125af9 .","accounts_in_message":[],"_revision_number":3},{"id":"9e61c384d81fc848bcbc2d67bdb06be1015f6a23","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2026-06-23 11:45:45.000000000","message":"Patch Set 3:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/1026fabb5a8b4feb94409fc776f11e1c\n\n- pure-devstack-cinder-tempest-iscsi-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f4161ca44b1b456eb3cdb1905f0874d2 : FAILURE in 33m 38s\n- pure-devstack-cinder-tempest-fc-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/67a3ce07e98248f28174fca8dc634b44 : FAILURE in 37m 50s\n- pure-devstack-cinder-tempest-nvme-roce-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/c66c8c101cca474fabd45ce85f064b0c : FAILURE in 1h 41m 43s (non-voting)\n- pure-devstack-cinder-tempest-nvme-tcp-aio-noble https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ca8cb6c265b8410fb62b9f77334746c7 : FAILURE in 35m 28s","accounts_in_message":[],"_revision_number":3},{"id":"6a59d598fc2240ca41771867699e86227e368aaf","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2026-06-23 13:38:59.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- cinder-storpool-tempest build/5e21687ff25349ca9fd819f016753e3f : FAILURE in 1h 25m 04s","accounts_in_message":[],"_revision_number":3},{"id":"f4232a23d0918908c8100d541197b80db8333c9b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-23 13:55:02.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/10c7319b772a46ec9a531afa3e841733\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0995bd759b884f11ad36f8a2fc62c852 : SUCCESS in 1h 47m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c3bd726291f49c6a6dc352a30c10ae5 : SUCCESS in 3m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9c507a417caf440fae2138480ee9217b : SUCCESS in 4m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8edd7efe02ce46428ad8f6412e41eaf2 : SUCCESS in 3m 33s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c7a02a42f9c747c7a5d44033302e2220 : SUCCESS in 6m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96be7bc81e5f4b8fad4b7621e6b4e20e : SUCCESS in 8m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cc3be2015e724537b9fdc7a5ce261495 : SUCCESS in 6m 25s\n- os-brick-code-coverage https://zuul.opendev.org/t/openstack/build/a7cefb6d37c54964a9149c00884cad4e : SUCCESS in 4m 37s (non-voting)\n- os-brick-src-devstack-plugin-ceph-nv https://zuul.opendev.org/t/openstack/build/7d9c54087b184976a2df0b1898ee83d6 : SUCCESS in 1h 50m 14s (non-voting)\n- os-brick-src-tempest-lvm-lio-barbican https://zuul.opendev.org/t/openstack/build/706fd48403c84b34bea294b5bc51150a : SUCCESS in 1h 57m 26s\n- os-brick-src-tempest-nfs https://zuul.opendev.org/t/openstack/build/0a04ddc7a8bf41608f3bd5c7d5dcc0d7 : TIMED_OUT in 3h 10m 13s (non-voting)\n- os-brick-mypy https://zuul.opendev.org/t/openstack/build/eb0713a2eaff43358a0884f96a2d68eb : SUCCESS in 2m 53s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"bc9fe305a38455cd4ee76b9e64b21372a5ab3bda","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-06-23 14:11:24.000000000","message":"Patch Set 3:\n\nBuild succeeded. log path: https://github.com/LightBitsLabs/logs/tree/main/23_06_2026_14_03_57\nLeave a comment with","accounts_in_message":[],"_revision_number":3},{"id":"4eda78e34c2538f22aa2bafe2bdc3291a32ae535","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-06-23 14:12:17.000000000","message":"Patch Set 2:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4486/.","accounts_in_message":[],"_revision_number":2},{"id":"16d24d6943d9cbcc09cd5bda08e5efd92e0457b6","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-06-23 17:35:03.000000000","message":"Patch Set 3:\n\nBuild succeeded (iSCSI).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-iscsi/4489/.","accounts_in_message":[],"_revision_number":3},{"id":"005993d3835e2ab5932ac609d7431d33266960ac","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8213,"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","username":"claudiub"},"date":"2026-06-28 16:18:48.000000000","message":"Topic set to oslotest-autospec","accounts_in_message":[],"_revision_number":3},{"id":"d6883a13ca988bd75e7e12d7fbcebd0884bfb8e6","tag":"autogenerated:NEC_Cinder_CI","author":{"_account_id":23613,"name":"NEC Cinder CI","email":"nec-cinder-ci@istorage.jp.nec.com","username":"nec-cinder-ci","tags":["SERVICE_USER"]},"date":"2026-06-29 08:44:14.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- NEC-ISCSI https://cinder-cilogs.openstack.nec-solutioninnovators.com/53/994253/3/20260629114328/NEC-ISCSI : SUCCESS in 5h 32m 56s\n- NEC-FC https://cinder-cilogs.openstack.nec-solutioninnovators.com/53/994253/3/20260629114328/NEC-FC : SUCCESS in 6h 00m 43s","accounts_in_message":[],"_revision_number":3},{"id":"bf2186b11a1e346b1334f33bb99a7e6910305c78","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-12 16:24:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4295/.","accounts_in_message":[],"_revision_number":1},{"id":"84e6afdc8910fb6619b2bcde422a73f4f1c050c4","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-13 08:23:13.000000000","message":"Patch Set 2:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4310/.","accounts_in_message":[],"_revision_number":2},{"id":"f4ce7d4fb6ac130a6d367b459a322c212a91723f","author":{"_account_id":33746,"name":"YADRO TATLIN CI","email":"tatlin_ci@yadro.com","username":"tatlin-ci"},"date":"2026-07-13 11:25:24.000000000","message":"Patch Set 3:\n\nBuild succeeded (FC).\nLogs: http://tempest-openstack.spb.yadro.com/openstack-gerrit-trigger-fc/4313/.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"08050dad6a8ff003aec442f21839d18c414027a6","revisions":{"a3179d95e03635da14fe68e7d86f89ecfb1a3993":{"kind":"REWORK","_number":1,"created":"2026-06-22 09:42:13.000000000","uploader":{"_account_id":8213,"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","username":"claudiub"},"ref":"refs/changes/53/994253/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/53/994253/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/53/994253/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/53/994253/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/53/994253/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/53/994253/1"}}},"commit":{"parents":[{"commit":"379b7a5fe27ee22148fe48b68c454059485b8ec9","subject":"Update python tested versions for hibiscus","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/379b7a5fe27ee22148fe48b68c454059485b8ec9"}]}],"author":{"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","date":"2026-06-13 11:35:31.000000000","tz":0},"committer":{"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","date":"2026-06-22 09:42:11.000000000","tz":0},"subject":"tests: Enforce autospec\u003dTrue on mock patches","message":"tests: Enforce autospec\u003dTrue on mock patches\n\nUsing mock_fixture.patch_mock_module() will enforce autospec\u003dTrue on all\nthe mock.patch and similar calls. This will enforce signature checks on\nthe mocked methods and functions, catching potential bugs missed by\notherwise false positive unit tests.\n\nCommon issues / changes:\n- time.sleep mocked twice\n- open is now patched wth mock.mock_open\n\nChange-Id: Iec82b571594d1b453e3aeafaf84d954c14125af9\nSigned-off-by: Claudiu Belu \u003ccbelu@cloudbasesolutions.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/a3179d95e03635da14fe68e7d86f89ecfb1a3993"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/a3179d95e03635da14fe68e7d86f89ecfb1a3993"}]},"branch":"refs/heads/master"},"5bf75c51ef6490801bcf3599b0c555754b1a31fd":{"kind":"REWORK","_number":2,"created":"2026-06-23 09:23:37.000000000","uploader":{"_account_id":8213,"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","username":"claudiub"},"ref":"refs/changes/53/994253/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/53/994253/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/53/994253/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/53/994253/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/53/994253/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/53/994253/2"}}},"commit":{"parents":[{"commit":"379b7a5fe27ee22148fe48b68c454059485b8ec9","subject":"Update python tested versions for hibiscus","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/379b7a5fe27ee22148fe48b68c454059485b8ec9"}]}],"author":{"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","date":"2026-06-13 11:35:31.000000000","tz":0},"committer":{"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","date":"2026-06-23 09:23:27.000000000","tz":0},"subject":"tests: Enforce autospec\u003dTrue on mock patches","message":"tests: Enforce autospec\u003dTrue on mock patches\n\nUsing mock_fixture.patch_mock_module() will enforce autospec\u003dTrue on all\nthe mock.patch and similar calls. This will enforce signature checks on\nthe mocked methods and functions, catching potential bugs missed by\notherwise false positive unit tests.\n\nCommon issues / changes:\n- time.sleep mocked twice\n- open is now patched wth mock.mock_open\n\nChange-Id: Iec82b571594d1b453e3aeafaf84d954c14125af9\nSigned-off-by: Claudiu Belu \u003ccbelu@cloudbasesolutions.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/5bf75c51ef6490801bcf3599b0c555754b1a31fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/5bf75c51ef6490801bcf3599b0c555754b1a31fd"}]},"branch":"refs/heads/master"},"08050dad6a8ff003aec442f21839d18c414027a6":{"kind":"REWORK","_number":3,"created":"2026-06-23 10:03:29.000000000","uploader":{"_account_id":8213,"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","username":"claudiub"},"ref":"refs/changes/53/994253/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-brick","ref":"refs/changes/53/994253/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/53/994253/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/53/994253/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-brick refs/changes/53/994253/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-brick refs/changes/53/994253/3"}}},"commit":{"parents":[{"commit":"379b7a5fe27ee22148fe48b68c454059485b8ec9","subject":"Update python tested versions for hibiscus","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/379b7a5fe27ee22148fe48b68c454059485b8ec9"}]}],"author":{"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","date":"2026-06-13 11:35:31.000000000","tz":0},"committer":{"name":"Claudiu Belu","email":"cbelu@cloudbasesolutions.com","date":"2026-06-23 10:03:21.000000000","tz":0},"subject":"tests: Enforce autospec\u003dTrue on mock patches","message":"tests: Enforce autospec\u003dTrue on mock patches\n\nUsing mock_fixture.patch_mock_module() will enforce autospec\u003dTrue on all\nthe mock.patch and similar calls. This will enforce signature checks on\nthe mocked methods and functions, catching potential bugs missed by\notherwise false positive unit tests.\n\nCommon issues / changes:\n- time.sleep mocked twice\n- open is now patched wth mock.mock_open\n\nChange-Id: Iec82b571594d1b453e3aeafaf84d954c14125af9\nSigned-off-by: Claudiu Belu \u003ccbelu@cloudbasesolutions.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/08050dad6a8ff003aec442f21839d18c414027a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-brick/commit/08050dad6a8ff003aec442f21839d18c414027a6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
