)]}'
{"id":"openstack%2Fos-client-config~549307","triplet_id":"openstack%2Fos-client-config~master~I7180e20aa675c0ccbe3718493a6eff2be7e1b8b6","project":"openstack/os-client-config","branch":"master","topic":"os-client-config-shim","hashtags":[],"change_id":"I7180e20aa675c0ccbe3718493a6eff2be7e1b8b6","subject":"Replace guts with openstack.config","status":"MERGED","created":"2018-03-02 19:09:49.000000000","updated":"2018-05-02 13:52:19.000000000","submitted":"2018-05-02 13:52:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"549307-1525269139938-b5c247cf","meta_rev_id":"83762854f6257259f53a05df7427430fe1f70492","_number":549307,"virtual_id_number":549307,"owner":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2018-05-01 11:38:47.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":3099,"name":"David Shrewsbury","email":"dshrewsb@redhat.com","username":"dshrews"},{"value":2,"date":"2018-05-02 13:52:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":2,"date":"2018-05-01 14:34:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2018-05-02 12:28:48.000000000","_account_id":3099,"name":"David Shrewsbury","email":"dshrewsb@redhat.com","username":"dshrews"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3099,"name":"David Shrewsbury","email":"dshrewsb@redhat.com","username":"dshrews"},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2018-05-02 12:28:48.000000000","_account_id":3099,"name":"David Shrewsbury","email":"dshrewsb@redhat.com","username":"dshrews"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Allow-Post-Review":{"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":3099,"name":"David Shrewsbury","email":"dshrewsb@redhat.com","username":"dshrews"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Abstain","+1":"Approved for Zuul Post Review Pipeline"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":3099,"name":"David Shrewsbury","email":"dshrewsb@redhat.com","username":"dshrews"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-05-01 14:34:10.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2018-05-02 12:28:48.000000000","updated_by":{"_account_id":3099,"name":"David Shrewsbury","email":"dshrewsb@redhat.com","username":"dshrews"},"reviewer":{"_account_id":3099,"name":"David Shrewsbury","email":"dshrewsb@redhat.com","username":"dshrews"},"state":"REVIEWER"},{"updated":"2018-05-02 13:52:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c486f6192c7636c18f3aeacc2d262d52fd7fd812","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2018-03-02 19:09:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5eae21c64b8114007be2acdb25e8cfa16bb5c793","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-02 20:56:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/07/549307/1/check/requirements-check/0b8ef73/ : SUCCESS in 2m 44s\n- build-openstack-sphinx-docs http://logs.openstack.org/07/549307/1/check/build-openstack-sphinx-docs/c01741a/ : FAILURE in 3m 38s\n- openstack-tox-pep8 http://logs.openstack.org/07/549307/1/check/openstack-tox-pep8/078fa90/ : FAILURE in 4m 03s\n- openstack-tox-py27 http://logs.openstack.org/07/549307/1/check/openstack-tox-py27/2fd55ea/ : FAILURE in 4m 00s\n- openstack-tox-py35 http://logs.openstack.org/07/549307/1/check/openstack-tox-py35/374c524/ : FAILURE in 3m 42s\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/07/549307/1/check/legacy-tempest-dsvm-neutron-src/1764db6/ : SUCCESS in 1h 40m 55s\n- build-openstack-releasenotes http://logs.openstack.org/07/549307/1/check/build-openstack-releasenotes/bbec394/html/ : SUCCESS in 3m 06s\n- shade-functional-devstack-tips http://logs.openstack.org/07/549307/1/check/shade-functional-devstack-tips/19d81d3/ : FAILURE in 32m 35s\n- shade-functional-devstack-tips-python3 http://logs.openstack.org/07/549307/1/check/shade-functional-devstack-tips-python3/d12b082/ : FAILURE in 34m 47s\n- shade-tox-py27-tips http://logs.openstack.org/07/549307/1/check/shade-tox-py27-tips/0d00981/ : FAILURE in 4m 12s\n- shade-tox-py35-tips http://logs.openstack.org/07/549307/1/check/shade-tox-py35-tips/cb36586/ : FAILURE in 4m 23s","accounts_in_message":[],"_revision_number":1},{"id":"490003859835bbfe8dc2a12c1cbfc42a84dfc05f","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2018-03-03 14:38:07.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"62689f19af75924e4dc8f3bd4cd9734bf16f7095","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-03 14:42:12.000000000","message":"Patch Set 2: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"a810e15f66a0b454a52f43ce1af56cf19d172b8b","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2018-03-04 13:17:45.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"4bdbea9e583f138c040b7b4b074c3ce739842e2e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-04 14:46:33.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/07/549307/2/check/requirements-check/a692181/ : SUCCESS in 3m 06s\n- build-openstack-sphinx-docs http://logs.openstack.org/07/549307/2/check/build-openstack-sphinx-docs/36203ad/ : FAILURE in 4m 35s\n- openstack-tox-pep8 http://logs.openstack.org/07/549307/2/check/openstack-tox-pep8/a7d0e7c/ : FAILURE in 2m 37s\n- openstack-tox-py27 http://logs.openstack.org/07/549307/2/check/openstack-tox-py27/33e3053/ : FAILURE in 3m 32s\n- openstack-tox-py35 http://logs.openstack.org/07/549307/2/check/openstack-tox-py35/28fe2de/ : FAILURE in 3m 24s\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/07/549307/2/check/legacy-tempest-dsvm-neutron-src/56970ca/ : SUCCESS in 1h 21m 27s\n- build-openstack-releasenotes http://logs.openstack.org/07/549307/2/check/build-openstack-releasenotes/422d7ca/html/ : SUCCESS in 2m 50s\n- os-client-config-tox-py27-tips http://logs.openstack.org/07/549307/2/check/os-client-config-tox-py27-tips/5f2f5a0/ : FAILURE in 4m 13s\n- os-client-config-tox-py35-tips http://logs.openstack.org/07/549307/2/check/os-client-config-tox-py35-tips/500c340/ : FAILURE in 3m 36s\n- shade-functional-devstack-tips http://logs.openstack.org/07/549307/2/check/shade-functional-devstack-tips/1de7eb8/ : FAILURE in 46m 06s\n- shade-functional-devstack-tips-python3 http://logs.openstack.org/07/549307/2/check/shade-functional-devstack-tips-python3/1420218/ : FAILURE in 34m 50s\n- shade-tox-py27-tips http://logs.openstack.org/07/549307/2/check/shade-tox-py27-tips/406401a/ : SUCCESS in 2m 13s\n- shade-tox-py35-tips http://logs.openstack.org/07/549307/2/check/shade-tox-py35-tips/53fc420/ : SUCCESS in 3m 26s","accounts_in_message":[],"_revision_number":2},{"id":"9ac2aac440eccbe9e8320860dff986ebde56484c","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2018-03-10 14:11:21.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5a53ddeffa3efd57b6bbd38328aa2507d455036e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-10 15:42:30.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/07/549307/3/check/requirements-check/9e0fa39/ : SUCCESS in 3m 43s\n- build-openstack-sphinx-docs http://logs.openstack.org/07/549307/3/check/build-openstack-sphinx-docs/5ec9ba0/html/ : SUCCESS in 3m 42s\n- openstack-tox-pep8 http://logs.openstack.org/07/549307/3/check/openstack-tox-pep8/412cbf2/ : FAILURE in 3m 42s\n- openstack-tox-py27 http://logs.openstack.org/07/549307/3/check/openstack-tox-py27/2eb1d95/ : FAILURE in 3m 39s\n- openstack-tox-py35 http://logs.openstack.org/07/549307/3/check/openstack-tox-py35/57f07d4/ : FAILURE in 3m 44s\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/07/549307/3/check/legacy-tempest-dsvm-neutron-src/f4e384b/ : SUCCESS in 1h 25m 35s\n- build-openstack-releasenotes http://logs.openstack.org/07/549307/3/check/build-openstack-releasenotes/32ec406/html/ : SUCCESS in 3m 10s\n- shade-functional-devstack-tips http://logs.openstack.org/07/549307/3/check/shade-functional-devstack-tips/102dbf6/ : FAILURE in 39m 19s\n- shade-functional-devstack-tips-python3 http://logs.openstack.org/07/549307/3/check/shade-functional-devstack-tips-python3/5b160f7/ : FAILURE in 33m 32s\n- shade-tox-py27-tips http://logs.openstack.org/07/549307/3/check/shade-tox-py27-tips/156ea3b/ : SUCCESS in 3m 23s\n- shade-tox-py35-tips http://logs.openstack.org/07/549307/3/check/shade-tox-py35-tips/aed38ec/ : SUCCESS in 2m 40s\n- os-client-config-tox-py27-tips http://logs.openstack.org/07/549307/3/check/os-client-config-tox-py27-tips/3dd502a/ : SUCCESS in 4m 14s\n- os-client-config-tox-py35-tips http://logs.openstack.org/07/549307/3/check/os-client-config-tox-py35-tips/333449a/ : SUCCESS in 3m 24s","accounts_in_message":[],"_revision_number":3},{"id":"38181d8c6756859b89a759640bc1fe339eef3762","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2018-03-29 16:08:40.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"b3d724eddf8a878d00b1c53d44c716f66867e703","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-29 17:34:07.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/07/549307/4/check/requirements-check/6c9afce/ : SUCCESS in 3m 30s\n- build-openstack-sphinx-docs http://logs.openstack.org/07/549307/4/check/build-openstack-sphinx-docs/b839729/html/ : SUCCESS in 3m 52s\n- openstack-tox-pep8 http://logs.openstack.org/07/549307/4/check/openstack-tox-pep8/03da6fe/ : FAILURE in 4m 21s\n- openstack-tox-py27 http://logs.openstack.org/07/549307/4/check/openstack-tox-py27/fd27b2e/ : FAILURE in 4m 02s\n- openstack-tox-py35 http://logs.openstack.org/07/549307/4/check/openstack-tox-py35/8de861b/ : FAILURE in 3m 39s\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/07/549307/4/check/legacy-tempest-dsvm-neutron-src/65005b8/ : SUCCESS in 1h 23m 24s\n- build-openstack-releasenotes http://logs.openstack.org/07/549307/4/check/build-openstack-releasenotes/633742c/html/ : SUCCESS in 3m 06s\n- os-client-config-tox-py27-tips http://logs.openstack.org/07/549307/4/check/os-client-config-tox-py27-tips/53f62b3/ : SUCCESS in 4m 22s\n- os-client-config-tox-py35-tips http://logs.openstack.org/07/549307/4/check/os-client-config-tox-py35-tips/a3f08f1/ : SUCCESS in 5m 09s\n- shade-functional-devstack-tips http://logs.openstack.org/07/549307/4/check/shade-functional-devstack-tips/6492e10/ : FAILURE in 40m 10s\n- shade-functional-devstack-tips-python3 http://logs.openstack.org/07/549307/4/check/shade-functional-devstack-tips-python3/caeeca9/ : FAILURE in 36m 06s\n- shade-tox-py27-tips http://logs.openstack.org/07/549307/4/check/shade-tox-py27-tips/2eae8eb/ : SUCCESS in 2m 47s\n- shade-tox-py35-tips http://logs.openstack.org/07/549307/4/check/shade-tox-py35-tips/4e765ea/ : SUCCESS in 2m 33s","accounts_in_message":[],"_revision_number":4},{"id":"bbe4159fb262cc5a5c215833ba00397b19d8a392","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2018-04-27 11:04:28.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"3504d3f0b5e2ce43b496d6fc7f26a13d66484580","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-27 12:28:39.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/07/549307/5/check/requirements-check/1a4a75c/ : FAILURE in 2m 41s\n- build-openstack-sphinx-docs http://logs.openstack.org/07/549307/5/check/build-openstack-sphinx-docs/83b0896/html/ : SUCCESS in 3m 24s\n- openstack-tox-pep8 http://logs.openstack.org/07/549307/5/check/openstack-tox-pep8/341e913/ : SUCCESS in 3m 16s\n- openstack-tox-py27 http://logs.openstack.org/07/549307/5/check/openstack-tox-py27/171517f/ : SUCCESS in 3m 41s\n- openstack-tox-py35 http://logs.openstack.org/07/549307/5/check/openstack-tox-py35/3415a42/ : SUCCESS in 3m 38s\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/07/549307/5/check/legacy-tempest-dsvm-neutron-src/d7be610/ : SUCCESS in 1h 22m 48s\n- build-openstack-releasenotes http://logs.openstack.org/07/549307/5/check/build-openstack-releasenotes/26f8b26/html/ : SUCCESS in 3m 11s\n- shade-functional-devstack-tips http://logs.openstack.org/07/549307/5/check/shade-functional-devstack-tips/a411e83/ : FAILURE in 50m 24s\n- shade-functional-devstack-tips-python3 http://logs.openstack.org/07/549307/5/check/shade-functional-devstack-tips-python3/b334927/ : FAILURE in 33m 47s\n- shade-tox-py27-tips http://logs.openstack.org/07/549307/5/check/shade-tox-py27-tips/51d32b3/ : FAILURE in 3m 11s\n- shade-tox-py35-tips http://logs.openstack.org/07/549307/5/check/shade-tox-py35-tips/486f0dc/ : FAILURE in 3m 08s\n- os-client-config-tox-py27-tips http://logs.openstack.org/07/549307/5/check/os-client-config-tox-py27-tips/6c8a31f/ : SUCCESS in 3m 53s\n- os-client-config-tox-py35-tips http://logs.openstack.org/07/549307/5/check/os-client-config-tox-py35-tips/e9d8435/ : SUCCESS in 4m 20s\n- osc-functional-devstack-tips http://logs.openstack.org/07/549307/5/check/osc-functional-devstack-tips/7a3d4af/ : SUCCESS in 39m 36s\n- openstack-tox-lower-constraints http://logs.openstack.org/07/549307/5/check/openstack-tox-lower-constraints/c3a3310/ : SUCCESS in 4m 35s","accounts_in_message":[],"_revision_number":5},{"id":"29e7fd741a2720aadb0dd9dc25d298e1c8f33fcd","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2018-04-27 13:44:50.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"34a5cc77e8b117b4ada8adacf39fa5e82a0fc69c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-27 15:18:20.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/07/549307/6/check/requirements-check/e6756ff/ : SUCCESS in 3m 07s\n- build-openstack-sphinx-docs http://logs.openstack.org/07/549307/6/check/build-openstack-sphinx-docs/ec9b608/html/ : SUCCESS in 3m 46s\n- openstack-tox-pep8 http://logs.openstack.org/07/549307/6/check/openstack-tox-pep8/69ee67f/ : SUCCESS in 3m 30s\n- openstack-tox-py27 http://logs.openstack.org/07/549307/6/check/openstack-tox-py27/445ab58/ : SUCCESS in 3m 39s\n- openstack-tox-py35 http://logs.openstack.org/07/549307/6/check/openstack-tox-py35/6226d74/ : SUCCESS in 3m 49s\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/07/549307/6/check/legacy-tempest-dsvm-neutron-src/89b67b8/ : SUCCESS in 1h 27m 12s\n- build-openstack-releasenotes http://logs.openstack.org/07/549307/6/check/build-openstack-releasenotes/28e0d1f/html/ : SUCCESS in 3m 48s\n- os-client-config-tox-py27-tips http://logs.openstack.org/07/549307/6/check/os-client-config-tox-py27-tips/5331022/ : SUCCESS in 4m 34s\n- os-client-config-tox-py35-tips http://logs.openstack.org/07/549307/6/check/os-client-config-tox-py35-tips/828ddc9/ : SUCCESS in 4m 30s\n- shade-functional-devstack-tips http://logs.openstack.org/07/549307/6/check/shade-functional-devstack-tips/f87178a/ : FAILURE in 41m 28s\n- shade-functional-devstack-tips-python3 http://logs.openstack.org/07/549307/6/check/shade-functional-devstack-tips-python3/9fa2e80/ : FAILURE in 40m 35s\n- shade-tox-py27-tips http://logs.openstack.org/07/549307/6/check/shade-tox-py27-tips/6e66776/ : SUCCESS in 3m 02s\n- shade-tox-py35-tips http://logs.openstack.org/07/549307/6/check/shade-tox-py35-tips/0e5b80e/ : SUCCESS in 3m 12s\n- osc-functional-devstack-tips http://logs.openstack.org/07/549307/6/check/osc-functional-devstack-tips/dbd60c5/ : SUCCESS in 45m 41s\n- openstack-tox-lower-constraints http://logs.openstack.org/07/549307/6/check/openstack-tox-lower-constraints/2877ba3/ : SUCCESS in 3m 55s","accounts_in_message":[],"_revision_number":6},{"id":"815a68fe3df4c3ecaa7dd9dbd5916243289d945b","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2018-04-30 14:49:37.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"4389523fd56b4cd562d6b9a4f3338cd77894b2c7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-30 16:19:08.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/07/549307/6/check/requirements-check/f57ca3f/ : SUCCESS in 3m 13s\n- build-openstack-sphinx-docs http://logs.openstack.org/07/549307/6/check/build-openstack-sphinx-docs/5232bb6/html/ : SUCCESS in 3m 40s\n- openstack-tox-pep8 http://logs.openstack.org/07/549307/6/check/openstack-tox-pep8/c648460/ : SUCCESS in 4m 20s\n- openstack-tox-py27 http://logs.openstack.org/07/549307/6/check/openstack-tox-py27/31ecc62/ : SUCCESS in 4m 34s\n- openstack-tox-py35 http://logs.openstack.org/07/549307/6/check/openstack-tox-py35/26fda1c/ : SUCCESS in 4m 13s\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/07/549307/6/check/legacy-tempest-dsvm-neutron-src/d152505/ : SUCCESS in 1h 22m 52s\n- build-openstack-releasenotes http://logs.openstack.org/07/549307/6/check/build-openstack-releasenotes/6177d24/html/ : SUCCESS in 3m 21s\n- os-client-config-tox-py27-tips http://logs.openstack.org/07/549307/6/check/os-client-config-tox-py27-tips/d6d5b98/ : SUCCESS in 3m 55s\n- os-client-config-tox-py35-tips http://logs.openstack.org/07/549307/6/check/os-client-config-tox-py35-tips/5e07b98/ : SUCCESS in 4m 37s\n- shade-functional-devstack-tips http://logs.openstack.org/07/549307/6/check/shade-functional-devstack-tips/543738e/ : SUCCESS in 40m 26s\n- shade-functional-devstack-tips-python3 http://logs.openstack.org/07/549307/6/check/shade-functional-devstack-tips-python3/12df344/ : SUCCESS in 38m 18s\n- shade-tox-py27-tips http://logs.openstack.org/07/549307/6/check/shade-tox-py27-tips/5d5173f/ : SUCCESS in 4m 00s\n- shade-tox-py35-tips http://logs.openstack.org/07/549307/6/check/shade-tox-py35-tips/704072f/ : SUCCESS in 3m 10s\n- osc-functional-devstack-tips http://logs.openstack.org/07/549307/6/check/osc-functional-devstack-tips/e20ffdd/ : FAILURE in 42m 12s\n- openstack-tox-lower-constraints http://logs.openstack.org/07/549307/6/check/openstack-tox-lower-constraints/8769bc1/ : SUCCESS in 3m 50s","accounts_in_message":[],"_revision_number":6},{"id":"f71b3119817d51b9732022fc3b5f52b5bde6b6fd","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2018-05-01 11:38:47.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"418072e51a8264efaa5f42367a3de9f80f34da6c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-01 13:05:02.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/07/549307/6/check/requirements-check/60dfda9/ : SUCCESS in 2m 44s\n- build-openstack-sphinx-docs http://logs.openstack.org/07/549307/6/check/build-openstack-sphinx-docs/53eb0c9/html/ : SUCCESS in 3m 05s\n- openstack-tox-pep8 http://logs.openstack.org/07/549307/6/check/openstack-tox-pep8/9cd5e3a/ : SUCCESS in 3m 20s\n- openstack-tox-py27 http://logs.openstack.org/07/549307/6/check/openstack-tox-py27/2dacc88/ : SUCCESS in 3m 30s\n- openstack-tox-py35 http://logs.openstack.org/07/549307/6/check/openstack-tox-py35/e6088fa/ : SUCCESS in 3m 53s\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/07/549307/6/check/legacy-tempest-dsvm-neutron-src/b836f51/ : SUCCESS in 1h 25m 15s\n- build-openstack-releasenotes http://logs.openstack.org/07/549307/6/check/build-openstack-releasenotes/39c2071/html/ : SUCCESS in 3m 00s\n- os-client-config-tox-py27-tips http://logs.openstack.org/07/549307/6/check/os-client-config-tox-py27-tips/25146f6/ : SUCCESS in 5m 12s\n- os-client-config-tox-py35-tips http://logs.openstack.org/07/549307/6/check/os-client-config-tox-py35-tips/f944d56/ : SUCCESS in 4m 44s\n- shade-functional-devstack-tips http://logs.openstack.org/07/549307/6/check/shade-functional-devstack-tips/be73558/ : SUCCESS in 40m 41s\n- shade-functional-devstack-tips-python3 http://logs.openstack.org/07/549307/6/check/shade-functional-devstack-tips-python3/12770c0/ : SUCCESS in 34m 23s\n- shade-tox-py27-tips http://logs.openstack.org/07/549307/6/check/shade-tox-py27-tips/54a85f8/ : SUCCESS in 2m 59s\n- shade-tox-py35-tips http://logs.openstack.org/07/549307/6/check/shade-tox-py35-tips/f8c5282/ : SUCCESS in 2m 47s\n- osc-functional-devstack-tips http://logs.openstack.org/07/549307/6/check/osc-functional-devstack-tips/9d65d21/ : SUCCESS in 48m 57s\n- openstack-tox-lower-constraints http://logs.openstack.org/07/549307/6/check/openstack-tox-lower-constraints/e8ca549/ : SUCCESS in 3m 47s","accounts_in_message":[],"_revision_number":6},{"id":"a0941a8e6c964bc3ae8366579bb8a5b97e1c5226","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2018-05-01 14:34:10.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"9eab033f4129dee0d92989fc2c6e4388f6b9b42c","author":{"_account_id":3099,"name":"David Shrewsbury","email":"dshrewsb@redhat.com","username":"dshrews"},"date":"2018-05-02 12:28:48.000000000","message":"Patch Set 6: Workflow+1 Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"7ca60a680e3d9a0a77e1d24175670ab5849d83c4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-02 12:29:52.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"b4e9d010aa34bf080f5515d36d7cee78ac53cc78","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-02 13:52:19.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check http://logs.openstack.org/07/549307/6/gate/requirements-check/8c8051e/ : SUCCESS in 3m 31s\n- build-openstack-sphinx-docs http://logs.openstack.org/07/549307/6/gate/build-openstack-sphinx-docs/3b31754/html/ : SUCCESS in 4m 18s\n- openstack-tox-pep8 http://logs.openstack.org/07/549307/6/gate/openstack-tox-pep8/cf9da40/ : SUCCESS in 2m 53s\n- openstack-tox-py27 http://logs.openstack.org/07/549307/6/gate/openstack-tox-py27/0bfac34/ : SUCCESS in 3m 28s\n- openstack-tox-py35 http://logs.openstack.org/07/549307/6/gate/openstack-tox-py35/1e4fa6b/ : SUCCESS in 4m 40s\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/07/549307/6/gate/legacy-tempest-dsvm-neutron-src/414ea68/ : SUCCESS in 1h 19m 56s\n- build-openstack-releasenotes http://logs.openstack.org/07/549307/6/gate/build-openstack-releasenotes/0421f7b/html/ : SUCCESS in 3m 26s\n- os-client-config-tox-py27-tips http://logs.openstack.org/07/549307/6/gate/os-client-config-tox-py27-tips/c9a88a8/ : SUCCESS in 3m 36s\n- os-client-config-tox-py35-tips http://logs.openstack.org/07/549307/6/gate/os-client-config-tox-py35-tips/687f56d/ : SUCCESS in 4m 01s\n- shade-functional-devstack-tips http://logs.openstack.org/07/549307/6/gate/shade-functional-devstack-tips/c59588c/ : SUCCESS in 43m 35s\n- shade-functional-devstack-tips-python3 http://logs.openstack.org/07/549307/6/gate/shade-functional-devstack-tips-python3/854cbda/ : SUCCESS in 39m 47s\n- shade-tox-py27-tips http://logs.openstack.org/07/549307/6/gate/shade-tox-py27-tips/a6e5c28/ : SUCCESS in 3m 10s\n- shade-tox-py35-tips http://logs.openstack.org/07/549307/6/gate/shade-tox-py35-tips/c680e6a/ : SUCCESS in 4m 03s\n- osc-functional-devstack-tips http://logs.openstack.org/07/549307/6/gate/osc-functional-devstack-tips/8701053/ : SUCCESS in 37m 24s\n- openstack-tox-lower-constraints http://logs.openstack.org/07/549307/6/gate/openstack-tox-lower-constraints/4e3d279/ : SUCCESS in 3m 28s","accounts_in_message":[],"_revision_number":6},{"id":"817ada62255e573ec328eddc6f07c332d7e47b3d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-02 13:52:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"5c2eae8f8242332aff59d18b45a7d17f01524aaa","revisions":{"73f037e38213b6944eeb8e616afe71c989f994cd":{"kind":"REWORK","_number":1,"created":"2018-03-02 19:09:49.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/07/549307/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-client-config","ref":"refs/changes/07/549307/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-client-config refs/changes/07/549307/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-client-config refs/changes/07/549307/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-client-config refs/changes/07/549307/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-client-config refs/changes/07/549307/1"}}},"commit":{"parents":[{"commit":"007793f9c5d6a2e873ef18a99980e3d72160145a","subject":"Updated from global requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-client-config/commit/007793f9c5d6a2e873ef18a99980e3d72160145a"}]}],"author":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2018-03-02 09:42:37.000000000","tz":60},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2018-03-02 19:09:33.000000000","tz":-300},"subject":"Replace guts with openstack.config","message":"Replace guts with openstack.config\n\nThe code has been moved into openstacksdk. Make os-client-config a\nthin wrapper around that code.\n\nThe legacy client code needs to stay, as it was removed in openstacksdk.\n\nWe also have to plumb os_client_config.defaults and\nos_client_config.cloud_region.CloudRegion in so that users get the\nos-client-config versions from the openstacksdk methods that return the\nrelevant objects.\n\nSupport for per-service region_name has been dropped. This was never\nused anywhere and was added for compatibility with openstacksdk Profile\nobjects. Since Profile objects are no more, we can remove it. It never\nmade any sense.\n\nDepends-On: https://review.openstack.org/549306\nChange-Id: I7180e20aa675c0ccbe3718493a6eff2be7e1b8b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-client-config/commit/73f037e38213b6944eeb8e616afe71c989f994cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-client-config/commit/73f037e38213b6944eeb8e616afe71c989f994cd"}]},"branch":"refs/heads/master"},"f1754f78f479d3e18e0875b57ae52dbb2ce6e676":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2018-03-03 14:38:07.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/07/549307/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-client-config","ref":"refs/changes/07/549307/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-client-config refs/changes/07/549307/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-client-config refs/changes/07/549307/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-client-config refs/changes/07/549307/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-client-config refs/changes/07/549307/2"}}},"commit":{"parents":[{"commit":"c842c06d8cbc564985ae33d2bbb53dd79b5a40c1","subject":"Add os-client-config tips jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-client-config/commit/c842c06d8cbc564985ae33d2bbb53dd79b5a40c1"}]}],"author":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2018-03-02 09:42:37.000000000","tz":60},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2018-03-03 14:36:40.000000000","tz":-360},"subject":"Replace guts with openstack.config","message":"Replace guts with openstack.config\n\nThe code has been moved into openstacksdk. Make os-client-config a\nthin wrapper around that code.\n\nThe legacy client code needs to stay, as it was removed in openstacksdk.\n\nWe also have to plumb os_client_config.defaults and\nos_client_config.cloud_region.CloudRegion in so that users get the\nos-client-config versions from the openstacksdk methods that return the\nrelevant objects.\n\nSupport for per-service region_name has been dropped. This was never\nused anywhere and was added for compatibility with openstacksdk Profile\nobjects. Since Profile objects are no more, we can remove it. It never\nmade any sense.\n\nDepends-On: https://review.openstack.org/549306\nChange-Id: I7180e20aa675c0ccbe3718493a6eff2be7e1b8b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-client-config/commit/f1754f78f479d3e18e0875b57ae52dbb2ce6e676"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-client-config/commit/f1754f78f479d3e18e0875b57ae52dbb2ce6e676"}]},"branch":"refs/heads/master"},"0bb4d445fa5e9b1443ba908dbd108916627fe93f":{"kind":"REWORK","_number":3,"created":"2018-03-10 14:11:21.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/07/549307/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-client-config","ref":"refs/changes/07/549307/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-client-config refs/changes/07/549307/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-client-config refs/changes/07/549307/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-client-config refs/changes/07/549307/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-client-config refs/changes/07/549307/3"}}},"commit":{"parents":[{"commit":"34e2b565046903d70bcdba56f0c93330f0437a93","subject":"Merge \"Add os-client-config tips jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-client-config/commit/34e2b565046903d70bcdba56f0c93330f0437a93"}]}],"author":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2018-03-02 09:42:37.000000000","tz":60},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2018-03-10 14:11:17.000000000","tz":-360},"subject":"Replace guts with openstack.config","message":"Replace guts with openstack.config\n\nThe code has been moved into openstacksdk. Make os-client-config a\nthin wrapper around that code.\n\nThe legacy client code needs to stay, as it was removed in openstacksdk.\n\nWe also have to plumb os_client_config.defaults and\nos_client_config.cloud_region.CloudRegion in so that users get the\nos-client-config versions from the openstacksdk methods that return the\nrelevant objects.\n\nSupport for per-service region_name has been dropped. This was never\nused anywhere and was added for compatibility with openstacksdk Profile\nobjects. Since Profile objects are no more, we can remove it. It never\nmade any sense.\n\nDepends-On: https://review.openstack.org/549306\nChange-Id: I7180e20aa675c0ccbe3718493a6eff2be7e1b8b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-client-config/commit/0bb4d445fa5e9b1443ba908dbd108916627fe93f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-client-config/commit/0bb4d445fa5e9b1443ba908dbd108916627fe93f"}]},"branch":"refs/heads/master"},"3d23f5dd7b31378cca9a1141356fe0d52063481a":{"kind":"REWORK","_number":4,"created":"2018-03-29 16:08:40.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/07/549307/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-client-config","ref":"refs/changes/07/549307/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-client-config refs/changes/07/549307/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-client-config refs/changes/07/549307/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-client-config refs/changes/07/549307/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-client-config refs/changes/07/549307/4"}}},"commit":{"parents":[{"commit":"e86641fd0b45b76a40bd459fff7d977cfa59004e","subject":"pypy is not checked at gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-client-config/commit/e86641fd0b45b76a40bd459fff7d977cfa59004e"}]}],"author":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2018-03-02 09:42:37.000000000","tz":60},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2018-03-29 16:08:29.000000000","tz":-300},"subject":"Replace guts with openstack.config","message":"Replace guts with openstack.config\n\nThe code has been moved into openstacksdk. Make os-client-config a\nthin wrapper around that code.\n\nThe legacy client code needs to stay, as it was removed in openstacksdk.\n\nWe also have to plumb os_client_config.defaults and\nos_client_config.cloud_region.CloudRegion in so that users get the\nos-client-config versions from the openstacksdk methods that return the\nrelevant objects.\n\nSupport for per-service region_name has been dropped. This was never\nused anywhere and was added for compatibility with openstacksdk Profile\nobjects. Since Profile objects are no more, we can remove it. It never\nmade any sense.\n\nDepends-On: https://review.openstack.org/549306\nChange-Id: I7180e20aa675c0ccbe3718493a6eff2be7e1b8b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-client-config/commit/3d23f5dd7b31378cca9a1141356fe0d52063481a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-client-config/commit/3d23f5dd7b31378cca9a1141356fe0d52063481a"}]},"branch":"refs/heads/master"},"e77dbc85c1a89ef3ce11baba32760e8f2c1a4cc9":{"kind":"REWORK","_number":5,"created":"2018-04-27 11:04:28.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/07/549307/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-client-config","ref":"refs/changes/07/549307/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-client-config refs/changes/07/549307/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-client-config refs/changes/07/549307/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-client-config refs/changes/07/549307/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-client-config refs/changes/07/549307/5"}}},"commit":{"parents":[{"commit":"cf5f446f27a08a41366847e2b586c51f9a6d696f","subject":"Remove DataCentred from list of vendors","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-client-config/commit/cf5f446f27a08a41366847e2b586c51f9a6d696f"}]}],"author":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2018-03-02 09:42:37.000000000","tz":60},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2018-04-27 11:03:40.000000000","tz":-300},"subject":"Replace guts with openstack.config","message":"Replace guts with openstack.config\n\nThe code has been moved into openstacksdk. Make os-client-config a\nthin wrapper around that code.\n\nThe legacy client code needs to stay, as it was removed in openstacksdk.\n\nWe also have to plumb os_client_config.defaults and\nos_client_config.cloud_region.CloudRegion in so that users get the\nos-client-config versions from the openstacksdk methods that return the\nrelevant objects.\n\nSupport for per-service region_name has been dropped. This was never\nused anywhere and was added for compatibility with openstacksdk Profile\nobjects. Since Profile objects are no more, we can remove it. It never\nmade any sense.\n\nChange-Id: I7180e20aa675c0ccbe3718493a6eff2be7e1b8b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-client-config/commit/e77dbc85c1a89ef3ce11baba32760e8f2c1a4cc9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-client-config/commit/e77dbc85c1a89ef3ce11baba32760e8f2c1a4cc9"}]},"branch":"refs/heads/master"},"5c2eae8f8242332aff59d18b45a7d17f01524aaa":{"kind":"REWORK","_number":6,"created":"2018-04-27 13:44:50.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/07/549307/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-client-config","ref":"refs/changes/07/549307/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-client-config refs/changes/07/549307/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-client-config refs/changes/07/549307/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-client-config refs/changes/07/549307/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-client-config refs/changes/07/549307/6"}}},"commit":{"parents":[{"commit":"cf5f446f27a08a41366847e2b586c51f9a6d696f","subject":"Remove DataCentred from list of vendors","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-client-config/commit/cf5f446f27a08a41366847e2b586c51f9a6d696f"}]}],"author":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2018-03-02 09:42:37.000000000","tz":60},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2018-04-27 13:42:17.000000000","tz":-300},"subject":"Replace guts with openstack.config","message":"Replace guts with openstack.config\n\nThe code has been moved into openstacksdk. Make os-client-config a\nthin wrapper around that code.\n\nThe legacy client code needs to stay, as it was removed in openstacksdk.\n\nWe also have to plumb os_client_config.defaults and\nos_client_config.cloud_region.CloudRegion in so that users get the\nos-client-config versions from the openstacksdk methods that return the\nrelevant objects.\n\nSupport for per-service region_name has been dropped. This was never\nused anywhere and was added for compatibility with openstacksdk Profile\nobjects. Since Profile objects are no more, we can remove it. It never\nmade any sense.\n\nChange-Id: I7180e20aa675c0ccbe3718493a6eff2be7e1b8b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-client-config/commit/5c2eae8f8242332aff59d18b45a7d17f01524aaa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-client-config/commit/5c2eae8f8242332aff59d18b45a7d17f01524aaa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
