)]}'
{"id":"openstack%2Fos-client-config~686589","triplet_id":"openstack%2Fos-client-config~master~If1fe0e7d8897e56ca485157dcb75bea7b1d77e47","project":"openstack/os-client-config","branch":"master","hashtags":[],"change_id":"If1fe0e7d8897e56ca485157dcb75bea7b1d77e47","subject":"Stop testing python2.7","status":"MERGED","created":"2019-10-04 12:39:21.000000000","updated":"2019-10-14 21:13:18.000000000","submitted":"2019-10-14 21:10:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"686589-1571087409522-9bbf562d","meta_rev_id":"514d6935af5b7f4d018af005ff7b3732de49005a","_number":686589,"virtual_id_number":686589,"owner":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-10-04 14:44:12.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":3099,"name":"David Shrewsbury","email":"dshrewsb@redhat.com","username":"dshrews"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-14 21:10:09.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3099,"name":"David Shrewsbury","email":"dshrewsb@redhat.com","username":"dshrews"},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":2,"date":"2019-10-14 19:40:43.000000000","_account_id":3099,"name":"David Shrewsbury","email":"dshrewsb@redhat.com","username":"dshrews"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3099,"name":"David Shrewsbury","email":"dshrewsb@redhat.com","username":"dshrews"},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":1,"date":"2019-10-14 19:40:43.000000000","_account_id":3099,"name":"David Shrewsbury","email":"dshrewsb@redhat.com","username":"dshrews"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Allow-Post-Review":{"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":3099,"name":"David Shrewsbury","email":"dshrewsb@redhat.com","username":"dshrews"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Abstain","+1":"Approved for Zuul Post Review Pipeline"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":3099,"name":"David Shrewsbury","email":"dshrewsb@redhat.com","username":"dshrews"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-14 19:40:43.000000000","updated_by":{"_account_id":3099,"name":"David Shrewsbury","email":"dshrewsb@redhat.com","username":"dshrews"},"reviewer":{"_account_id":3099,"name":"David Shrewsbury","email":"dshrewsb@redhat.com","username":"dshrews"},"state":"REVIEWER"},{"updated":"2019-10-14 21:10:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dbe6be0d0261ef5b8629ca9543817c5c55f699c4","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2019-10-04 12:39:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c0545208b83a9a60194f17ac9deaddcbe8eec923","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-04 14:03:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b1f9f315e94e4989a20f910a8011081c : SUCCESS in 1m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/694b9ace4b7543b3b88fc9245b2323de : SUCCESS in 1h 22m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/00aa2fb26237447a9b0d810e286c4a12 : SUCCESS in 3m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4e7413bc29da4d01bc82b289c38d91f5 : SUCCESS in 2m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/958ce589a9a14ce2bea0282f9db87282 : SUCCESS in 3m 23s\n- os-client-config-tox-py37-tips https://zuul.opendev.org/t/openstack/build/7476e68bf0f24a4286768bef092cf32c : SUCCESS in 3m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12510642ee2f42eaa908492fdfe8a1db : SUCCESS in 3m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6cf7c4056f364f1ea86489dfaf1359d3 : SUCCESS in 3m 08s\n- shade-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/0cd873e3b5424e109e9eca516905ff5d : SUCCESS in 44m 32s\n- shade-functional-devstack-tips-python3 https://zuul.opendev.org/t/openstack/build/b6b0aa00833c47acab6c4f7fb7f19086 : SUCCESS in 45m 55s\n- shade-tox-py27-tips https://zuul.opendev.org/t/openstack/build/4a437296b50d41528fec2cca3df9abd4 : SUCCESS in 3m 35s\n- shade-tox-py35-tips https://zuul.opendev.org/t/openstack/build/2c2e60c77ae04acaaf898d223feeaa0f : SUCCESS in 3m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85079efd71dd45ce8ecf3063be0649d2 : SUCCESS in 2m 59s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/de578de9f9c14ca9aa8f03b80e54b509 : FAILURE in 49m 33s","accounts_in_message":[],"_revision_number":1},{"id":"f08b62c1c0bdd25d2a8fc3801c37359369a34d90","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2019-10-04 14:44:12.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b6264a9efa95226eeaadfdcf7ca1dc90bc0e7d1d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-04 16:30:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/02b0ec60aa95467b92805b60649441dc : SUCCESS in 3m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8f2b64d924124ec3ab951caafdbf92bc : SUCCESS in 1h 39m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a3f4d3563b664d1d9779fd28ea64a4f3 : SUCCESS in 3m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e7daf55c402e47aaa5efee6f956acb93 : SUCCESS in 3m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/56cee2039889490492efebc19248edc9 : SUCCESS in 3m 35s\n- os-client-config-tox-py37-tips https://zuul.opendev.org/t/openstack/build/9c547fc7ce3e4b12896d3dd6875bf8d2 : SUCCESS in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ffc0db23c341421ca8cd57599a514d24 : SUCCESS in 3m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/322cc45a703b428d9500f29c3ad03291 : SUCCESS in 3m 47s\n- shade-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/a288792c2e76434c86e93e2e27e24620 : SUCCESS in 46m 21s\n- shade-functional-devstack-tips-python3 https://zuul.opendev.org/t/openstack/build/2f4ba5f63b814498ab84cb71ea33a28c : SUCCESS in 48m 09s\n- shade-tox-py27-tips https://zuul.opendev.org/t/openstack/build/3672c66436984520bcea352ade383fa0 : SUCCESS in 4m 16s\n- shade-tox-py35-tips https://zuul.opendev.org/t/openstack/build/d9d52340d4544bdeaa47e5d30a526d10 : SUCCESS in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/535f7ed0d0c7431ba09d54889047e83f : SUCCESS in 2m 46s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/b306b7600ad64ff0b228ff1a801b97af : SUCCESS in 1h 02m 33s","accounts_in_message":[],"_revision_number":1},{"id":"fb96968262547344106365e9b63aa0d151c5aad6","author":{"_account_id":3099,"name":"David Shrewsbury","email":"dshrewsb@redhat.com","username":"dshrews"},"date":"2019-10-14 19:40:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"241bad90a8581e16e280c8a894a8ec385b535b3e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-14 19:41:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cc2c3e20baa3de4415ab2565ffd619ece24604e9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-14 21:10:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"cec2a230cce116439e62f96f3738e9dc9ab68522","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-14 21:10:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/322d6fccacde402988dd739c4e28a123 : SUCCESS in 1m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fb8f359f83c24480869048f5cd160f87 : SUCCESS in 1h 25m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6e8ad13624b5479bba313b5c123f7e27 : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/08a743a866344391af75f94f900f753e : SUCCESS in 4m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2e9e39ec78124f64a7a7474913941efd : SUCCESS in 6m 46s\n- os-client-config-tox-py37-tips https://zuul.opendev.org/t/openstack/build/396ffefb8e3846d8832c94116e422091 : SUCCESS in 7m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ea8cf4c0ff7446195a07ca70dc297ff : SUCCESS in 3m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/51f4b170ac81438a93750cc08d7f5cf5 : SUCCESS in 3m 57s\n- shade-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/eea39187a2aa4beabb0a9b90b3122be4 : SUCCESS in 47m 18s\n- shade-functional-devstack-tips-python3 https://zuul.opendev.org/t/openstack/build/0b4581c564624cd2bd7e3d0f5d44e068 : SUCCESS in 58m 30s\n- shade-tox-py27-tips https://zuul.opendev.org/t/openstack/build/c08d92559ff64fb5bde352eff7f10428 : SUCCESS in 5m 34s\n- shade-tox-py35-tips https://zuul.opendev.org/t/openstack/build/fa9c65dcc08042699f88780e8fe28ac3 : SUCCESS in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d75737910160438fb90e99bd33b5a3e0 : SUCCESS in 2m 50s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/cce666ac879048abb29f94a7c18f07b8 : SUCCESS in 57m 27s","accounts_in_message":[],"_revision_number":1},{"id":"514d6935af5b7f4d018af005ff7b3732de49005a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-14 21:13:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f163e4f4f936414fba155e3cd3453ae3 : SUCCESS in 58s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dfb1cf7f20914879854e9d3460675dac : SUCCESS in 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cee637ad9c8bc1f19d785e658d80d1a9f106981f","revisions":{"cee637ad9c8bc1f19d785e658d80d1a9f106981f":{"kind":"REWORK","_number":1,"created":"2019-10-04 12:39:21.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/89/686589/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-client-config","ref":"refs/changes/89/686589/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-client-config refs/changes/89/686589/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-client-config refs/changes/89/686589/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-client-config refs/changes/89/686589/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-client-config refs/changes/89/686589/1"}}},"commit":{"parents":[{"commit":"33823db9593307ad1c6c0ce25e43ad8ff4f3a0ba","subject":"Update master for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-client-config/commit/33823db9593307ad1c6c0ce25e43ad8ff4f3a0ba"}]}],"author":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2019-10-04 12:38:20.000000000","tz":120},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2019-10-04 12:39:19.000000000","tz":120},"subject":"Stop testing python2.7","message":"Stop testing python2.7\n\nTrain was the last release supporting python2. Let\u0027s rip the bandaid off.\n\nChange-Id: If1fe0e7d8897e56ca485157dcb75bea7b1d77e47\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-client-config/commit/cee637ad9c8bc1f19d785e658d80d1a9f106981f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-client-config/commit/cee637ad9c8bc1f19d785e658d80d1a9f106981f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
