)]}'
{"id":"openstack%2Fos-client-config~713029","triplet_id":"openstack%2Fos-client-config~master~Ieb1746b1ad7571cb87f5b1e7a6876ce72e6b76c3","project":"openstack/os-client-config","branch":"master","topic":"unittest.mock","hashtags":[],"change_id":"Ieb1746b1ad7571cb87f5b1e7a6876ce72e6b76c3","subject":"Use unittest.mock instead of third party mock","status":"MERGED","created":"2020-03-13 16:54:24.000000000","updated":"2020-03-13 22:13:48.000000000","submitted":"2020-03-13 22:12:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"713029-1584137552181-7e8b974d","meta_rev_id":"76fbcb51e4d989583c7d14678b2df22200eb15f8","_number":713029,"virtual_id_number":713029,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-13 22:12:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-03-13 19:36:43.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-03-13 19:36:43.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Allow-Post-Review":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Abstain","+1":"Approved for Zuul Post Review Pipeline"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-13 19:36:43.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2020-03-13 22:12:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dd16a92a2be98fb3a3052f1e6091473167efff59","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-03-13 16:54:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"14080a0cc8dee75b2933e448145b9a6a718f7503","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-03-13 17:06:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b3e3b323c350a4259241979a39c348b2cf1b86d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-13 18:36:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/372a6f230cc34836af1348c0c283ccb6 : FAILURE in 2m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b8c0d121e6d04097b74f53478a7bd9f8 : SUCCESS in 1h 40m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c1f01451f5f24740b147685016f50e87 : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/217fc5af7651412dad7704f81e32e95e : SUCCESS in 3m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e5d081ebb29042f08343f3c29dfe74d2 : SUCCESS in 3m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5af24b2524544c04bf56265674817d7b : SUCCESS in 4m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9f95e75c86d740a885934c399ba092aa : SUCCESS in 4m 08s (non-voting)\n- os-client-config-tox-py37-tips https://zuul.opendev.org/t/openstack/build/c6227127372b4dae83cc9ba18232f57a : SUCCESS in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87a4b38b8b15407fbebbe121d5928f3e : SUCCESS in 3m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/59ad1c958494431f9eae42ba9e9f4a39 : SUCCESS in 5m 21s\n- shade-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/4aca642ac0904c4c8cafb69d0d0c7efd : SUCCESS in 50m 37s\n- shade-functional-devstack-tips-python3 https://zuul.opendev.org/t/openstack/build/61571b31f5234d30907514f93548d1ef : SUCCESS in 53m 54s\n- shade-tox-py27-tips https://zuul.opendev.org/t/openstack/build/e07624613ad746d084649782bcc1000c : SUCCESS in 5m 06s\n- shade-tox-py35-tips https://zuul.opendev.org/t/openstack/build/9f6c57c86f9f41ebbbac8480c3b5f116 : SUCCESS in 4m 23s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/e4c8ab00549f41c78edd1dbf717ebbdc : FAILURE in 51m 56s","accounts_in_message":[],"_revision_number":1},{"id":"40564610561ded0ab5b6bc3ac3e5150bb521ba8a","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-03-13 18:39:25.000000000","message":"Patch Set 1:\n\nHmm, appears to be an unrelated issue we need to address first:\n\n\"ERROR: Requirement for package jsonschema excludes a version not excluded in the global list.\"","accounts_in_message":[],"_revision_number":1},{"id":"dce89e2819468665fc6f209e348e2fc65aec0071","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-03-13 18:51:29.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"219e7d6233f5eaa40d42fec831c21ae4ff93ad90","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-03-13 19:36:43.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"074af9222633af74a33af104d27ca314ca91909d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-13 20:14:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/39acb720bf94481db611c0a7df057d6b : SUCCESS in 3m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c18bb4cde89e4548bebc368645c844a6 : SUCCESS in 1h 14m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7c8f94c091de4b3da628a49adacf9981 : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c474fa1656ad4d4090b14c3988247587 : SUCCESS in 2m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d79db0ebd9e14434bbea8c63d1745cdd : SUCCESS in 4m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/780f387b072d4a5fbd290742da65d0ba : SUCCESS in 5m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8c0db04a7b1a4e5ea6dad829f0c7894b : SUCCESS in 6m 17s (non-voting)\n- os-client-config-tox-py37-tips https://zuul.opendev.org/t/openstack/build/633fd4c222844d4faeccd6664ce7f2c8 : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc4dd4bd04334bfbacf12d5a3fce6b26 : SUCCESS in 3m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b5baea5180f346a89b6b8036e64e233a : SUCCESS in 4m 26s\n- shade-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/6e507f319c1c4c919c62cb865e5cb981 : SUCCESS in 45m 43s\n- shade-functional-devstack-tips-python3 https://zuul.opendev.org/t/openstack/build/b9fa952ca81c4b418183f37e7890be41 : SUCCESS in 45m 53s\n- shade-tox-py27-tips https://zuul.opendev.org/t/openstack/build/b5b6d0d5279b449d9334190ce860ed2a : SUCCESS in 5m 12s\n- shade-tox-py35-tips https://zuul.opendev.org/t/openstack/build/6f19e4602abf4f1ab43c6b459467a90f : SUCCESS in 5m 52s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/98e17a37937244f6822073ef6d8ad00b : SUCCESS in 53m 16s","accounts_in_message":[],"_revision_number":2},{"id":"829e18078a66cc1a727cb133157b09234cdc8eb8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-13 20:30:00.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e3ae222407f0d2547e7b9d242f154d47d0767738","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-13 22:12:31.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/8e72582a4ead46ca8b67cb4a6a446bff : SUCCESS in 3m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/40110265331c4d83922f973b27361a33 : SUCCESS in 1h 21m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/37dfd42e757948298c5d9161237a7d4a : SUCCESS in 3m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e1f20896a444a41b5dd56e7ba228d95 : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7382f3074b45465f9e8e695de8463bec : SUCCESS in 3m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5e84995acc07419c890842ef54ea9e3b : SUCCESS in 3m 08s\n- os-client-config-tox-py37-tips https://zuul.opendev.org/t/openstack/build/71d2fc79bba1403a811f93d4f4625553 : SUCCESS in 4m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6691c4bab35489ca5f7fc1f59afdbd9 : SUCCESS in 2m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fca11795ce404ebb869c851744f4d49f : SUCCESS in 5m 06s\n- shade-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/f5952472e88f43c98b76dc69a17181f3 : SUCCESS in 44m 50s\n- shade-functional-devstack-tips-python3 https://zuul.opendev.org/t/openstack/build/c709235fc36443b8ae061a053ff940f8 : SUCCESS in 47m 01s\n- shade-tox-py27-tips https://zuul.opendev.org/t/openstack/build/538b99786f944347bec1949c4d0bd31e : SUCCESS in 5m 19s\n- shade-tox-py35-tips https://zuul.opendev.org/t/openstack/build/9da51c38ff9946a89ec7eff9aa167613 : SUCCESS in 4m 57s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/de2b6d8cccd34bef890e2524b6975ad4 : SUCCESS in 1h 40m 27s","accounts_in_message":[],"_revision_number":2},{"id":"0dc6e69dc974109c103d0290494f164802f74035","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-13 22:12:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"ba88574455b2703d677da4f0887d28aa0dfded6b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-13 22:13:48.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e58b78d02de44d3b92b5b471500c5108 : SUCCESS in 1m 01s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fab3d40a17dc40ae86628ae9bb364729 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"bc96c23a9c44cee8f54d7349b288a406cec3f10a","revisions":{"dd229544dbe180a27cb5a0c3ae25a7cda9d69448":{"kind":"REWORK","_number":1,"created":"2020-03-13 16:54:24.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/29/713029/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-client-config","ref":"refs/changes/29/713029/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-client-config refs/changes/29/713029/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-client-config refs/changes/29/713029/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-client-config refs/changes/29/713029/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-client-config refs/changes/29/713029/1"}}},"commit":{"parents":[{"commit":"efd56a806e5d1ffad3dc8f31116c7e045b81ad04","subject":"Add a release note for Python 2 support removal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-client-config/commit/efd56a806e5d1ffad3dc8f31116c7e045b81ad04"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-03-13 16:53:51.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-03-13 16:53:51.000000000","tz":-300},"subject":"Use unittest.mock instead of third party mock","message":"Use unittest.mock instead of third party mock\n\nNow that we no longer support py27, we can use the standard library\nunittest.mock module instead of the third party mock lib.\n\nChange-Id: Ieb1746b1ad7571cb87f5b1e7a6876ce72e6b76c3\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-client-config/commit/dd229544dbe180a27cb5a0c3ae25a7cda9d69448"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-client-config/commit/dd229544dbe180a27cb5a0c3ae25a7cda9d69448"}]},"branch":"refs/heads/master"},"bc96c23a9c44cee8f54d7349b288a406cec3f10a":{"kind":"REWORK","_number":2,"created":"2020-03-13 18:51:29.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/29/713029/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-client-config","ref":"refs/changes/29/713029/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-client-config refs/changes/29/713029/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-client-config refs/changes/29/713029/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-client-config refs/changes/29/713029/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-client-config refs/changes/29/713029/2"}}},"commit":{"parents":[{"commit":"c7bb5d65abb4c9bc3a1d8f480018a9d661f7b36e","subject":"Uncap jsonschema","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-client-config/commit/c7bb5d65abb4c9bc3a1d8f480018a9d661f7b36e"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-03-13 16:53:51.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-03-13 18:50:53.000000000","tz":-300},"subject":"Use unittest.mock instead of third party mock","message":"Use unittest.mock instead of third party mock\n\nNow that we no longer support py27, we can use the standard library\nunittest.mock module instead of the third party mock lib.\n\nChange-Id: Ieb1746b1ad7571cb87f5b1e7a6876ce72e6b76c3\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-client-config/commit/bc96c23a9c44cee8f54d7349b288a406cec3f10a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-client-config/commit/bc96c23a9c44cee8f54d7349b288a406cec3f10a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
