)]}'
{"id":"openstack%2Fos-client-config~717095","triplet_id":"openstack%2Fos-client-config~stable%2Fussuri~Ib3cc26205e2c5c329b7eb439f95484913fe42185","project":"openstack/os-client-config","branch":"stable/ussuri","topic":"create-ussuri","hashtags":[],"change_id":"Ib3cc26205e2c5c329b7eb439f95484913fe42185","subject":"Update .gitreview for stable/ussuri","status":"MERGED","created":"2020-04-02 18:17:26.000000000","updated":"2020-04-04 14:49:56.000000000","submitted":"2020-04-04 14:48:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"717095-1586011723015-6ca8d248","meta_rev_id":"ed7ec115521616c3cabedf8fa3e6bc3ca21dd25c","_number":717095,"virtual_id_number":717095,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-04 14:48:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":2,"date":"2020-04-04 14:42:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":1,"date":"2020-04-04 14:42:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Allow-Post-Review":{"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Abstain","+1":"Approved for Zuul Post Review Pipeline"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-04 14:42:40.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2020-04-04 14:48:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7d5313bd9d41b8d4cc1a8e49a93f3e803c4d5d8c","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2020-04-02 18:17:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1c21b46519ce3dd8344cc3892ee4ba136d0612ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 18:42:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7d3cfb576fb14204aeaa88bd5dccf2da : SUCCESS in 6m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4fabdc40a0a943738334805d4f4adbbb : SUCCESS in 6m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/af0859c7842d428d9e991a62fc78309f : SUCCESS in 6m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/97ac9a69dd094879a2bf6c9a6747b033 : SUCCESS in 5m 51s\n- os-client-config-tox-py37-tips https://zuul.opendev.org/t/openstack/build/189544520a114bbfaa54196a447dd94a : SUCCESS in 6m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c48864760fa54632b0b76a72c104996a : SUCCESS in 6m 53s","accounts_in_message":[],"_revision_number":1},{"id":"ffe5153feaffc4bdce6ac0743ed01328a3fbda39","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-04-04 14:42:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f34582c1ca18088af9f128ffb21a84c9d3148448","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-04 14:42:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2c7b7fc5b215414d52cc58afb26da7b91f516b8f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-04 14:48:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d850719d30bd414b9871da411e621b89 : SUCCESS in 3m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2336d096ccd4d04a86679a8de8bb924 : SUCCESS in 3m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1d911faf35634e7ebe04f2a8f31aec69 : SUCCESS in 3m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bae281bd57b443c7a542ee69e4a947e0 : SUCCESS in 4m 29s\n- os-client-config-tox-py37-tips https://zuul.opendev.org/t/openstack/build/bc6d61e4c5144f19a3d2967e62d4dee5 : SUCCESS in 3m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21215510c74a4a689fedc2b0cd5b6ea7 : SUCCESS in 4m 04s\n\nWarning:\n  Job openstack-tox-lower-constraints: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/lower-constraints/lib/python3.6/site-packages/setuptools/dist.py in the commit\u0027\n  Job openstack-tox-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/openstack/config/loader.py in the commit\u0027\n  Job os-client-config-tox-py37-tips: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py37/lib/python3.6/site-packages/openstack/config/loader.py in the commit\u0027\n  Job openstack-tox-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py37/lib/python3.6/site-packages/openstack/config/loader.py in the commit\u0027\n  Comments left for invalid file .tox/lower-constraints/lib/python3.6/site-packages/setuptools/dist.py\n  Comments left for invalid file .tox/py36/lib/python3.6/site-packages/openstack/config/loader.py\n  Comments left for invalid file .tox/py37/lib/python3.6/site-packages/openstack/config/loader.py","accounts_in_message":[],"_revision_number":1},{"id":"f5a64b3badb1b8ebda7d2d9355d47cf21f95d65e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-04 14:48:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"58b70d38274f6b2d000a19c9cca383677d25afbb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-04 14:49:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad88a618095f41c9b477e2125389803c : SUCCESS in 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ea9da3a333b9acafb58394a7ce8809f8963f5c44","revisions":{"ea9da3a333b9acafb58394a7ce8809f8963f5c44":{"kind":"REWORK","_number":1,"created":"2020-04-02 18:17:26.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/95/717095/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-client-config","ref":"refs/changes/95/717095/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-client-config refs/changes/95/717095/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-client-config refs/changes/95/717095/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-client-config refs/changes/95/717095/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-client-config refs/changes/95/717095/1"}}},"commit":{"parents":[{"commit":"bc96c23a9c44cee8f54d7349b288a406cec3f10a","subject":"Use unittest.mock instead of third party mock","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-client-config/commit/bc96c23a9c44cee8f54d7349b288a406cec3f10a"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-04-02 18:17:24.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-04-02 18:17:24.000000000","tz":0},"subject":"Update .gitreview for stable/ussuri","message":"Update .gitreview for stable/ussuri\n\nChange-Id: Ib3cc26205e2c5c329b7eb439f95484913fe42185\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-client-config/commit/ea9da3a333b9acafb58394a7ce8809f8963f5c44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-client-config/commit/ea9da3a333b9acafb58394a7ce8809f8963f5c44"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
