)]}'
{"id":"openstack%2Fos-client-config~717096","triplet_id":"openstack%2Fos-client-config~stable%2Fussuri~Ibb435d31c3223ecde33822db604ff06f32ab495e","project":"openstack/os-client-config","branch":"stable/ussuri","topic":"create-ussuri","hashtags":[],"change_id":"Ibb435d31c3223ecde33822db604ff06f32ab495e","subject":"Update TOX/UPPER_CONSTRAINTS_FILE for stable/ussuri","status":"MERGED","created":"2020-04-02 18:17:28.000000000","updated":"2020-04-04 14:52:58.000000000","submitted":"2020-04-04 14:51:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"717096-1586011881385-cde04409","meta_rev_id":"a083e30f532f710998d6679113b971258e79552c","_number":717096,"virtual_id_number":717096,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-04 14:51:21.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":2,"date":"2020-04-04 14:43:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":1,"date":"2020-04-04 14:43:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Allow-Post-Review":{"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Abstain","+1":"Approved for Zuul Post Review Pipeline"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-04 14:43:09.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2020-04-04 14:51:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1b4e2b4e02a737204649ad97e63a5910de485a49","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2020-04-02 18:17:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"484e790908d4a3afb9d3ab5764c663b0d58bb721","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 18:43:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fd9f604dca8149c19316d3572ed3a1f5 : SUCCESS in 5m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af4a7345154e4764ac5ce3fe6d39bd7f : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6bf379da809b4de197ec851927a8341b : SUCCESS in 7m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/91a8e7a08ab54912ac8d25b5f08d6458 : SUCCESS in 7m 13s\n- os-client-config-tox-py37-tips https://zuul.opendev.org/t/openstack/build/eb4b6270df4441af8c835ec37d5508ed : SUCCESS in 6m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a4ee2877b9b47249d3d26736985a838 : SUCCESS in 5m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5772c317c079414e88202d8247970a0e : SUCCESS in 6m 32s","accounts_in_message":[],"_revision_number":1},{"id":"bdcd291bd001faa41231d6b2f18bc13165f06cbc","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-04-04 14:43:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fd1936f065c3b45749025f29f2e85c32ffc5f2b6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-04 14:43:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"29490a484fa6182be4d4d6910f24d39608ad50b3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-04 14:51:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4ed6f3584fd24669a85f3f0c686c5c4c : SUCCESS in 3m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3e37977ce9c4d87bfda759329a29c7e : SUCCESS in 2m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bf53a5d6239e4cecbd4093f06ab77f04 : SUCCESS in 2m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0bd862a0fea64a3d91f1e9f724bb77b9 : SUCCESS in 3m 05s\n- os-client-config-tox-py37-tips https://zuul.opendev.org/t/openstack/build/07e2439bb2314ba0bbe12682898f4300 : SUCCESS in 3m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d9a888bebc445608038c7dc75430e1a : SUCCESS in 2m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/47534666675c4d5dbf1f57c658074a58 : SUCCESS in 2m 52s\n\nWarning:\n  Job openstack-tox-py37: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py37/lib/python3.6/site-packages/openstack/config/loader.py in the commit\u0027\n  Job openstack-tox-py36: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py36/lib/python3.6/site-packages/openstack/config/loader.py in the commit\u0027\n  Job openstack-tox-lower-constraints: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/lower-constraints/lib/python3.6/site-packages/setuptools/dist.py in the commit\u0027\n  Job os-client-config-tox-py37-tips: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py37/lib/python3.6/site-packages/openstack/config/loader.py in the commit\u0027\n  Comments left for invalid file .tox/lower-constraints/lib/python3.6/site-packages/setuptools/dist.py\n  Comments left for invalid file .tox/py36/lib/python3.6/site-packages/openstack/config/loader.py\n  Comments left for invalid file .tox/py37/lib/python3.6/site-packages/openstack/config/loader.py","accounts_in_message":[],"_revision_number":1},{"id":"e627e79e0b18775f517ce0a7bb5ac69d540051b5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-04 14:51:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b7cff7de2ddc887beaa0e5bdefe7ca7a075d8f36","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-04 14:52:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/665e2153376e4a348a54e57874fc2488 : SUCCESS in 1m 05s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/26daaca70db54ddbab9cc2aeab1c71c4 : SUCCESS in 1m 22s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7868a144a48a17648f4ceb712932b63c2d76e457","revisions":{"7868a144a48a17648f4ceb712932b63c2d76e457":{"kind":"REWORK","_number":1,"created":"2020-04-02 18:17:28.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/96/717096/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-client-config","ref":"refs/changes/96/717096/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-client-config refs/changes/96/717096/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-client-config refs/changes/96/717096/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-client-config refs/changes/96/717096/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-client-config refs/changes/96/717096/1"}}},"commit":{"parents":[{"commit":"ea9da3a333b9acafb58394a7ce8809f8963f5c44","subject":"Update .gitreview for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-client-config/commit/ea9da3a333b9acafb58394a7ce8809f8963f5c44"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-04-02 18:17:26.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-04-02 18:17:26.000000000","tz":0},"subject":"Update TOX/UPPER_CONSTRAINTS_FILE for stable/ussuri","message":"Update TOX/UPPER_CONSTRAINTS_FILE for stable/ussuri\n\nUpdate the URL to the upper-constraints file to point to the redirect\nrule on releases.openstack.org so that anyone working on this branch\nwill switch to the correct upper-constraints list automatically when\nthe requirements repository branches.\n\nUntil the requirements repository has as stable/ussuri branch, tests will\ncontinue to use the upper-constraints list on master.\n\nChange-Id: Ibb435d31c3223ecde33822db604ff06f32ab495e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-client-config/commit/7868a144a48a17648f4ceb712932b63c2d76e457"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-client-config/commit/7868a144a48a17648f4ceb712932b63c2d76e457"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
