)]}'
{"id":"openstack%2Fos-client-config~717609","triplet_id":"openstack%2Fos-client-config~master~I55b4938edae85aa8108be62c49f0ac7fc6682997","project":"openstack/os-client-config","branch":"master","topic":"cleanup-py27-support","hashtags":[],"change_id":"I55b4938edae85aa8108be62c49f0ac7fc6682997","subject":"Cleanup py27 support","status":"MERGED","created":"2020-04-05 15:41:20.000000000","updated":"2020-04-17 20:43:54.000000000","submitted":"2020-04-17 20:42:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"717609-1587156131984-58e2f1df","meta_rev_id":"0ddb139394ab59302f932552820b764fc36f682b","_number":717609,"virtual_id_number":717609,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"date":"2020-04-17 16:03:33.000000000","_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-17 20:42:11.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-04-17 12:13:36.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-04-17 12:06:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-04-17 12:13:36.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Allow-Post-Review":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{" 0":"Abstain","+1":"Approved for Zuul Post Review Pipeline"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-07 11:35:00.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-04-14 10:22:56.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-04-17 12:06:02.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2020-04-17 12:13:36.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2020-04-17 20:42:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4b0e993a26acebaa9c906b9af9899cd6fd730b77","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-05 15:41:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"054702d9333d81d67cd5245f6ecb0ef8aa6116b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-05 17:04:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ed3f9be044124c6e9fe2e3fe1622e8bd : SUCCESS in 3m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a76ab21de0174e3da85971cb42ebb11f : SUCCESS in 1h 22m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a2a1329969ba4e61ac6a075c05c47cff : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20a20bf5b12e43d4b802bd7dfa436832 : SUCCESS in 3m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/39c3854775494d63a5d14794d7b68bdb : SUCCESS in 3m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/af29bcd1a8bc468b895d4cfe34a6f952 : SUCCESS in 3m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fded85470cd2495ba674df9ef3655c2e : SUCCESS in 3m 49s (non-voting)\n- os-client-config-tox-py37-tips https://zuul.opendev.org/t/openstack/build/ec8278d75ac54762b8330de9679a8363 : SUCCESS in 4m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d611cf86234f4f74ad2dfd703a1b2442 : SUCCESS in 3m 17s\n- shade-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/f1ab60828dfd461f9f2b7e075aebfba4 : FAILURE in 29m 20s\n- shade-functional-devstack-tips-python3 https://zuul.opendev.org/t/openstack/build/2184e6d5a6e2440d848fa8e9621e57a4 : SUCCESS in 44m 47s\n- shade-tox-py27-tips https://zuul.opendev.org/t/openstack/build/52ff5164dbe94fe5b7cf49d5f92e6583 : FAILURE in 3m 24s\n- shade-tox-py35-tips https://zuul.opendev.org/t/openstack/build/a74b913d4ecd4111a731b97049b3af9f : FAILURE in 4m 19s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/911bc734840e43ac8c9a0f40c476a8c2 : SUCCESS in 48m 04s","accounts_in_message":[],"_revision_number":1},{"id":"7b0bbf970af7f783137843f558596099bd51645b","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-06 19:54:23.000000000","message":"Patch Set 1:\n\nLooks like we should at least drop shade-tox-py27-tips.","accounts_in_message":[],"_revision_number":1},{"id":"941d333e3744e543e6c5f20de063db7d98164609","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-07 05:18:16.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"358222df941a1936e859a30ca2263a2dfb00aaba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 06:43:52.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/841d4fed6bd7492cb8c1d8d070c13b07 : SUCCESS in 2m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5aba8d7d8f5945258f98ae656629b90a : SUCCESS in 1h 23m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/115f0a0861594b6a83689046d746a64a : SUCCESS in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4d68118e1c344dc990b7dc2e5f93178 : SUCCESS in 3m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/00199c1138df4c448b5de866b1c4865a : SUCCESS in 3m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e7740f912d974c0fbfae84b30180c8e7 : SUCCESS in 4m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c502cae77e054c038e4eb852bd7eff85 : SUCCESS in 3m 41s (non-voting)\n- os-client-config-tox-py37-tips https://zuul.opendev.org/t/openstack/build/c5748a5d6aab471f93189a4e4493da77 : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2728e3b9b1244392bbf3ff14ff2ff1c8 : SUCCESS in 3m 33s\n- shade-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/32348f7f3fac42269ed7711893d44ab1 : FAILURE in 37m 15s\n- shade-functional-devstack-tips-python3 https://zuul.opendev.org/t/openstack/build/51bc473fe6e842efbe0c5369ac1c1d55 : SUCCESS in 41m 18s\n- shade-tox-py27-tips https://zuul.opendev.org/t/openstack/build/03e8d99bedcd4922984546073e582637 : FAILURE in 4m 19s\n- shade-tox-py35-tips https://zuul.opendev.org/t/openstack/build/892e28c6cf3042169c5cf26f669e09cf : SUCCESS in 3m 49s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/56ca28a229f14b91a5ee37852bf43ab1 : SUCCESS in 55m 52s\n\nWarning:\n  Job shade-tox-py35-tips: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path shade/openstackcloud.py in the commit\u0027\n  Comments left for invalid file shade/openstackcloud.py","accounts_in_message":[],"_revision_number":2},{"id":"a6c9d66d460698d1163b042141d24e3491c91bea","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-07 11:35:00.000000000","message":"Patch Set 2:\n\nProposed https://review.opendev.org/718068 to fix the issue of still running py27 jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7832a9c575d0c9495a542779ba5f9ef52d9515bf","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-07 11:40:38.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"741559dd1b89727096988b353fb982fa9a3ee82d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 12:58:06.000000000","message":"Patch Set 3: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":3},{"id":"4a30b06c5da0b560a88c33c1cb47b17c8e61967b","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-07 12:59:44.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"5441ca6bf3aa3432fa64571e111a9764bd5f20a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 15:06:08.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d7e63ae3399c473dba8b0f9e5a9f3d38 : SUCCESS in 5m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e208e7d63d264d7293049978d9bc4a44 : SUCCESS in 1h 22m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4bbbf9541c2c43b0b2c5be21d65fbf96 : SUCCESS in 5m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb9b5c13316f48bba06d5bef4bf656bb : SUCCESS in 5m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d1908d9008ad4e7fbc90fe048e73809c : SUCCESS in 5m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c33e94b7537a4f27ab2a280cd57fe9a2 : SUCCESS in 5m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1903fb43f1c94e33ba1f226ed2af1725 : SUCCESS in 6m 01s (non-voting)\n- os-client-config-tox-py37-tips https://zuul.opendev.org/t/openstack/build/4efcf506ede247c2bb595bb4782d9137 : SUCCESS in 6m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d5d5ffc5acc4a2eab06a0aacb3de266 : SUCCESS in 4m 13s\n- shade-functional-devstack-tips-python3 https://zuul.opendev.org/t/openstack/build/9bd5744be0824208b65b002f287cde91 : RETRY_LIMIT in 36m 26s\n- shade-tox-py35-tips https://zuul.opendev.org/t/openstack/build/fab1744af01d47fb876af1cc7a7f54de : SUCCESS in 6m 30s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/649b589e472a4924bdd01a23b5ee3a47 : SUCCESS in 53m 48s\n\nWarning:\n  Job shade-tox-py35-tips: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path shade/openstackcloud.py in the commit\u0027\n  Comments left for invalid file shade/openstackcloud.py","accounts_in_message":[],"_revision_number":3},{"id":"b4461534247f17052f1260dfaa7c386f35f30403","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-08 05:30:10.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"b678b5942bd5250a7dea8453638fbdac3f95a1f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 07:41:44.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/8a940d79a1d14100943d9fb28395e40c : SUCCESS in 2m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0ad3df01fd7149b1aebd992c9d6c3ac3 : SUCCESS in 1h 13m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/470c00f0dbc94e9f927a00b882a09aba : SUCCESS in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ba084fcf39541b58ac3084d55bd3bfb : SUCCESS in 2m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f690de5f92cb4578b1776bc7fd6c4c93 : SUCCESS in 3m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/43074e99b41b485290596ad11c9e7544 : SUCCESS in 3m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b0becd956d884e569309412d796db595 : SUCCESS in 3m 33s (non-voting)\n- os-client-config-tox-py37-tips https://zuul.opendev.org/t/openstack/build/aad52fdb191e4c6f865ff1821cb0658c : SUCCESS in 5m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52326a2ea3754460b2ecb91b1dad5928 : SUCCESS in 3m 19s\n- shade-functional-devstack-tips-python3 https://zuul.opendev.org/t/openstack/build/4af4798c312e4e40ad999e91fa2bf1ef : RETRY_LIMIT in 39m 37s\n- shade-tox-py35-tips https://zuul.opendev.org/t/openstack/build/8dab6f62ea774726803747a9db20e43d : SUCCESS in 4m 20s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/1c32535a92f2483a8e6ffc087e13ffa3 : SUCCESS in 51m 25s\n\nWarning:\n  Job shade-tox-py35-tips: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path shade/openstackcloud.py in the commit\u0027\n  Comments left for invalid file shade/openstackcloud.py","accounts_in_message":[],"_revision_number":3},{"id":"0c3fde8cfe7e38b5795d29776088fcc4d5e5cd7a","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-11 19:44:05.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"771621519591262175fe3c0c5a7da0d77ee7a4f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-11 21:03:58.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/8b1adcfeba1c48baad6f6b5af4ca94ca : SUCCESS in 2m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b9d0b38a77794c78a7a6cc68cb424837 : SUCCESS in 1h 18m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/252331f2f214426ba740d9134e66987f : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d95fa104498042e8a559662ac2fa819b : SUCCESS in 2m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dfdc532c746343339fc52bf71e0380c5 : SUCCESS in 3m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c2ccf3d2bfb84c818a909be7f75d296a : SUCCESS in 3m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1b2b2d66b72a4e2ea2519ea557ca2548 : SUCCESS in 4m 12s (non-voting)\n- os-client-config-tox-py37-tips https://zuul.opendev.org/t/openstack/build/e1972d52a51b4d718554eb54070f748f : SUCCESS in 4m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cff0d2f14982493f8d2fdd56b365f200 : SUCCESS in 3m 52s\n- shade-functional-devstack-tips-python3 https://zuul.opendev.org/t/openstack/build/a9767cee25124842b85793d99393113f : SUCCESS in 41m 26s\n- shade-tox-py35-tips https://zuul.opendev.org/t/openstack/build/698e04bbe0de4a91b96578f1b3277e10 : SUCCESS in 3m 40s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/8d638fab8f0747c9b1e6b2882bf08d8b : SUCCESS in 55m 24s","accounts_in_message":[],"_revision_number":3},{"id":"29f0e86f12fafce1f7da3bafd24186fd882b8706","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-04-14 10:22:56.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"df77be04a82cb565bdf975a02cb7d92d8ce6d213","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-14 10:30:22.000000000","message":"Patch Set 3:\n\nshade is running py35 tests","accounts_in_message":[],"_revision_number":3},{"id":"f869f3bd3471b0142126af8e6088999cd00e7396","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-04-17 12:01:46.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a6780c6eae4864a42193c345cf1e26a3df94c8cf","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-17 12:04:05.000000000","message":"Patch Set 3: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c29545e98b8fa3e8c66bb9e1ef7c38c48f806d58","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-17 12:04:26.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"1323d143c0ffdb02a22aaebc7571614cb3586eee","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-04-17 12:06:02.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"bb9068b6ff26844c62d0cac67a1a75f1feb5345b","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-04-17 12:13:36.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"15a9830d6db02019724f641074219c451ad45cdc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-17 13:37:56.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a1bf4650141241028d1954a526307194 : SUCCESS in 2m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0690b5c6c53c4eee809ec0f2809e6b5a : SUCCESS in 1h 23m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/216a054f32eb46628d4e8c5cdd3d4a41 : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/188c0a2e59ec48d9bca733f6a01883d1 : SUCCESS in 4m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c7563063d1a84f0d9ebcee0da59cb6da : SUCCESS in 3m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/35c412c780814892986efd3719f02a16 : SUCCESS in 4m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2305176a72414bbc96755bd5fdbe3b93 : SUCCESS in 4m 48s (non-voting)\n- os-client-config-tox-py37-tips https://zuul.opendev.org/t/openstack/build/53ae7464de1543fbb2fe91cf15a3c9ff : SUCCESS in 4m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89aa61974fdb43e3a338f53f1acec736 : SUCCESS in 7m 26s\n- shade-functional-devstack-tips-python3 https://zuul.opendev.org/t/openstack/build/6bc4f39493f944d5ad1ffe2350755865 : SUCCESS in 45m 23s\n- shade-tox-py35-tips https://zuul.opendev.org/t/openstack/build/d91d923cb4d040f59f1950c95d7b91db : SUCCESS in 6m 12s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/c0c2c6ad9b4f4cb588be7e20b87b4971 : SUCCESS in 49m 57s","accounts_in_message":[],"_revision_number":4},{"id":"ba8542bb296b7ee25fd760db8f058a611e650067","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-17 16:03:33.000000000","message":"Patch Set 4:\n\nrecheck - dependency merged","accounts_in_message":[],"_revision_number":4},{"id":"54657e15d3f501e2d927639baba8aac1ceb1827c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-17 19:23:56.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7ef3f99c2d9e4121add411725b2f3fb2 : SUCCESS in 3m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0db30f4d4b984015b88b5ed5858ba746 : SUCCESS in 1h 32m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0e4e3109b4224d50bb9b6b656e6eb25a : SUCCESS in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/369a4c6549b445afa799cf328f4ebfbd : SUCCESS in 3m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1b8359831cf244a8b59bc21730edfa55 : SUCCESS in 5m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d38af0684f154f618dcedbe3f54af299 : SUCCESS in 4m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7eadc1eaab004431a6a6d806dfe5ca65 : SUCCESS in 5m 32s (non-voting)\n- os-client-config-tox-py37-tips https://zuul.opendev.org/t/openstack/build/d7cfad99b07b487a878cc31d90a6ce5f : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c46769bf644a44a8a86411df0dae6e36 : SUCCESS in 3m 44s\n- shade-functional-devstack-tips-python3 https://zuul.opendev.org/t/openstack/build/e4711f610d7a4346afd9fdc161cddef3 : SUCCESS in 48m 51s\n- shade-tox-py35-tips https://zuul.opendev.org/t/openstack/build/8ea2c6e706994e3dacf2a5999b541ed8 : SUCCESS in 6m 36s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/b3030cbc0f464fda98098bad0225febc : SUCCESS in 53m 25s","accounts_in_message":[],"_revision_number":4},{"id":"7d09d6d8d8ecb54723ea7166d9090ce62ba7ad97","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-17 19:24:08.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"cd0c65057765178ae4d610e973a052264338c847","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-17 20:42:11.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/0994b0034dac43bfa6bb6223e0197c63 : SUCCESS in 3m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/af691a0586ba42ada8643933908f9e96 : SUCCESS in 1h 15m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6e41afafa9db4a3d9e0601f32f565886 : SUCCESS in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55ac3a9ff4974de78bd61d798958c7b5 : SUCCESS in 3m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a7aed767872348328ad74f5e5516bdcc : SUCCESS in 3m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f413df48ce634879826dc02a5a398763 : SUCCESS in 4m 48s\n- os-client-config-tox-py37-tips https://zuul.opendev.org/t/openstack/build/541ac95c1ca946978220a68dd1a8e526 : SUCCESS in 5m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de13ce622e62427a823a56738638b73e : SUCCESS in 4m 56s\n- shade-functional-devstack-tips-python3 https://zuul.opendev.org/t/openstack/build/9d0248f6670a44fb8ab820cdfe0ec22a : SUCCESS in 42m 03s\n- shade-tox-py35-tips https://zuul.opendev.org/t/openstack/build/6f3201adb85a4450b50ba6ee451faec3 : SUCCESS in 5m 18s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/afc04d0eade44a7dbcd8a612b7e632cf : SUCCESS in 53m 01s","accounts_in_message":[],"_revision_number":4},{"id":"747275b5933ae34f307ea91b7c9bce0f0e97a4f3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-17 20:42:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"edff8858b4156c8c5402e31965db8d39713fed1e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-17 20:43:54.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1cd67880ff3744e88153b852109e95a0 : SUCCESS in 1m 22s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"0ef60e6249e0bc41ab3c29fc9cbc86fe17808765","revisions":{"c5fb09ab77de6cf7243c6530f1bbc5a625837dbf":{"kind":"REWORK","_number":1,"created":"2020-04-05 15:41:20.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/09/717609/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-client-config","ref":"refs/changes/09/717609/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-client-config refs/changes/09/717609/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-client-config refs/changes/09/717609/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-client-config refs/changes/09/717609/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-client-config refs/changes/09/717609/1"}}},"commit":{"parents":[{"commit":"fd0db53db1b22ba1df239dc2978c192361dd8244","subject":"Update master for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-client-config/commit/fd0db53db1b22ba1df239dc2978c192361dd8244"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-05 15:40:33.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-05 15:41:00.000000000","tz":120},"subject":"Cleanup py27 support","message":"Cleanup py27 support\n\nMake a few cleanups:\n- Remove python 2.7 stanza from setup.py\n- Add requires on python \u003e\u003d 3.6 to setup.cfg so that pypi and pip\n  know about the requirement\n- Remove obsolete sections from setup.cfg\n- Update classifiers\n\nChange-Id: I55b4938edae85aa8108be62c49f0ac7fc6682997\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-client-config/commit/c5fb09ab77de6cf7243c6530f1bbc5a625837dbf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-client-config/commit/c5fb09ab77de6cf7243c6530f1bbc5a625837dbf"}]},"branch":"refs/heads/master"},"3c25946f4b69daaaa756d9db161c443e9f94a4cc":{"kind":"REWORK","_number":2,"created":"2020-04-07 05:18:16.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/09/717609/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-client-config","ref":"refs/changes/09/717609/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-client-config refs/changes/09/717609/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-client-config refs/changes/09/717609/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-client-config refs/changes/09/717609/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-client-config refs/changes/09/717609/2"}}},"commit":{"parents":[{"commit":"fd0db53db1b22ba1df239dc2978c192361dd8244","subject":"Update master for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-client-config/commit/fd0db53db1b22ba1df239dc2978c192361dd8244"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-05 15:40:33.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-07 05:18:04.000000000","tz":120},"subject":"Cleanup py27 support","message":"Cleanup py27 support\n\nMake a few cleanups:\n- Remove python 2.7 stanza from setup.py\n- Add requires on python \u003e\u003d 3.5 to setup.cfg so that pypi and pip\n  know about the requirement\n- Remove obsolete sections from setup.cfg\n- Update classifiers\n\nChange-Id: I55b4938edae85aa8108be62c49f0ac7fc6682997\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-client-config/commit/3c25946f4b69daaaa756d9db161c443e9f94a4cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-client-config/commit/3c25946f4b69daaaa756d9db161c443e9f94a4cc"}]},"branch":"refs/heads/master"},"286f76019345b5af3fa6dc29e216124f1aa7c015":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-04-07 11:40:38.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/09/717609/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-client-config","ref":"refs/changes/09/717609/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-client-config refs/changes/09/717609/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-client-config refs/changes/09/717609/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-client-config refs/changes/09/717609/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-client-config refs/changes/09/717609/3"}}},"commit":{"parents":[{"commit":"fd0db53db1b22ba1df239dc2978c192361dd8244","subject":"Update master for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-client-config/commit/fd0db53db1b22ba1df239dc2978c192361dd8244"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-05 15:40:33.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","date":"2020-04-07 11:40:35.000000000","tz":0},"subject":"Cleanup py27 support","message":"Cleanup py27 support\n\nMake a few cleanups:\n- Remove python 2.7 stanza from setup.py\n- Add requires on python \u003e\u003d 3.5 to setup.cfg so that pypi and pip\n  know about the requirement\n- Remove obsolete sections from setup.cfg\n- Update classifiers\n\nDepends-On: https://review.opendev.org/718068\nChange-Id: I55b4938edae85aa8108be62c49f0ac7fc6682997\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-client-config/commit/286f76019345b5af3fa6dc29e216124f1aa7c015"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-client-config/commit/286f76019345b5af3fa6dc29e216124f1aa7c015"}]},"branch":"refs/heads/master"},"0ef60e6249e0bc41ab3c29fc9cbc86fe17808765":{"kind":"REWORK","_number":4,"created":"2020-04-17 12:04:26.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/09/717609/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-client-config","ref":"refs/changes/09/717609/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-client-config refs/changes/09/717609/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-client-config refs/changes/09/717609/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-client-config refs/changes/09/717609/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-client-config refs/changes/09/717609/4"}}},"commit":{"parents":[{"commit":"fd0db53db1b22ba1df239dc2978c192361dd8244","subject":"Update master for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-client-config/commit/fd0db53db1b22ba1df239dc2978c192361dd8244"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-05 15:40:33.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","date":"2020-04-17 12:04:22.000000000","tz":0},"subject":"Cleanup py27 support","message":"Cleanup py27 support\n\nMake a few cleanups:\n- Remove python 2.7 stanza from setup.py\n- Add requires on python \u003e\u003d 3.5 to setup.cfg so that pypi and pip\n  know about the requirement\n- Remove obsolete sections from setup.cfg\n- Update classifiers\n\nDepends-On: https://review.opendev.org/718068\nChange-Id: I55b4938edae85aa8108be62c49f0ac7fc6682997\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-client-config/commit/0ef60e6249e0bc41ab3c29fc9cbc86fe17808765"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-client-config/commit/0ef60e6249e0bc41ab3c29fc9cbc86fe17808765"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
