)]}'
{"id":"openstack%2Fos-client-config~770598","triplet_id":"openstack%2Fos-client-config~stable%2Ftrain~Ia6ad2c7e62aebc81386610e9d16c8bb8a162d70f","project":"openstack/os-client-config","branch":"stable/train","topic":"pip-20.3","hashtags":[],"change_id":"Ia6ad2c7e62aebc81386610e9d16c8bb8a162d70f","subject":"Uncap jsonschema","status":"MERGED","created":"2021-01-13 11:19:37.000000000","updated":"2021-01-13 19:45:04.000000000","submitted":"2021-01-13 19:43:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"770598-pip-20.3","meta_rev_id":"4ca124c1613c13eb481249f047394f753c4b1182","_number":770598,"virtual_id_number":770598,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-13 19:43:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":2,"date":"2021-01-13 17:51:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-01-13 17:53:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Allow-Post-Review":{"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{" 0":"Abstain","+1":"Approved for Zuul Post Review Pipeline"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-13 11:19:37.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2021-01-13 11:20:11.000000000","updated_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"REVIEWER"},{"updated":"2021-01-13 12:55:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-13 17:51:30.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2021-01-13 17:53:18.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"}],"messages":[{"id":"fabd8e2c42e0fbef173ef87cc657cc71930689ce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-01-13 11:19:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8eae23c5aff15910c0bdfabc235f251242f4fee2","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2021-01-13 11:20:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2d22a398fdeec8a3488ba8ec974c72b4e621145b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 12:55:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/8afad8018dc943af9fdd0b9435498178 : FAILURE in 3m 24s\n- tempest-full https://zuul.opendev.org/t/openstack/build/109b52e43623421491da78a3e2798dc1 : SUCCESS in 1h 34m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0bdfa6eee25f42b7963c36bc5a3cf99b : SUCCESS in 1h 20m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/62607c99ae5d4df286b0ba33b1191958 : FAILURE in 2m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/443e47e51dc14126b6d1ad8e46ccc6fe : FAILURE in 3m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e60a4b9ac9cb478e90fd216d3a134f0c : SUCCESS in 4m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/be6828299d094ecf9c883f4689a5d1c3 : SUCCESS in 3m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/16ac59152fad4c17a683599b7fd248a5 : SUCCESS in 3m 32s\n- os-client-config-tox-py27-tips https://zuul.opendev.org/t/openstack/build/3b3845f2ca424c1498495828292fb0f6 : RETRY_LIMIT in 2m 16s\n- os-client-config-tox-py37-tips https://zuul.opendev.org/t/openstack/build/e41a470652c94387bc0451e31658ff34 : SUCCESS in 11m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3fe33440c8674aa9adc8e282704ae103 : SUCCESS in 4m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/472e8332e82b45ef8f0e2e849daaadf3 : SUCCESS in 3m 15s\n- shade-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/4c1583cc2eb342359715df4f33b93db9 : SUCCESS in 52m 49s\n- shade-functional-devstack-tips-python3 https://zuul.opendev.org/t/openstack/build/6f48618a91684b16bf2c23c182816d3e : SUCCESS in 46m 05s\n- shade-tox-py27-tips https://zuul.opendev.org/t/openstack/build/79dc4a72d5c14c57a89caf8952d1f1d8 : SUCCESS in 5m 18s\n- shade-tox-py35-tips https://zuul.opendev.org/t/openstack/build/9351dae06a58441f8b1a2eb9b5a12e58 : RETRY_LIMIT in 1m 50s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/76d0ba01c71a40cfa0846ddf3b45bc0b : SUCCESS in 41m 46s\n\nWarning:\n  Comments left for invalid file os_client_config/__init__.py\n  Comments left for invalid file os_client_config/tests/test_cloud_config.py","accounts_in_message":[],"_revision_number":1},{"id":"7cf2ed8c01f4f94e32bddb36932bc72cbac894aa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-01-13 15:58:00.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"dc678614694cfb0a47a6ce66991e55341af57c30","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 17:36:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/71c9c4f116384ebca55795f9d5d1a2cc : SUCCESS in 2m 55s\n- tempest-full https://zuul.opendev.org/t/openstack/build/ba6fcc3ceb3e4191ab323a9727bb98ad : SUCCESS in 1h 28m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/648c65aabe65498d8e7ef33a6c0f5077 : SUCCESS in 1h 18m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9ff4a6673a3f4305a77e67edee80c856 : SUCCESS in 3m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9e23f52a7f441d694238550fe76af49 : SUCCESS in 2m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0f28a4c76df747c7905808dbad37e0f0 : SUCCESS in 11m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/96ed54b9f2c0458c8b86834f3abb3ba9 : SUCCESS in 4m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f248242cacb440fb9b7c045ca7b0a08e : SUCCESS in 4m 16s\n- os-client-config-tox-py27-tips https://zuul.opendev.org/t/openstack/build/7e642e0dfbe34e5d9e70b7ed6541d49f : SUCCESS in 4m 30s\n- os-client-config-tox-py37-tips https://zuul.opendev.org/t/openstack/build/804caf36a3684e65913431ed5af84203 : SUCCESS in 3m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04e69d765ffd4444ad184eba73384984 : SUCCESS in 8m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/70f0c10410754e76912ac502680b11e7 : SUCCESS in 2m 42s\n- shade-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/8da7af5aba6f4a29bd99e1d7ec7a7d27 : SUCCESS in 45m 40s\n- shade-functional-devstack-tips-python3 https://zuul.opendev.org/t/openstack/build/b30ce8b57f3a480bbc163bdab5abf060 : SUCCESS in 45m 14s\n- shade-tox-py27-tips https://zuul.opendev.org/t/openstack/build/25b9b52616b8410cb0c3e8aa1dc628be : SUCCESS in 4m 19s\n- shade-tox-py35-tips https://zuul.opendev.org/t/openstack/build/a6080ea9d6aa488ca1732d1e72a2c695 : SUCCESS in 4m 25s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/10a27d4657574448a3f9412f17a5ae64 : SUCCESS in 1h 00m 13s","accounts_in_message":[],"_revision_number":2},{"id":"73a4b2401b027e9d81971e242e89dc113ded0485","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2021-01-13 17:51:30.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"35e7848d2792d74a6f80a2ed84df0c87658aaf26","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2021-01-13 17:53:18.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"54211ee218b36310692ceb7be9ecd5b71a0c3421","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 17:53:36.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5dedee3c475fb884887aaf75f9ee21cf6be6e919","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 19:43:43.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/76ed85dc368042268bfbd372a23d04d3 : SUCCESS in 3m 09s\n- tempest-full https://zuul.opendev.org/t/openstack/build/8f5c469f3d5846b18f755011e0776abd : SUCCESS in 1h 43m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d28698ea3ead411a8ccc007a0d5e85bb : SUCCESS in 1h 36m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/923c95d2ed5f4c6e85cbfb914b6e23bd : SUCCESS in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1878f468b3c49e9b32445732647d759 : SUCCESS in 2m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2bc690faf871491d9d48f8076798557c : SUCCESS in 6m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1bd600a22c984be09530b11cf10a3d59 : SUCCESS in 5m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/db15a716898345c9b8dec20a45d83b13 : SUCCESS in 7m 28s\n- os-client-config-tox-py27-tips https://zuul.opendev.org/t/openstack/build/563f1040007745f3a87546f4ee6867bd : SUCCESS in 3m 15s\n- os-client-config-tox-py37-tips https://zuul.opendev.org/t/openstack/build/e6b301465d874a6d9d4efacc5e1d13fd : SUCCESS in 4m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/516b01801d38455892806d206603b434 : SUCCESS in 4m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/335f7844110a4c7cb9bd242c97c28715 : SUCCESS in 4m 07s\n- shade-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/480978b54d524746992563a88cefad9c : SUCCESS in 43m 19s\n- shade-functional-devstack-tips-python3 https://zuul.opendev.org/t/openstack/build/7fe2b13ba32c4b7a87ba53ed8a6dffde : SUCCESS in 41m 44s\n- shade-tox-py27-tips https://zuul.opendev.org/t/openstack/build/24a695bca8a6449393d569e9920c4f56 : SUCCESS in 4m 24s\n- shade-tox-py35-tips https://zuul.opendev.org/t/openstack/build/c92cd2ef4a154ccb9aa50f2b4b37cfd9 : SUCCESS in 5m 54s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/e3131629efe24e1c993f2d8b04708aea : SUCCESS in 50m 11s","accounts_in_message":[],"_revision_number":2},{"id":"0b2353c6ebc5d7b8c735b85bb6f7327b9fe7e805","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 19:43:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"4ca124c1613c13eb481249f047394f753c4b1182","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 19:45:04.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/241d0ce74aa94bbfa4f2bdc954a9ff45 : SUCCESS in 51s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2bf89e291884452eb992170b94a8da47 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"68c7bac81d5713a8b9a885db1ab824afaf28813c","revisions":{"d9af7186e9a32ad1880c43413e34add82c605f07":{"kind":"REWORK","_number":1,"created":"2021-01-13 11:19:37.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/98/770598/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-client-config","ref":"refs/changes/98/770598/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-client-config refs/changes/98/770598/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-client-config refs/changes/98/770598/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-client-config refs/changes/98/770598/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-client-config refs/changes/98/770598/1"}}},"commit":{"parents":[{"commit":"248f892c0d97c4ed7f68ec1cf0f3605373c9d994","subject":"Update TOX/UPPER_CONSTRAINTS_FILE for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-client-config/commit/248f892c0d97c4ed7f68ec1cf0f3605373c9d994"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-03-13 18:49:02.000000000","tz":-300},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-01-13 11:19:35.000000000","tz":0},"subject":"Uncap jsonschema","message":"Uncap jsonschema\n\njsonschema was uncapped in the global requirements. We need to uncap it\nlocally now too.\n\nNOTE(stephenfin): This is needed to resolve conflicts with the new\ndependency resolver found in pip 20.3.\n\nChange-Id: Ia6ad2c7e62aebc81386610e9d16c8bb8a162d70f\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n(cherry picked from commit c7bb5d65abb4c9bc3a1d8f480018a9d661f7b36e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-client-config/commit/d9af7186e9a32ad1880c43413e34add82c605f07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-client-config/commit/d9af7186e9a32ad1880c43413e34add82c605f07"}]},"branch":"refs/heads/stable/train"},"68c7bac81d5713a8b9a885db1ab824afaf28813c":{"kind":"REWORK","_number":2,"created":"2021-01-13 15:58:00.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/98/770598/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-client-config","ref":"refs/changes/98/770598/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-client-config refs/changes/98/770598/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-client-config refs/changes/98/770598/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-client-config refs/changes/98/770598/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-client-config refs/changes/98/770598/2"}}},"commit":{"parents":[{"commit":"248f892c0d97c4ed7f68ec1cf0f3605373c9d994","subject":"Update TOX/UPPER_CONSTRAINTS_FILE for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-client-config/commit/248f892c0d97c4ed7f68ec1cf0f3605373c9d994"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-03-13 18:49:02.000000000","tz":-300},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-01-13 15:57:48.000000000","tz":0},"subject":"Uncap jsonschema","message":"Uncap jsonschema\n\njsonschema was uncapped in the global requirements. We need to uncap it\nlocally now too.\n\nNOTE(stephenfin): This is needed to resolve conflicts with the new\ndependency resolver found in pip 20.3. We also need to include the\nfollowing additional requirement-related changes to get things passing:\n\n- doc/requirements.txt\n  Ensure specific versions of Sphinx are used on different Python\n  versions.\n\n- lower-constraints.txt\n  Remove hacking since it\u0027s not necessary\n\n- test-requirements.txt\n  Remove hacking since it\u0027s not necessary\n\n- tox.ini\n  Set a hacking version to avoid installing the latest and greatest\n  versions\n\nChange-Id: Ia6ad2c7e62aebc81386610e9d16c8bb8a162d70f\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n(cherry picked from commit c7bb5d65abb4c9bc3a1d8f480018a9d661f7b36e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-client-config/commit/68c7bac81d5713a8b9a885db1ab824afaf28813c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-client-config/commit/68c7bac81d5713a8b9a885db1ab824afaf28813c"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
