)]}'
{"id":"openstack%2Fos-client-config~791407","triplet_id":"openstack%2Fos-client-config~master~I596f8ec16ce08d6a7476a9acd9a0f5bc79d32a1e","project":"openstack/os-client-config","branch":"master","hashtags":[],"change_id":"I596f8ec16ce08d6a7476a9acd9a0f5bc79d32a1e","subject":"setup.cfg: Replace dashes with underscores","status":"ABANDONED","created":"2021-05-14 08:09:10.000000000","updated":"2021-07-16 08:01:16.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"97f1c668a5398cd434d22f1a324959bde33cdce5","_number":791407,"virtual_id_number":791407,"owner":{"_account_id":31828,"name":"yang yawei","email":"yangyawei@inspur.com","username":"yangyawei"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-05-14 09:33:28.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Allow-Post-Review":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Abstain","+1":"Approved for Zuul Post Review Pipeline"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-14 09:33:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6ff935afddfaaff4d2b2f5905e3c95020e429fbd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31828,"name":"yang yawei","email":"yangyawei@inspur.com","username":"yangyawei"},"date":"2021-05-14 08:09:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5693e7898e2b99ad044864b845faef2eabeaa7e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-14 09:33:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a4ba939d06604985a62eee9bd0419c33 : SUCCESS in 3m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/44c3b8f48d0043988b751f11e4a63906 : SUCCESS in 1h 22m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/22cac5a640fc4c30b6486c6aa69188b4 : FAILURE in 3m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6a9eba1b2a24ec39c1c99190b77af83 : FAILURE in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/042e910a24de426fae62cac59e0e24d2 : SUCCESS in 4m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a31e17005bcf4e67a19795fca5345296 : SUCCESS in 5m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c1f0e43f1fa249f38fbed3b61ffe8531 : SUCCESS in 4m 01s (non-voting)\n- os-client-config-tox-py37-tips https://zuul.opendev.org/t/openstack/build/21420f0ef9154f17908d189ab62bd369 : SUCCESS in 5m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0bd3c7882e3c467c9b9767b40329aff0 : SUCCESS in 4m 13s\n- shade-functional-devstack-tips-python3 https://zuul.opendev.org/t/openstack/build/713ac3fbd24d4963919fcd577ef218b5 : SUCCESS in 39m 39s\n- shade-tox-py38-tips https://zuul.opendev.org/t/openstack/build/8a3157503cf34c3aac69748c08748cf9 : SUCCESS in 5m 51s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/60bb4b5654e04f73a87e23d23878f893 : FAILURE in 53m 41s\n\nWarning:\n  Comments left for invalid file os_client_config/tests/test_cloud_config.py","accounts_in_message":[],"_revision_number":1},{"id":"b853b8d9f006549cc1bb18b760f865a4da5e4597","author":{"_account_id":31828,"name":"yang yawei","email":"yangyawei@inspur.com","username":"yangyawei"},"date":"2021-07-16 07:20:12.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"97f1c668a5398cd434d22f1a324959bde33cdce5","tag":"autogenerated:gerrit:abandon","author":{"_account_id":31828,"name":"yang yawei","email":"yangyawei@inspur.com","username":"yangyawei"},"date":"2021-07-16 08:01:16.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"46211b308644534882e48921ed115fcbef9117c0","revisions":{"46211b308644534882e48921ed115fcbef9117c0":{"kind":"REWORK","_number":1,"created":"2021-05-14 08:09:10.000000000","uploader":{"_account_id":31828,"name":"yang yawei","email":"yangyawei@inspur.com","username":"yangyawei"},"ref":"refs/changes/07/791407/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-client-config","ref":"refs/changes/07/791407/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-client-config refs/changes/07/791407/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-client-config refs/changes/07/791407/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-client-config refs/changes/07/791407/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-client-config refs/changes/07/791407/1"}}},"commit":{"parents":[{"commit":"82a708b09f944708a770e8c5ffae5ea98848ed46","subject":"Switch to newer openstackdocstheme and reno versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-client-config/commit/82a708b09f944708a770e8c5ffae5ea98848ed46"}]}],"author":{"name":"yangyawei","email":"yangyawei@inspur.com","date":"2021-05-14 08:08:28.000000000","tz":480},"committer":{"name":"yangyawei","email":"yangyawei@inspur.com","date":"2021-05-14 08:09:05.000000000","tz":480},"subject":"setup.cfg: Replace dashes with underscores","message":"setup.cfg: Replace dashes with underscores\n\nSetuptools v54.1.0 introduces a warning that the use of dash-separated\noptions in \u0027setup.cfg\u0027 will not be supported in a future version [1].\nGet ahead of the issue by replacing the dashes with underscores. Without\nthis, we see \u0027UserWarning\u0027 messages like the following on new enough\nversions of setuptools:\n\n  UserWarning: Usage of dash-separated \u0027description-file\u0027 will not be\n  supported in future versions. Please use the underscore name\n  \u0027description_file\u0027 instead\n\n[1] https://github.com/pypa/setuptools/commit/a2e9ae4cb\n\nChange-Id: I596f8ec16ce08d6a7476a9acd9a0f5bc79d32a1e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-client-config/commit/46211b308644534882e48921ed115fcbef9117c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-client-config/commit/46211b308644534882e48921ed115fcbef9117c0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
