)]}'
{"id":"openstack%2Fos-client-config~833817","triplet_id":"openstack%2Fos-client-config~master~I90b4c8aca5f35c2e27dcaef21c6333c2bf7ae993","project":"openstack/os-client-config","branch":"master","topic":"add-python3-unversioned-jobs","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2023-11-13 17:16:38.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I90b4c8aca5f35c2e27dcaef21c6333c2bf7ae993","subject":"Update CI to use unversioned jobs template","status":"ABANDONED","created":"2022-03-15 15:07:04.000000000","updated":"2023-11-13 17:16:38.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"860ec33ee8ae61f70c88bf7362a87cb2cf3e9311","_number":833817,"virtual_id_number":833817,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-03-15 16:27:49.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Allow-Post-Review":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Abstain","+1":"Approved for Zuul Post Review Pipeline"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-15 16:27:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-17 09:28:57.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"CC"}],"messages":[{"id":"7bcdb67541a4876c8ff8142149c87b0e0d346ba5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-03-15 15:07:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"59bd03aaa3452ba03d1362435fd6c44d6fa9794f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-15 16:27:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9e84517b0f014301ba6446b7a57a025d : FAILURE in 1h 08m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1cdc525839364548be4664d19f74558b : FAILURE in 5m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/655d534ae9444c73b1d9e4fbc2ce7e94 : FAILURE in 5m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7b9fb88a5c8b4ac6ae2be4a6c9f59599 : SUCCESS in 5m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0ac2a4d7566c4d1c99c0a26956a01a3e : SUCCESS in 5m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0fb65e764698494994cb30dd8f3d6434 : SUCCESS in 8m 01s (non-voting)\n- os-client-config-tox-py37-tips https://zuul.opendev.org/t/openstack/build/16e8e50145c447359f5d712ffd345e6f : SUCCESS in 7m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9fe17bc4429b4492860d8ce303b26fe9 : SUCCESS in 6m 51s\n- shade-functional-devstack-tips-python3 https://zuul.opendev.org/t/openstack/build/2f63c5b1551048348704e20c4a282728 : FAILURE in 35m 58s\n- shade-tox-py38-tips https://zuul.opendev.org/t/openstack/build/a9bcb3dcc4924a12a487e0f4be78eb34 : FAILURE in 8m 54s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/c8fee343410b4003935fac9bf1514693 : SUCCESS in 1h 09m 57s\n\nWarning:\n  Comments left for invalid file os_client_config/tests/test_cloud_config.py","accounts_in_message":[],"_revision_number":1},{"id":"b68dcf8f2b3e1a194d9173c1161581d1e5d171ff","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-03-17 09:28:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"60040c2200e9ddc516b3ed72bfe667f13aff7939","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-17 11:05:03.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/64432ab6851041cb9f48b6f242fd4b4b : SUCCESS in 1h 33m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/efa7dd6e3b334b4dba9990f3906c709a : FAILURE in 5m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18c6270944174450a9a1ff13cce4a242 : FAILURE in 5m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ecef9893cd8d418990471ac81c122b06 : SUCCESS in 5m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eec6859cdc8941ab9210e977b198b7df : SUCCESS in 5m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e4228626b79846a1ae5ad38e9ad84958 : SUCCESS in 7m 11s (non-voting)\n- os-client-config-tox-py37-tips https://zuul.opendev.org/t/openstack/build/a8df82c9569443dd8219a3e5aa46d627 : SUCCESS in 5m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/050b3a0fc0a84a059e70b16d8c75d541 : SUCCESS in 5m 43s\n- shade-functional-devstack-tips-python3 https://zuul.opendev.org/t/openstack/build/12966ed65fcd4ea78003311845000443 : FAILURE in 42m 03s\n- shade-tox-py38-tips https://zuul.opendev.org/t/openstack/build/c9f1831e9f504f3097990d6143f0cca4 : FAILURE in 7m 11s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/0ba9c1ec95644c5bb5134e6dd3ceb3fb : SUCCESS in 1h 08m 14s\n\nWarning:\n  Comments left for invalid file os_client_config/tests/test_cloud_config.py","accounts_in_message":[],"_revision_number":1},{"id":"860ec33ee8ae61f70c88bf7362a87cb2cf3e9311","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-11-13 17:16:38.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9425194297f605ce1d23dd01ae94f6f2dd3d8ba3","revisions":{"9425194297f605ce1d23dd01ae94f6f2dd3d8ba3":{"kind":"REWORK","_number":1,"created":"2022-03-15 15:07:04.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/17/833817/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-client-config","ref":"refs/changes/17/833817/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-client-config refs/changes/17/833817/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-client-config refs/changes/17/833817/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-client-config refs/changes/17/833817/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-client-config refs/changes/17/833817/1"}}},"commit":{"parents":[{"commit":"82a708b09f944708a770e8c5ffae5ea98848ed46","subject":"Switch to newer openstackdocstheme and reno versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-client-config/commit/82a708b09f944708a770e8c5ffae5ea98848ed46"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-03-15 15:07:01.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-03-15 15:07:01.000000000","tz":0},"subject":"Update CI to use unversioned jobs template","message":"Update CI to use unversioned jobs template\n\nAs part of the migration of this project to the independent release\nmodel, we failed to notice that the job template was still tied to a\nspecific release. We\u0027ve now introduced a new unversioned job template,\n\u0027openstack-python3-jobs\u0027 [1], which was can and should use. Do this.\n\n[1] https://review.opendev.org/c/openstack/openstack-zuul-jobs/+/833286/\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-zuul-jobs/+/833286/\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: I90b4c8aca5f35c2e27dcaef21c6333c2bf7ae993\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-client-config/commit/9425194297f605ce1d23dd01ae94f6f2dd3d8ba3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-client-config/commit/9425194297f605ce1d23dd01ae94f6f2dd3d8ba3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Allow-Post-Review","description":"Allow Post Review","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
