)]}'
{"id":"openstack%2Fos-client-config~904989","triplet_id":"openstack%2Fos-client-config~master~I729a469a719b52767607938a91175f8c9285d4f6","project":"openstack/os-client-config","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-05-30 14:48:10.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I729a469a719b52767607938a91175f8c9285d4f6","subject":"Remove untested lower-constraints.txt","status":"MERGED","created":"2024-01-08 14:38:32.000000000","updated":"2025-05-30 14:48:54.000000000","submitted":"2025-05-30 14:48:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"904989","meta_rev_id":"039d5c6f9635ba985f5b869e6e410807092e6404","_number":904989,"virtual_id_number":904989,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-30 14:48:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2025-05-30 12:51:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2025-05-30 12:51:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Allow-Post-Review":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Abstain","+1":"Approved for Zuul Post Review Pipeline"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-08 16:24:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-30 12:51:15.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"01b5ebad9c3d97f41804b8e554c82851b0354c89","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-08 14:38:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4216f85f5d0a1096c2421f7beedf7329b0373c56","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-08 16:24:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/12752d4d2d80490783f1416302e80449\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/a31b8b6c3234425698ccf77f4dc2ea32 : SUCCESS in 1m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/aed8b494c97048768149d30127414368 : SUCCESS in 1h 45m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e280770aae744c7b1fc3f7ea9d8e721 : SUCCESS in 2m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/61eb6adb6e224e68b833ccc602581dba : SUCCESS in 2m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6457ee4df1ed4304ad161f85c46f9f0b : SUCCESS in 3m 37s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/28eb82ecdc5d4855b93e58e6b56ee8ce : SUCCESS in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec93b39bc22f4009bb2521cb1f444789 : SUCCESS in 2m 38s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/dce298f2fb084eeeac390f8e06c6ffad : SUCCESS in 1h 06m 29s","accounts_in_message":[],"_revision_number":1},{"id":"28fc1e7045a2b13b90e5b52946c174ecd56c7803","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-05-30 12:51:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"578572ee68c7718a9f384996427091ef1ec013fb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-30 12:51:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"de88ac107c1745eb0742f512257854e304269776","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-30 14:48:10.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/822cf19eec0343588e7f9cede9637165\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/343277c2a31f4b728c42a361773056cc : SUCCESS in 2m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8ef0d1df9a694a0d91c2bc6638c4e363 : SUCCESS in 1h 41m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69641521dc0e49c0892a19c17f28c20d : SUCCESS in 2m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f9a50e1f5dd646e9905a3ac6387a271e : SUCCESS in 2m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9489421198f44b33b0d61f025d5e4f83 : SUCCESS in 2m 51s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/f793428836b4401a857be87cedfdfa25 : SUCCESS in 5m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2ee49abebfa4d3ebd23326a45833a79 : SUCCESS in 3m 37s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/f11163590a63410ebc6ac73a6a70597b : SUCCESS in 1h 03m 01s","accounts_in_message":[],"_revision_number":1},{"id":"9068fa1b49e68e18ecf93b266d5a04cb39aee24e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-30 14:48:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"039d5c6f9635ba985f5b869e6e410807092e6404","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-30 14:48:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c5f73a9d689740b4925391122c7e0444\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8cc7b2f50e64a8db5000c6b3860e6cc : SUCCESS in 34s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3fa63cf0cceeb89863d80ef46a61f69a8957bf57","revisions":{"3fa63cf0cceeb89863d80ef46a61f69a8957bf57":{"kind":"REWORK","_number":1,"created":"2024-01-08 14:38:32.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/89/904989/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-client-config","ref":"refs/changes/89/904989/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-client-config refs/changes/89/904989/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-client-config refs/changes/89/904989/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-client-config refs/changes/89/904989/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-client-config refs/changes/89/904989/1"}}},"commit":{"parents":[{"commit":"4256fe1146bcc1b0145e2b9536c4eed10f735082","subject":"Merge \"Remove ceilometerclient\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-client-config/commit/4256fe1146bcc1b0145e2b9536c4eed10f735082"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-08 14:37:29.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-08 14:38:29.000000000","tz":540},"subject":"Remove untested lower-constraints.txt","message":"Remove untested lower-constraints.txt\n\nThe lower constraints job was already removed[1] and the file content\nis no longer tested.\n\n[1] 624cb972224ba473c98189be6f59a0ea57e7dec6\n\nChange-Id: I729a469a719b52767607938a91175f8c9285d4f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-client-config/commit/3fa63cf0cceeb89863d80ef46a61f69a8957bf57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-client-config/commit/3fa63cf0cceeb89863d80ef46a61f69a8957bf57"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Allow-Post-Review","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Allow-Post-Review","description":"Allow Post Review","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
