)]}'
{"id":"openstack%2Fos-client-config~945087","triplet_id":"openstack%2Fos-client-config~master~I2e9da526c271716a02dda47cd74ac998d5133af8","project":"openstack/os-client-config","branch":"master","topic":"tests-no-private-methods","attention_set":{},"removed_from_attention_set":{"9542":{"account":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"last_update":"2025-03-20 19:28:45.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2e9da526c271716a02dda47cd74ac998d5133af8","subject":"Stop using private SDK methods in config unit tests","status":"MERGED","created":"2025-03-20 12:01:28.000000000","updated":"2025-03-20 19:29:39.000000000","submitted":"2025-03-20 19:28:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"945087-tests-no-private-methods","meta_rev_id":"59671f4e6e9ab119b438345e96d8d4272e868eeb","_number":945087,"virtual_id_number":945087,"owner":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-20 19:28:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2025-03-20 17:47:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2025-03-20 17:47:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Allow-Post-Review":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Abstain","+1":"Approved for Zuul Post Review Pipeline"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-20 13:39:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-20 17:47:35.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"bbcb72dda45e43c49239ffc311b8c326665b643b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2025-03-20 12:01:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3a6547835b80341c45fc095e16d6f85651fc5970","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-20 13:39:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0de3219f6aa949018fff3cb36ccf2598\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/63bfa9735bd64f218b079ea55488638a : SUCCESS in 1h 35m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14b943607756446cb04f7d75ec29281a : SUCCESS in 1m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c6d056a3309a4dff84925f335bf88c36 : SUCCESS in 3m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/99900e121f314c858da49188e00dd6ad : SUCCESS in 1m 56s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/3ad0cc6f2ac843ea8146c10a4f09f384 : SUCCESS in 3m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eff6b340a2634508b4313570d0b94258 : SUCCESS in 1m 39s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/362512075a5a4a60b6693f15e8559ea0 : SUCCESS in 59m 48s","accounts_in_message":[],"_revision_number":1},{"id":"c90d9a6107fa0d437ec575257ef403da4e5963dc","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-03-20 17:47:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b561b1dc4886091a73fc81ce4be1dcb2e8172aee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-20 17:48:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"905f14c2dd89e2946993ac80658de865dce6c69b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-20 19:28:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2cbcda4180a1467fb9d5e1dc40dbb2a7\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5d275624bc9a417fb8f04abb3462b0c3 : SUCCESS in 1h 39m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1043f9a3e164ab4ab0713bb0dbdf815 : SUCCESS in 1m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/279c1f056e304cc4955a90d4adfb62bc : SUCCESS in 2m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cc38a4a03fb046efb38cb37e0b5edc89 : SUCCESS in 2m 48s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/9d400a8e47d14fb3815a52272231392a : SUCCESS in 4m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2773ee5d7d74989a934c8fa2f519748 : SUCCESS in 2m 47s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/44c886074f224ea8ba73f5e59b278329 : SUCCESS in 59m 18s","accounts_in_message":[],"_revision_number":1},{"id":"e6f1c242d3d60af39a4b08da8f88478d548a4fa6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-20 19:28:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"59671f4e6e9ab119b438345e96d8d4272e868eeb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-20 19:29:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7cb59ae61c7f40428fc165d910f9178d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01ab112fcdb54f24b9719dc88878ecbf : SUCCESS in 33s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"46bc2deb4c6762dc1dd674686283eb3fa4c1d5e6","revisions":{"46bc2deb4c6762dc1dd674686283eb3fa4c1d5e6":{"kind":"REWORK","_number":1,"created":"2025-03-20 12:01:28.000000000","uploader":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"ref":"refs/changes/87/945087/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-client-config","ref":"refs/changes/87/945087/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-client-config refs/changes/87/945087/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-client-config refs/changes/87/945087/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-client-config refs/changes/87/945087/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-client-config refs/changes/87/945087/1"}}},"commit":{"parents":[{"commit":"1f53c701c6fdf90dbeb5ac3d2a395ea884a04672","subject":"Fix unit tests to pass locally","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-client-config/commit/1f53c701c6fdf90dbeb5ac3d2a395ea884a04672"}]}],"author":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2025-03-20 10:25:04.000000000","tz":0},"committer":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2025-03-20 12:01:18.000000000","tz":0},"subject":"Stop using private SDK methods in config unit tests","message":"Stop using private SDK methods in config unit tests\n\nChange-Id: I2e9da526c271716a02dda47cd74ac998d5133af8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-client-config/commit/46bc2deb4c6762dc1dd674686283eb3fa4c1d5e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-client-config/commit/46bc2deb4c6762dc1dd674686283eb3fa4c1d5e6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Allow-Post-Review","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Allow-Post-Review","description":"Allow Post Review","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
