)]}'
{"id":"openstack%2Fos-client-config~951401","triplet_id":"openstack%2Fos-client-config~master~Idba69929e020c487f98d16ec7adaf69bf98d2f30","project":"openstack/os-client-config","branch":"master","topic":"pre-commit","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-05-30 17:15:56.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}},"hashtags":[],"change_id":"Idba69929e020c487f98d16ec7adaf69bf98d2f30","subject":"tests: Remove duplicated tests","status":"MERGED","created":"2025-05-30 12:45:13.000000000","updated":"2025-05-30 21:24:55.000000000","submitted":"2025-05-30 21:23:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"951401-pre-commit","meta_rev_id":"813f44d9340402ae52a8303b4a992604b229b0ef","_number":951401,"virtual_id_number":951401,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-30 21:23:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2025-05-30 17:15:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2025-05-30 17:15:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Allow-Post-Review":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Abstain","+1":"Approved for Zuul Post Review Pipeline"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-30 14:59:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d79cae516dab541643fc14a87fba2fa54bccc897","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-05-30 12:45:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f85f6c4cebdec92187889a27040f1dd12790ff9b","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-05-30 12:49:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6cfb77254ba56031deabdc7aad34d4fd1d74f928","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-30 14:59:17.000000000","message":"Patch Set 1: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3c8bbeded8ed45abad22b6d5ae45c0aa\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4731d2cf57ca483586b31302d0c59f66 : SUCCESS in 1h 32m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5196e050751400886ec887a978edea3 : FAILURE in 2m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9d3945abebfe4352b5b7fd9d7b9e500f : SUCCESS in 3m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8086196755e047efae4dbe864ebbb923 : SUCCESS in 2m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ae221a88c65f4e7b81db3ba168fc7f29 : SUCCESS in 5m 10s (non-voting)\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/cb305ec469684a4aa3111b9ed46a63b2 : SUCCESS in 3m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7ff5e6b867e4d37b93ba4e84fe62050 : SUCCESS in 3m 37s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/3339c8f2385845c78a2bbee5e2cac73b : SUCCESS in 58m 45s","accounts_in_message":[],"_revision_number":1},{"id":"46e35b56ad9bdd8584bb908aed462c35b61d886c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-05-30 16:59:15.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":2},{"id":"577c7aa70e940fb5a2517f4f95c84a47e072d3e0","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-05-30 17:15:56.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d2e5677fb8d3febce948c3d9d8c61f348f8670cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-30 19:34:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7116f3eec9664d52b2a1f5ff4e43a1e0\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0a6e221605784eea814e3713e1deecc1 : SUCCESS in 1h 59m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23e40682707845d1b974d180b1891a80 : SUCCESS in 3m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/55d5f0658c6e42c080e7b7ba9a9f6dc6 : SUCCESS in 3m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/857f187d6252421f89338ec8a8276be6 : SUCCESS in 2m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f70b10522a8446d2be378800058e3987 : SUCCESS in 5m 26s (non-voting)\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/a97d1e69f2fd4695a07a6bc1fd4bdcda : SUCCESS in 3m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59bfbfee7c6545ceb0a89af940c76a33 : SUCCESS in 3m 33s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/f0bd13f464ee4ba29547b332f4c122ff : SUCCESS in 1h 12m 29s","accounts_in_message":[],"_revision_number":2},{"id":"74ff7db6159c4111e3d64bf44d086a182a042e87","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-30 19:35:10.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f99ace7d528194c9a283c67cfb7a79bd1152bdd0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-30 21:23:59.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c9d4e72bf1d4b829e08c576e85f1a91\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b6424f2635184fcba930fe150bdbbccf : SUCCESS in 1h 17m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e363152d2a114462978cd068b46cdc3b : SUCCESS in 2m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f179862527394dc8a621fa0b266de7d8 : SUCCESS in 4m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/26ef50a928a046338d607544e370edc1 : SUCCESS in 3m 55s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/ee5dc4976a58471e9c50d3916189a5d8 : SUCCESS in 3m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0223d6d1e4ba4622a81def3276610275 : SUCCESS in 2m 38s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/e6875fc9e2df40f398deef41bcc8b01d : SUCCESS in 1h 10m 06s","accounts_in_message":[],"_revision_number":2},{"id":"ff18c5b9ddba23c0501385e9aac4469eedf450d1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-30 21:23:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"813f44d9340402ae52a8303b4a992604b229b0ef","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-30 21:24:55.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/76f296c5e7c24323bb4e313e3d52fd41\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da5e005c26154f828d05e79e11ad0446 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6d5283eb9af5b43307d91f1bc3a9d72ecce7df4b","revisions":{"c5f3e91f4e74a5cff638cee8f7c43f91a34560a7":{"kind":"REWORK","_number":1,"created":"2025-05-30 12:45:13.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/01/951401/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-client-config","ref":"refs/changes/01/951401/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-client-config refs/changes/01/951401/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-client-config refs/changes/01/951401/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-client-config refs/changes/01/951401/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-client-config refs/changes/01/951401/1"}}},"commit":{"parents":[{"commit":"46bc2deb4c6762dc1dd674686283eb3fa4c1d5e6","subject":"Stop using private SDK methods in config unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-client-config/commit/46bc2deb4c6762dc1dd674686283eb3fa4c1d5e6"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-05-30 12:39:06.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-05-30 12:39:09.000000000","tz":60},"subject":"tests: Remove duplicated tests","message":"tests: Remove duplicated tests\n\nWe should not be testing functionality from the parent class,\nparticularly when they\u0027re duplicating tests already found in the parent\nlibrary (openstack/tests/unit/config/test_cloud_config.py).\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: Idba69929e020c487f98d16ec7adaf69bf98d2f30\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-client-config/commit/c5f3e91f4e74a5cff638cee8f7c43f91a34560a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-client-config/commit/c5f3e91f4e74a5cff638cee8f7c43f91a34560a7"}]},"branch":"refs/heads/master"},"6d5283eb9af5b43307d91f1bc3a9d72ecce7df4b":{"kind":"REWORK","_number":2,"created":"2025-05-30 16:59:15.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/01/951401/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-client-config","ref":"refs/changes/01/951401/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-client-config refs/changes/01/951401/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-client-config refs/changes/01/951401/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-client-config refs/changes/01/951401/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-client-config refs/changes/01/951401/2"}}},"commit":{"parents":[{"commit":"46bc2deb4c6762dc1dd674686283eb3fa4c1d5e6","subject":"Stop using private SDK methods in config unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-client-config/commit/46bc2deb4c6762dc1dd674686283eb3fa4c1d5e6"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-05-30 12:39:06.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-05-30 16:57:28.000000000","tz":60},"subject":"tests: Remove duplicated tests","message":"tests: Remove duplicated tests\n\nWe should not be testing functionality from the parent class,\nparticularly when they\u0027re duplicating tests already found in the parent\nlibrary (openstack/tests/unit/config/test_cloud_config.py).\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: Idba69929e020c487f98d16ec7adaf69bf98d2f30\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-client-config/commit/6d5283eb9af5b43307d91f1bc3a9d72ecce7df4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-client-config/commit/6d5283eb9af5b43307d91f1bc3a9d72ecce7df4b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Allow-Post-Review","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Allow-Post-Review","description":"Allow Post Review","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
