)]}'
{"id":"openstack%2Fos-client-config~951405","triplet_id":"openstack%2Fos-client-config~master~Iedc7653f27dbe9bb77b6d9bb03117a457e195e58","project":"openstack/os-client-config","branch":"master","topic":"pre-commit","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-05-30 17:16:04.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}},"hashtags":[],"change_id":"Iedc7653f27dbe9bb77b6d9bb03117a457e195e58","subject":"Add pre-commit, apply ruff","status":"MERGED","created":"2025-05-30 12:50:33.000000000","updated":"2025-05-30 21:25:41.000000000","submitted":"2025-05-30 21:24:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"951405-pre-commit","meta_rev_id":"e16fcd573ce43a4a0755baf4036da766b48ec9b9","_number":951405,"virtual_id_number":951405,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-30 21:24:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2025-05-30 17:16:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2025-05-30 17:16:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Allow-Post-Review":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Abstain","+1":"Approved for Zuul Post Review Pipeline"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-30 14:33:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"070383b5f2817a5daa83a199be0fb6a48c822d2a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-05-30 12:50:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0bd0788103e37132101c1b12096201171ac315cb","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-05-30 12:50:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a025e4bfe77868be84118182edaf89bfe93f623d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-30 14:33:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/133424f99c6f494bac353cc3bb49a922\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/f04df5d500bc4eadbdac411a4008cc2f : SUCCESS in 3m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fae68ffe9a414c5e896fb4e6b027d1a9 : SUCCESS in 1h 41m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9238105bfb9440929e5a5da34952fcd7 : SUCCESS in 4m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a70c04e3069c4a21a3140344d6c687a5 : SUCCESS in 4m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/452bf019b7ad4e59965304b385b2fbfe : SUCCESS in 4m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3a746be19f5c470883ac7c725853ea50 : SUCCESS in 6m 55s (non-voting)\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/ecd2ce46a33e4ca38daac1c5874b45cc : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23a86cd96ed6443a8bb91843a55ad3d6 : SUCCESS in 4m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bee2c8d84977401d85d02b21b2234f00 : SUCCESS in 3m 14s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/47349098d0db4ef9b27fb2c29a81bf2a : SUCCESS in 1h 08m 57s","accounts_in_message":[],"_revision_number":1},{"id":"c304595fb9cb7094426c309c14f063e5b83c5218","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-05-30 16:59:15.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":2},{"id":"ac984fae51acba6ddbf06b793860951319e15fcc","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-05-30 17:16:04.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"cbed062c005cf5e4181af07a88548cfa6d545cb1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-30 19:32:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/89ae49d7474c49d8b5010e8873584c67\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/917b53f0a6ee472caab067a8a75e91bf : SUCCESS in 3m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b9443b615112424e8d6378715b77fbf6 : SUCCESS in 1h 57m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87fbf718641f4df69352a2ea15b58a30 : SUCCESS in 2m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f4aa7390bebb4954ba586baa4deed1b2 : SUCCESS in 2m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/30c1ecaa29f14a80abeb06656992f785 : SUCCESS in 4m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6c44aafb6629415499ffb0b6a0aa4413 : SUCCESS in 6m 14s (non-voting)\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/88e887dbbf7649848fb5428a0623588f : SUCCESS in 3m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6381a9a02e0341cdbe71ddced3240cf4 : SUCCESS in 3m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9b3c341440cd4921ab66eaacf52519f8 : SUCCESS in 2m 40s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/8a1678ef8f0646d3b13e9974777bbe6b : SUCCESS in 1h 08m 34s","accounts_in_message":[],"_revision_number":2},{"id":"b0577364588d5c3609239e23deb0bdbf87116597","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-30 19:35:14.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"63561b58cd6e8a49fbad2c52a809aa24da08df33","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-30 21:24:02.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9fefdaf266ff4c19bf129153c8e1f540\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/87c31a6917c548be9465fb285674501c : SUCCESS in 2m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/85f7ea0641274c84ac8bfbe5416ee0cf : SUCCESS in 1h 42m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2730f1db18884a97bf2d385aba7d9cb2 : SUCCESS in 2m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6f27dc9a8d794eb184a885d698eaf6ee : SUCCESS in 2m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/869c0ed1f3b14a8cb0733f203454ecc9 : SUCCESS in 2m 57s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/896f307442314630bf7a3cde69a62a92 : SUCCESS in 3m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2dc94769e7a84b33a2078781a24a1cd4 : SUCCESS in 2m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/113badfea1ba44ad995d8b882b66d32a : SUCCESS in 2m 38s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/85684547f6234893b6c5af51f2d7b1cb : SUCCESS in 1h 00m 32s","accounts_in_message":[],"_revision_number":2},{"id":"7d5fff0553ea797c83e2b5b5e5ec1fdef50500e0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-30 21:24:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"e16fcd573ce43a4a0755baf4036da766b48ec9b9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-30 21:25:41.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2490b536f4724139807ec6925e8938d8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a641685a8664a87997a5ad75108bfa2 : SUCCESS in 39s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b71555ea66374d6fb72aae3ee5e09032 : SUCCESS in 34s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"15b72f2214babb1f36e3e8fe69b940463c6602f6","revisions":{"54b41e6b6ca7c654bedfde4997e66d643250ea7d":{"kind":"REWORK","_number":1,"created":"2025-05-30 12:50:33.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/05/951405/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-client-config","ref":"refs/changes/05/951405/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-client-config refs/changes/05/951405/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-client-config refs/changes/05/951405/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-client-config refs/changes/05/951405/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-client-config refs/changes/05/951405/1"}}},"commit":{"parents":[{"commit":"790763b5ce568fab5cbef2c0db630c406750b402","subject":"Drop support for Python \u003c 3.10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-client-config/commit/790763b5ce568fab5cbef2c0db630c406750b402"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-05-30 12:41:26.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-05-30 12:49:01.000000000","tz":60},"subject":"Add pre-commit, apply ruff","message":"Add pre-commit, apply ruff\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: Iedc7653f27dbe9bb77b6d9bb03117a457e195e58\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-client-config/commit/54b41e6b6ca7c654bedfde4997e66d643250ea7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-client-config/commit/54b41e6b6ca7c654bedfde4997e66d643250ea7d"}]},"branch":"refs/heads/master"},"15b72f2214babb1f36e3e8fe69b940463c6602f6":{"kind":"REWORK","_number":2,"created":"2025-05-30 16:59:15.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/05/951405/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-client-config","ref":"refs/changes/05/951405/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-client-config refs/changes/05/951405/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-client-config refs/changes/05/951405/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-client-config refs/changes/05/951405/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-client-config refs/changes/05/951405/2"}}},"commit":{"parents":[{"commit":"964b7237e5629f701343bb1d4a3f96ca27f754e8","subject":"Drop support for Python \u003c 3.10","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-client-config/commit/964b7237e5629f701343bb1d4a3f96ca27f754e8"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-05-30 12:41:26.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-05-30 16:57:53.000000000","tz":60},"subject":"Add pre-commit, apply ruff","message":"Add pre-commit, apply ruff\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: Iedc7653f27dbe9bb77b6d9bb03117a457e195e58\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-client-config/commit/15b72f2214babb1f36e3e8fe69b940463c6602f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-client-config/commit/15b72f2214babb1f36e3e8fe69b940463c6602f6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Allow-Post-Review","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Allow-Post-Review","description":"Allow Post Review","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
