)]}'
{"id":"openstack%2Fos-client-config~953790","triplet_id":"openstack%2Fos-client-config~master~Ia9ed2ca179d95071e54b8611993f42c8417d0ff2","project":"openstack/os-client-config","branch":"master","topic":"retirement","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-06-30 19:23:34.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}},"hashtags":[],"change_id":"Ia9ed2ca179d95071e54b8611993f42c8417d0ff2","subject":"Log unmaintained warning on import","status":"MERGED","created":"2025-06-30 15:37:50.000000000","updated":"2025-06-30 22:37:58.000000000","submitted":"2025-06-30 22:37:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"953790-retirement","meta_rev_id":"b4f1e79cdfc4e0906a22ae045fd2b8c8afe899b6","_number":953790,"virtual_id_number":953790,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-30 22:37:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2025-06-30 19:23:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2025-06-30 19:23:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Allow-Post-Review":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Abstain","+1":"Approved for Zuul Post Review Pipeline"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-30 17:28:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5fdb8ac70e5e1dfada3fb59f630fcbaa4f4b64ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-06-30 15:37:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f3f3c39b4d58560513a1d26b29a36ad4452e9e40","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-06-30 15:53:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5331741aa6bca67bdf07a74951e5ff57648e0e00","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-30 17:28:32.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ca370df9c25b4cd28adc2f233da6e980\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1896dbdb10e84293afd1386478ae8842 : SUCCESS in 1h 47m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f1ffddf82ef44e5b3155a920358ba5b : FAILURE in 2m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d7786d276a3942f1b1525c91d22a7e95 : SUCCESS in 4m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b64408b864814f129fa4ddf8c2572657 : SUCCESS in 3m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c207bef7131b4bffb6274cbce544a594 : SUCCESS in 3m 03s (non-voting)\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/231ef233a924452198280a4b021a6b6c : SUCCESS in 3m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8c847e29477467595ae9ee685ce354f : SUCCESS in 2m 51s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/5555f69d7b0342998e0efc3121330f90 : SUCCESS in 1h 00m 21s","accounts_in_message":[],"_revision_number":1},{"id":"ddcb187de5c1a4c98e21d4ac7247e27a03ef90a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-06-30 19:23:31.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":2},{"id":"58523096651dd292bf00d1476609fe3aabfda45a","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-06-30 19:23:34.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a4ec72c839e13146389918fa3b37ad8afe174171","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-30 21:06:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/34cf365b1a754fedb1aff1b6a0963c85\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e231f4441c1e4b79a5bd74127e213239 : SUCCESS in 1h 31m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/663c824ca7ae4b87a372f9dc591491a0 : SUCCESS in 2m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a0faefb369ec468b8d1614e83bea7d9b : SUCCESS in 2m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6fd9745d8fd9453e953e890b5e9d427b : SUCCESS in 4m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d1ce888fd57f4b1fb759f7c29b3d6152 : SUCCESS in 4m 12s (non-voting)\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/c165f930c3334710b03388f28b82ff09 : SUCCESS in 3m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6623419a5d2e46e8a3e53c77da14339b : SUCCESS in 3m 39s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/9dce3cdcb68c443a9f23f825621fd1f9 : SUCCESS in 1h 01m 00s","accounts_in_message":[],"_revision_number":2},{"id":"33af1eb9f0dab71ed94805cd41d284d7703778d5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-30 21:07:11.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1d087b7d11b0fbc5975faec523500099368ee6bc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-30 22:37:07.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/21d91d7435c14f56b83a0b477f9ed33f\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/76d1f61f4e2d4c25a487ffd823e40edb : SUCCESS in 1h 28m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3e9665569e247658a7fb0ec12f935c0 : SUCCESS in 4m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1df3b77853a9419ea97d8f36b6c7758b : SUCCESS in 4m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cee43570d1a3404ca64161453dad3521 : SUCCESS in 2m 49s\n- os-client-config-tox-py311-tips https://zuul.opendev.org/t/openstack/build/15a060f7af354c2aa633c35cdae47a7b : SUCCESS in 3m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51ec4d60e1d143bdbb75a7e575518219 : SUCCESS in 1m 36s\n- osc-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/2d4ab7b9c1254e9797f93484f863a4e8 : SUCCESS in 58m 43s","accounts_in_message":[],"_revision_number":2},{"id":"67c5f726f0d83c9805e5f026bcff028ca971d3e8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-30 22:37:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"b4f1e79cdfc4e0906a22ae045fd2b8c8afe899b6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-30 22:37:58.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6acf2c9e2ed64840a0bc2d61cf192306\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f19e9b2f793743ccbdcd772ca9a26804 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d1e2fe7d2e07767f1a66ed0c057c20d7a95271f4","revisions":{"195a3aafc76c5b407424bcd76571c721d7322007":{"kind":"REWORK","_number":1,"created":"2025-06-30 15:37:50.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/90/953790/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-client-config","ref":"refs/changes/90/953790/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-client-config refs/changes/90/953790/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-client-config refs/changes/90/953790/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-client-config refs/changes/90/953790/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-client-config refs/changes/90/953790/1"}}},"commit":{"parents":[{"commit":"1c5a1f7dee3c7bf629b01353cb90e40329ec1052","subject":"Add pbr as dep, remove hacking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-client-config/commit/1c5a1f7dee3c7bf629b01353cb90e40329ec1052"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-06-30 15:35:53.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-06-30 15:35:53.000000000","tz":60},"subject":"Log unmaintained warning on import","message":"Log unmaintained warning on import\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: Ia9ed2ca179d95071e54b8611993f42c8417d0ff2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-client-config/commit/195a3aafc76c5b407424bcd76571c721d7322007"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-client-config/commit/195a3aafc76c5b407424bcd76571c721d7322007"}]},"branch":"refs/heads/master"},"d1e2fe7d2e07767f1a66ed0c057c20d7a95271f4":{"kind":"REWORK","_number":2,"created":"2025-06-30 19:23:31.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/90/953790/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-client-config","ref":"refs/changes/90/953790/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-client-config refs/changes/90/953790/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-client-config refs/changes/90/953790/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-client-config refs/changes/90/953790/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-client-config refs/changes/90/953790/2"}}},"commit":{"parents":[{"commit":"1c5a1f7dee3c7bf629b01353cb90e40329ec1052","subject":"Add pbr as dep, remove hacking","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-client-config/commit/1c5a1f7dee3c7bf629b01353cb90e40329ec1052"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-06-30 15:35:53.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-06-30 19:23:31.000000000","tz":0},"subject":"Log unmaintained warning on import","message":"Log unmaintained warning on import\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: Ia9ed2ca179d95071e54b8611993f42c8417d0ff2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-client-config/commit/d1e2fe7d2e07767f1a66ed0c057c20d7a95271f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-client-config/commit/d1e2fe7d2e07767f1a66ed0c057c20d7a95271f4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Allow-Post-Review","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Allow-Post-Review","description":"Allow Post Review","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
