)]}'
{"id":"openstack%2Fos-cloud-config~129656","triplet_id":"openstack%2Fos-cloud-config~master~Ic8f356d2d67324c0b2ccaa8a7b2cf7752635f68b","project":"openstack/os-cloud-config","branch":"master","topic":"feature/cyclomatic-complexity-cap","hashtags":[],"change_id":"Ic8f356d2d67324c0b2ccaa8a7b2cf7752635f68b","subject":"Put a cap on our cyclomatic complexity","status":"MERGED","created":"2014-10-20 16:10:50.000000000","updated":"2014-10-20 19:43:45.000000000","submitted":"2014-10-20 19:43:45.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ddc2d671ccdb62abf1140a4f3f8471972e903373","_number":129656,"virtual_id_number":129656,"owner":{"_account_id":10035,"name":"greghaynes","email":"greg@greghaynes.net","username":"greghaynes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":8688,"name":"Alexis Lee","email":"openstack@lxsli.co.uk","username":"lxsli"},{"value":2,"date":"2014-10-20 19:43:45.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":8399,"name":"Jay Dobies","email":"jason.dobies@redhat.com","username":"jdob"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8688,"name":"Alexis Lee","email":"openstack@lxsli.co.uk","username":"lxsli"},"all":[{"value":2,"date":"2014-10-20 16:17:01.000000000","_account_id":8688,"name":"Alexis Lee","email":"openstack@lxsli.co.uk","username":"lxsli"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":2,"date":"2014-10-20 19:37:45.000000000","_account_id":8399,"name":"Jay Dobies","email":"jason.dobies@redhat.com","username":"jdob"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8399,"name":"Jay Dobies","email":"jason.dobies@redhat.com","username":"jdob"},"all":[{"value":0,"_account_id":8688,"name":"Alexis Lee","email":"openstack@lxsli.co.uk","username":"lxsli"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2014-10-20 19:37:45.000000000","_account_id":8399,"name":"Jay Dobies","email":"jason.dobies@redhat.com","username":"jdob"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":8399,"name":"Jay Dobies","email":"jason.dobies@redhat.com","username":"jdob"},{"_account_id":8688,"name":"Alexis Lee","email":"openstack@lxsli.co.uk","username":"lxsli"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2014-10-20 16:17:01.000000000","updated_by":{"_account_id":8688,"name":"Alexis Lee","email":"openstack@lxsli.co.uk","username":"lxsli"},"reviewer":{"_account_id":8688,"name":"Alexis Lee","email":"openstack@lxsli.co.uk","username":"lxsli"},"state":"REVIEWER"},{"updated":"2014-10-20 19:37:45.000000000","updated_by":{"_account_id":8399,"name":"Jay Dobies","email":"jason.dobies@redhat.com","username":"jdob"},"reviewer":{"_account_id":8399,"name":"Jay Dobies","email":"jason.dobies@redhat.com","username":"jdob"},"state":"REVIEWER"},{"updated":"2014-10-20 19:43:45.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"f36829fc0c713558e065f1eabd49fadc5d5ae112","author":{"_account_id":10035,"name":"greghaynes","email":"greg@greghaynes.net","username":"greghaynes"},"date":"2014-10-20 16:10:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d3d39e693e5b11305287649ddd46ea874940974a","author":{"_account_id":8688,"name":"Alexis Lee","email":"openstack@lxsli.co.uk","username":"lxsli"},"date":"2014-10-20 16:17:01.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0fc725f13f8d1724325d47757206714c91806026","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-10-20 16:50:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-os-cloud-config-pep8 http://logs.openstack.org/56/129656/1/check/gate-os-cloud-config-pep8/8ad03f1 : SUCCESS in 2m 31s\n- gate-os-cloud-config-docs http://docs-draft.openstack.org/56/129656/1/check/gate-os-cloud-config-docs/cbff12d/doc/build/html/ : SUCCESS in 2m 35s\n- gate-os-cloud-config-python26 http://logs.openstack.org/56/129656/1/check/gate-os-cloud-config-python26/f902e52 : SUCCESS in 1m 57s\n- gate-os-cloud-config-python27 http://logs.openstack.org/56/129656/1/check/gate-os-cloud-config-python27/b26384b : SUCCESS in 2m 35s\n- gate-os-cloud-config-python33 http://logs.openstack.org/56/129656/1/check/gate-os-cloud-config-python33/3bb3693 : SUCCESS in 4m 32s","accounts_in_message":[],"_revision_number":1},{"id":"b6845c9895f223eecb39384de4a2f801b0ba1b13","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-10-20 18:23:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (check-tripleo pipeline).\n\n- check-tripleo-ironic-undercloud-precise-nonha http://logs.openstack.org/56/129656/1/check-tripleo/check-tripleo-ironic-undercloud-precise-nonha/b0312ab : SUCCESS in 50m 00s\n- check-tripleo-ironic-overcloud-f20-nonha http://logs.openstack.org/56/129656/1/check-tripleo/check-tripleo-ironic-overcloud-f20-nonha/ce070e5 : SUCCESS in 2h 12m 53s\n- check-tripleo-ironic-overcloud-precise-nonha http://logs.openstack.org/56/129656/1/check-tripleo/check-tripleo-ironic-overcloud-precise-nonha/587ad48 : SUCCESS in 1h 23m 32s","accounts_in_message":[],"_revision_number":1},{"id":"f167149cb018a94e1717cefb3d8fcb233ddb3b78","author":{"_account_id":8399,"name":"Jay Dobies","email":"jason.dobies@redhat.com","username":"jdob"},"date":"2014-10-20 19:37:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nI\u0027d like to see it get lower than this, but it\u0027s smart to put the cap in now and address it later.","accounts_in_message":[],"_revision_number":1},{"id":"dc8ec759f1a517a7a9813902755babb196a5732d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-10-20 19:37:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"396f1cd9125ed05f55e8627c41eeabe4b5224513","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-10-20 19:43:45.000000000","message":"Change has been successfully merged into the git repository.","accounts_in_message":[],"_revision_number":1},{"id":"769807f252180436556bd06bd60c212a8e29509f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-10-20 19:43:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-os-cloud-config-docs http://docs-draft.openstack.org/56/129656/1/gate/gate-os-cloud-config-docs/5716e0c/doc/build/html/ : SUCCESS in 2m 41s\n- gate-os-cloud-config-pep8 http://logs.openstack.org/56/129656/1/gate/gate-os-cloud-config-pep8/c0f992c : SUCCESS in 2m 34s\n- gate-os-cloud-config-python26 http://logs.openstack.org/56/129656/1/gate/gate-os-cloud-config-python26/72902cc : SUCCESS in 2m 36s\n- gate-os-cloud-config-python27 http://logs.openstack.org/56/129656/1/gate/gate-os-cloud-config-python27/325d0fa : SUCCESS in 2m 38s\n- gate-os-cloud-config-python33 http://logs.openstack.org/56/129656/1/gate/gate-os-cloud-config-python33/1889b73 : SUCCESS in 3m 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"97d6b61c4573141c1064132ecac733db10808e64","revisions":{"97d6b61c4573141c1064132ecac733db10808e64":{"kind":"REWORK","_number":1,"created":"2014-10-20 16:10:50.000000000","uploader":{"_account_id":10035,"name":"greghaynes","email":"greg@greghaynes.net","username":"greghaynes"},"ref":"refs/changes/56/129656/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-cloud-config","ref":"refs/changes/56/129656/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-cloud-config refs/changes/56/129656/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-cloud-config refs/changes/56/129656/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-cloud-config refs/changes/56/129656/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-cloud-config refs/changes/56/129656/1"}}},"commit":{"parents":[{"commit":"3ddf7862eebd8e4663dba79a4a4fca6af2d96f45","subject":"Merge \"Add support for injecting extra-specs into flavors\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-cloud-config/commit/3ddf7862eebd8e4663dba79a4a4fca6af2d96f45"}]}],"author":{"name":"Gregory Haynes","email":"greg@greghaynes.net","date":"2014-10-20 16:06:22.000000000","tz":-420},"committer":{"name":"Gregory Haynes","email":"greg@greghaynes.net","date":"2014-10-20 16:10:56.000000000","tz":-420},"subject":"Put a cap on our cyclomatic complexity","message":"Put a cap on our cyclomatic complexity\n\nflake8 supports cyclomatic complexity checking, so lets use it. Our\ncurrent worst offender is register_ironic_node (which I kind of agree\nwith) so setting our cap to 14 which is one higher than this.\n\nChange-Id: Ic8f356d2d67324c0b2ccaa8a7b2cf7752635f68b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-cloud-config/commit/97d6b61c4573141c1064132ecac733db10808e64"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-cloud-config/commit/97d6b61c4573141c1064132ecac733db10808e64"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"3ddf7862eebd8e4663dba79a4a4fca6af2d96f45","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
