)]}'
{"id":"openstack%2Fos-cloud-config~99849","triplet_id":"openstack%2Fos-cloud-config~master~I77eef1654b025e33b1b54d35f020d0b61d739e5a","project":"openstack/os-cloud-config","branch":"master","topic":"no-conflict-for-power-state","hashtags":[],"change_id":"I77eef1654b025e33b1b54d35f020d0b61d739e5a","subject":"Ignore Conflict exception when setting power state","status":"MERGED","created":"2014-06-13 05:55:52.000000000","updated":"2014-06-16 08:22:22.000000000","submitted":"2014-06-16 08:22:22.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"42902209f33c008bfe4ae0dc6c9e4a8328e0c119","_number":99849,"virtual_id_number":99849,"owner":{"_account_id":9369,"name":"Steve Kowalik","email":"steven@wedontsleep.org","username":"stevenk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":2,"date":"2014-06-16 08:22:22.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":4330,"name":"Tomas Sedovic","email":"tsedovic@redhat.com","username":"tomas-8c8"},{"value":0,"_account_id":4190,"name":"lifeless","email":"robertc@robertcollins.net","username":"lifeless"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4330,"name":"Tomas Sedovic","email":"tsedovic@redhat.com","username":"tomas-8c8"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":2,"date":"2014-06-13 09:41:00.000000000","_account_id":4330,"name":"Tomas Sedovic","email":"tsedovic@redhat.com","username":"tomas-8c8"},{"value":2,"date":"2014-06-13 09:31:26.000000000","_account_id":4190,"name":"lifeless","email":"robertc@robertcollins.net","username":"lifeless"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4330,"name":"Tomas Sedovic","email":"tsedovic@redhat.com","username":"tomas-8c8"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2014-06-16 08:15:29.000000000","_account_id":4330,"name":"Tomas Sedovic","email":"tsedovic@redhat.com","username":"tomas-8c8"},{"value":0,"_account_id":4190,"name":"lifeless","email":"robertc@robertcollins.net","username":"lifeless"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":4190,"name":"lifeless","email":"robertc@robertcollins.net","username":"lifeless"},{"_account_id":4330,"name":"Tomas Sedovic","email":"tsedovic@redhat.com","username":"tomas-8c8"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2014-06-13 09:31:26.000000000","updated_by":{"_account_id":4190,"name":"lifeless","email":"robertc@robertcollins.net","username":"lifeless"},"reviewer":{"_account_id":4190,"name":"lifeless","email":"robertc@robertcollins.net","username":"lifeless"},"state":"REVIEWER"},{"updated":"2014-06-16 08:15:29.000000000","updated_by":{"_account_id":4330,"name":"Tomas Sedovic","email":"tsedovic@redhat.com","username":"tomas-8c8"},"reviewer":{"_account_id":4330,"name":"Tomas Sedovic","email":"tsedovic@redhat.com","username":"tomas-8c8"},"state":"REVIEWER"},{"updated":"2014-06-16 08:22:22.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"13c15126fcda8df0b8cabbbb76d5e75f195935c5","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-06-13 05:55:52.000000000","message":"Patch Set 1:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"14fa07f74ab11f772f52ddb082a23d054e3fa658","author":{"_account_id":9369,"name":"Steve Kowalik","email":"steven@wedontsleep.org","username":"stevenk"},"date":"2014-06-13 05:55:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8781ee6d39035e4b44e712d024aa9e08e1c03828","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-06-13 06:02:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded.\n\n- gate-os-cloud-config-pep8 http://logs.openstack.org/49/99849/1/check/gate-os-cloud-config-pep8/ed8b28a : SUCCESS in 3m 41s\n- gate-os-cloud-config-python26 http://logs.openstack.org/49/99849/1/check/gate-os-cloud-config-python26/854be37 : SUCCESS in 3m 28s\n- gate-os-cloud-config-python27 http://logs.openstack.org/49/99849/1/check/gate-os-cloud-config-python27/6386ae9 : SUCCESS in 3m 37s","accounts_in_message":[],"_revision_number":1},{"id":"e2051c3d65307bec4c55e9c8be512db6b1eede7d","author":{"_account_id":4190,"name":"lifeless","email":"robertc@robertcollins.net","username":"lifeless"},"date":"2014-06-13 09:31:26.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b596a0f308faa51d15ff8c3c42267a1f3387e1de","author":{"_account_id":4330,"name":"Tomas Sedovic","email":"tsedovic@redhat.com","username":"tomas-8c8"},"date":"2014-06-13 09:41:00.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"eda5d938454788ef0db9afafc376656fdc50e837","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-06-14 01:45:28.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- check-tripleo-ironic-seed-precise http://logs.openstack.org/49/99849/1/check-tripleo/check-tripleo-ironic-seed-precise/24dfc75 : SUCCESS in 35m 54s\n- check-tripleo-undercloud-precise http://logs.openstack.org/49/99849/1/check-tripleo/check-tripleo-undercloud-precise/b307561 : SUCCESS in 47m 10s\n- check-tripleo-overcloud-f20 http://logs.openstack.org/49/99849/1/check-tripleo/check-tripleo-overcloud-f20/e9d4521 : SUCCESS in 1h 26m 59s","accounts_in_message":[],"_revision_number":1},{"id":"b175556b75a2937f99eac7236438b3ad62a066b6","author":{"_account_id":4330,"name":"Tomas Sedovic","email":"tsedovic@redhat.com","username":"tomas-8c8"},"date":"2014-06-16 08:15:29.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9ea67ab8d8faec11dda35ba52c42121a8072d8b2","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-06-16 08:15:30.000000000","message":"Patch Set 1: -Verified\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"8cc42f4f069ccafcd163613ccc29b1556a0a0e05","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-06-16 08:19:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"677a336e351c6da697e722a4dd3135028f99f45d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-06-16 08:19:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded.\n\n- gate-os-cloud-config-pep8 http://logs.openstack.org/49/99849/1/check/gate-os-cloud-config-pep8/bd84379 : SUCCESS in 3m 02s\n- gate-os-cloud-config-python26 http://logs.openstack.org/49/99849/1/check/gate-os-cloud-config-python26/66c1dcf : SUCCESS in 2m 27s\n- gate-os-cloud-config-python27 http://logs.openstack.org/49/99849/1/check/gate-os-cloud-config-python27/17c81d5 : SUCCESS in 3m 29s","accounts_in_message":[],"_revision_number":1},{"id":"441ec1ca2ba4cc42d0d3f09cadea1c8c3e6cc348","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-06-16 08:22:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded.\n\n- gate-os-cloud-config-pep8 http://logs.openstack.org/49/99849/1/gate/gate-os-cloud-config-pep8/943af5e : SUCCESS in 2m 11s\n- gate-os-cloud-config-python26 http://logs.openstack.org/49/99849/1/gate/gate-os-cloud-config-python26/c7b7e6a : SUCCESS in 3m 16s\n- gate-os-cloud-config-python27 http://logs.openstack.org/49/99849/1/gate/gate-os-cloud-config-python27/ef48baa : SUCCESS in 3m 13s","accounts_in_message":[],"_revision_number":1},{"id":"d1f5c7cbe3ca7199f31588f8263d293e7a233430","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-06-16 08:22:22.000000000","message":"Change has been successfully merged into the git repository.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3a217fa80fbe7fbf4aef13702d9782e24ca7338e","revisions":{"3a217fa80fbe7fbf4aef13702d9782e24ca7338e":{"kind":"REWORK","_number":1,"created":"2014-06-13 05:55:52.000000000","uploader":{"_account_id":9369,"name":"Steve Kowalik","email":"steven@wedontsleep.org","username":"stevenk"},"ref":"refs/changes/49/99849/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-cloud-config","ref":"refs/changes/49/99849/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-cloud-config refs/changes/49/99849/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-cloud-config refs/changes/49/99849/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-cloud-config refs/changes/49/99849/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-cloud-config refs/changes/49/99849/1"}}},"commit":{"parents":[{"commit":"d79eb04f6d1cee51119aa1c0d8720239f4cc1b81","subject":"Add logging to os_cloud_config/nodes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-cloud-config/commit/d79eb04f6d1cee51119aa1c0d8720239f4cc1b81"}]}],"author":{"name":"Steve Kowalik","email":"steven@wedontsleep.org","date":"2014-06-13 05:52:55.000000000","tz":600},"committer":{"name":"Steve Kowalik","email":"steven@wedontsleep.org","date":"2014-06-13 05:52:55.000000000","tz":600},"subject":"Ignore Conflict exception when setting power state","message":"Ignore Conflict exception when setting power state\n\nThe call to set_power_state() when registering an Ironic node can\nfail if the Ironic conductor has grabbed a lock first and started\nits work to assert power state. Since the conductor will power off\nthe node, we can safely ignore the exception.\n\nChange-Id: I77eef1654b025e33b1b54d35f020d0b61d739e5a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-cloud-config/commit/3a217fa80fbe7fbf4aef13702d9782e24ca7338e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-cloud-config/commit/3a217fa80fbe7fbf4aef13702d9782e24ca7338e"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"d79eb04f6d1cee51119aa1c0d8720239f4cc1b81","is_merged_in_target_branch":true,"change_id":"Idddb2e9eb391973c50ee06a7bcb62986eba177f8","change_number":96051,"patch_set_number":6,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
