)]}'
{"id":"openstack%2Fos-ken~655049","triplet_id":"openstack%2Fos-ken~master~I0449513aabe3fa5da7774e216a5e854c0c123803","project":"openstack/os-ken","branch":"master","hashtags":[],"change_id":"I0449513aabe3fa5da7774e216a5e854c0c123803","subject":"Replace git.openstack.org URLs with opendev.org URLs","status":"ABANDONED","created":"2019-04-23 11:39:41.000000000","updated":"2020-01-28 14:34:39.000000000","total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7507bb6e1d2b40928490f19de0eb59c17ec42a66","_number":655049,"virtual_id_number":655049,"owner":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":1,"date":"2019-08-05 03:32:56.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":-1,"date":"2019-09-17 06:36:26.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-10 09:37:57.000000000","updated_by":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"reviewer":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"state":"REVIEWER"},{"updated":"2019-05-29 14:14:08.000000000","updated_by":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"reviewer":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"state":"REVIEWER"},{"updated":"2019-06-07 14:19:06.000000000","updated_by":{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},"reviewer":{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},"state":"REVIEWER"},{"updated":"2019-06-09 16:00:05.000000000","updated_by":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"reviewer":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"state":"REVIEWER"},{"updated":"2019-06-09 16:32:09.000000000","updated_by":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"reviewer":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"state":"REVIEWER"},{"updated":"2019-06-12 02:47:01.000000000","updated_by":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"reviewer":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"state":"REVIEWER"},{"updated":"2019-08-03 20:00:41.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2019-08-03 20:00:41.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2019-08-03 20:00:41.000000000","updated_by":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"reviewer":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"state":"REVIEWER"},{"updated":"2019-08-03 20:00:41.000000000","updated_by":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"reviewer":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"state":"REVIEWER"},{"updated":"2019-08-03 20:00:41.000000000","updated_by":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"reviewer":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"state":"REVIEWER"},{"updated":"2019-08-04 14:18:30.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2019-08-05 03:32:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-17 06:36:26.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"}],"messages":[{"id":"7b96c32639f4c64272d157573f85d3fd0008c6a6","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-04-23 11:39:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"284dae71d7b7ca20b2c66f696f39bd3067ae9301","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 12:12:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/49/655049/1/check/test-release-openstack/6619efd/ : SUCCESS in 3m 45s\n- openstack-tox-lower-constraints http://logs.openstack.org/49/655049/1/check/openstack-tox-lower-constraints/c533deb/ : SUCCESS in 6m 05s\n- openstack-tox-pep8 http://logs.openstack.org/49/655049/1/check/openstack-tox-pep8/7f5109e/ : SUCCESS in 5m 13s\n- openstack-tox-py27 http://logs.openstack.org/49/655049/1/check/openstack-tox-py27/929ff94/ : SUCCESS in 5m 13s\n- openstack-tox-py35 http://logs.openstack.org/49/655049/1/check/openstack-tox-py35/4293fc4/ : SUCCESS in 5m 37s\n- openstack-tox-py36 http://logs.openstack.org/49/655049/1/check/openstack-tox-py36/8eee8d7/ : SUCCESS in 5m 29s\n- openstack-tox-docs http://logs.openstack.org/49/655049/1/check/openstack-tox-docs/84dcdd6/html/ : SUCCESS in 4m 55s\n- build-openstack-releasenotes http://logs.openstack.org/49/655049/1/check/build-openstack-releasenotes/d503a71/html/ : SUCCESS in 3m 36s","accounts_in_message":[],"_revision_number":1},{"id":"86c14c0e32cebf8a907bbb7ce28d3c0545e070a5","author":{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},"date":"2019-05-07 23:22:55.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6ff96139fe7567196f9d6eb32a8bf42440784333","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-05-08 13:24:23.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"42fd519f1fd86d5a7f134f8e76dabcd1defa3d31","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-05-08 13:28:11.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"1f158dfeb4b942c6165a5280c2deec52f8a3e1de","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-08 13:39:34.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/49/655049/3/check/test-release-openstack/1c0a099/ : SUCCESS in 3m 16s\n- openstack-tox-lower-constraints http://logs.openstack.org/49/655049/3/check/openstack-tox-lower-constraints/f572272/ : SUCCESS in 5m 16s\n- openstack-tox-pep8 http://logs.openstack.org/49/655049/3/check/openstack-tox-pep8/499adb6/ : SUCCESS in 5m 04s\n- openstack-tox-py27 http://logs.openstack.org/49/655049/3/check/openstack-tox-py27/88138f0/ : SUCCESS in 5m 59s\n- openstack-tox-py36 http://logs.openstack.org/49/655049/3/check/openstack-tox-py36/792b4b8/ : SUCCESS in 5m 40s\n- openstack-tox-py37 http://logs.openstack.org/49/655049/3/check/openstack-tox-py37/a8ac881/ : SUCCESS in 5m 35s\n- openstack-tox-docs http://logs.openstack.org/49/655049/3/check/openstack-tox-docs/86bf5bf/html/ : SUCCESS in 4m 43s\n- build-openstack-releasenotes http://logs.openstack.org/49/655049/3/check/build-openstack-releasenotes/93ede89/html/ : SUCCESS in 3m 34s","accounts_in_message":[],"_revision_number":3},{"id":"ffc0772573222b8468749420de673bd29552d69a","author":{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},"date":"2019-05-08 13:58:31.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"5caef2682bb17730c826ba2d311c5cec6bcb579a","author":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"date":"2019-05-09 02:50:50.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"62eaf0e63db49029a07ab043988fae0cf88de565","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2019-05-10 09:37:57.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"bf60889d7dd20ec875aee975d888773cd34d15b5","author":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"date":"2019-05-20 03:15:06.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"cb5b7725c8434b03a96ad46e7adf790d2527d5fb","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2019-05-23 08:57:38.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"73ac7b2b8b632e14139826fd06b17f3d3dcfe96f","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-05-27 05:32:22.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"1d085c062cb1a88e0e53ffdda25aeed8c07821ad","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-27 05:43:16.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/49/655049/4/check/test-release-openstack/8ff3811/ : SUCCESS in 3m 24s\n- openstack-tox-lower-constraints http://logs.openstack.org/49/655049/4/check/openstack-tox-lower-constraints/208b9e6/ : SUCCESS in 5m 20s\n- openstack-tox-pep8 http://logs.openstack.org/49/655049/4/check/openstack-tox-pep8/77eb2a3/ : SUCCESS in 4m 30s\n- openstack-tox-py27 http://logs.openstack.org/49/655049/4/check/openstack-tox-py27/2224954/ : SUCCESS in 5m 26s\n- openstack-tox-py36 http://logs.openstack.org/49/655049/4/check/openstack-tox-py36/f454adb/ : SUCCESS in 5m 29s\n- openstack-tox-py37 http://logs.openstack.org/49/655049/4/check/openstack-tox-py37/a258169/ : SUCCESS in 5m 20s\n- openstack-tox-docs http://logs.openstack.org/49/655049/4/check/openstack-tox-docs/cb08baf/html/ : SUCCESS in 4m 41s\n- build-openstack-releasenotes http://logs.openstack.org/49/655049/4/check/build-openstack-releasenotes/e5048c6/html/ : SUCCESS in 3m 45s","accounts_in_message":[],"_revision_number":4},{"id":"ca5e3af9cd66258e576af6b62d748f2800df5643","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-05-29 05:33:11.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"1b81f71b321f11197b62c87d91cfbc75678b3c13","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-29 06:17:49.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/49/655049/5/check/test-release-openstack/9081e78/ : SUCCESS in 3m 06s\n- openstack-tox-lower-constraints http://logs.openstack.org/49/655049/5/check/openstack-tox-lower-constraints/27b5ec5/ : SUCCESS in 5m 23s\n- openstack-tox-pep8 http://logs.openstack.org/49/655049/5/check/openstack-tox-pep8/548721b/ : SUCCESS in 4m 13s\n- openstack-tox-py27 http://logs.openstack.org/49/655049/5/check/openstack-tox-py27/9a6744c/ : SUCCESS in 5m 46s\n- openstack-tox-py36 http://logs.openstack.org/49/655049/5/check/openstack-tox-py36/22ef4da/ : SUCCESS in 5m 01s\n- openstack-tox-py37 http://logs.openstack.org/49/655049/5/check/openstack-tox-py37/3af8106/ : SUCCESS in 5m 12s\n- openstack-tox-docs http://logs.openstack.org/49/655049/5/check/openstack-tox-docs/a62f5a3/html/ : SUCCESS in 7m 00s\n- build-openstack-releasenotes http://logs.openstack.org/49/655049/5/check/build-openstack-releasenotes/8c8aa15/html/ : SUCCESS in 3m 27s","accounts_in_message":[],"_revision_number":5},{"id":"9dd10fbf7be25ed7f7b9b1edc7e7c7847f8b0d2c","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2019-05-29 14:14:08.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"a9b8cfde30e608e4fb6ae147aee00cc961e1508b","author":{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},"date":"2019-06-07 14:19:06.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"ce2a6d5c1ba305cb51df03d506b05b050271754a","author":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"date":"2019-06-09 16:00:05.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"b06a76212449b347f2199393ab5ae72b9ad47d61","author":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"date":"2019-06-09 16:32:09.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"c550026d661818c8bf0574f13416fe57f1f22619","author":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"date":"2019-06-12 02:47:01.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"4c5c194ad1e60e4871a8e7eb37a50e1148427bca","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-08-04 14:18:30.000000000","message":"Patch Set 5: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"1aae83b9da0384e724028c6c9cda4348cd33796e","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-08-05 03:26:55.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"1e822da999d38cc6debd74e9b9477851f8832781","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 03:32:56.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://logs.opendev.org/49/655049/6/check/test-release-openstack/2c64a48/ : SUCCESS in 2m 13s\n- openstack-tox-lower-constraints https://logs.opendev.org/49/655049/6/check/openstack-tox-lower-constraints/d628483/ : SUCCESS in 4m 35s\n- openstack-tox-pep8 https://logs.opendev.org/49/655049/6/check/openstack-tox-pep8/746441e/ : SUCCESS in 3m 38s\n- openstack-tox-py27 https://logs.opendev.org/49/655049/6/check/openstack-tox-py27/9823bc6/ : SUCCESS in 4m 30s\n- openstack-tox-py36 https://logs.opendev.org/49/655049/6/check/openstack-tox-py36/f642905/ : SUCCESS in 3m 34s\n- openstack-tox-py37 https://logs.opendev.org/49/655049/6/check/openstack-tox-py37/d66a684/ : SUCCESS in 4m 17s\n- openstack-tox-docs https://logs.opendev.org/49/655049/6/check/openstack-tox-docs/dc4a85e/html/ : SUCCESS in 2m 38s\n- build-openstack-releasenotes https://logs.opendev.org/49/655049/6/check/build-openstack-releasenotes/b1f0406/html/ : SUCCESS in 2m 45s","accounts_in_message":[],"_revision_number":6},{"id":"f5814d224aff13edba74287b307ef436e02035d7","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-08-05 11:51:42.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"f48fe43ddfb92ef4c5de71289aff0c8752e00bc3","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-09-17 06:36:26.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)\n\n-1 due to no response from the author.","accounts_in_message":[],"_revision_number":6},{"id":"7507bb6e1d2b40928490f19de0eb59c17ec42a66","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2020-01-28 14:34:39.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"6370134f5ed6418aa6579a80afc69c5c8bc5ed90","revisions":{"df455aa7bed345ade5c5ce64b3ad57f691f6cd59":{"kind":"REWORK","_number":1,"created":"2019-04-23 11:39:41.000000000","uploader":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"ref":"refs/changes/49/655049/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-ken","ref":"refs/changes/49/655049/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-ken refs/changes/49/655049/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-ken refs/changes/49/655049/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-ken refs/changes/49/655049/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-ken refs/changes/49/655049/1"}}},"commit":{"parents":[{"commit":"379a7694c3129cc0156343af71f4fca8830d9de5","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-ken/commit/379a7694c3129cc0156343af71f4fca8830d9de5"}]}],"author":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-04-23 11:39:27.000000000","tz":480},"committer":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-04-23 11:39:27.000000000","tz":480},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nChange-Id: I0449513aabe3fa5da7774e216a5e854c0c123803\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-ken/commit/df455aa7bed345ade5c5ce64b3ad57f691f6cd59"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-ken/commit/df455aa7bed345ade5c5ce64b3ad57f691f6cd59"}]},"branch":"refs/heads/master"},"203415d158330bf6928ae99fe2ca2c6a662f883c":{"kind":"REWORK","_number":2,"created":"2019-05-08 13:24:23.000000000","uploader":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"ref":"refs/changes/49/655049/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-ken","ref":"refs/changes/49/655049/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-ken refs/changes/49/655049/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-ken refs/changes/49/655049/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-ken refs/changes/49/655049/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-ken refs/changes/49/655049/2"}}},"commit":{"parents":[{"commit":"379a7694c3129cc0156343af71f4fca8830d9de5","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-ken/commit/379a7694c3129cc0156343af71f4fca8830d9de5"}]}],"author":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-04-23 11:39:27.000000000","tz":480},"committer":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-05-08 13:23:31.000000000","tz":480},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\n1. Replace git.openstack.org URLs with opendev.org URLs\n2. Remove unncessary \"\u003d\u003d\"\n3. Fix the wrong bug url\n\nChange-Id: I0449513aabe3fa5da7774e216a5e854c0c123803\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-ken/commit/203415d158330bf6928ae99fe2ca2c6a662f883c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-ken/commit/203415d158330bf6928ae99fe2ca2c6a662f883c"}]},"branch":"refs/heads/master"},"d99de1a8d54b7b4fd8b9baf65cb31e676391806d":{"kind":"REWORK","_number":3,"created":"2019-05-08 13:28:11.000000000","uploader":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"ref":"refs/changes/49/655049/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-ken","ref":"refs/changes/49/655049/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-ken refs/changes/49/655049/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-ken refs/changes/49/655049/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-ken refs/changes/49/655049/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-ken refs/changes/49/655049/3"}}},"commit":{"parents":[{"commit":"379a7694c3129cc0156343af71f4fca8830d9de5","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-ken/commit/379a7694c3129cc0156343af71f4fca8830d9de5"}]}],"author":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-04-23 11:39:27.000000000","tz":480},"committer":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-05-08 13:27:52.000000000","tz":480},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\n1. Replace git.openstack.org URLs with opendev.org URLs\n2. Remove unncessary \"\u003d\u003d\"\n3. Fix the wrong bug url\n4. Update the home page\n\nChange-Id: I0449513aabe3fa5da7774e216a5e854c0c123803\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-ken/commit/d99de1a8d54b7b4fd8b9baf65cb31e676391806d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-ken/commit/d99de1a8d54b7b4fd8b9baf65cb31e676391806d"}]},"branch":"refs/heads/master"},"c5449307fdbf34e5a23a4f78ce92a625e0ecd00e":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-05-27 05:32:22.000000000","uploader":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"ref":"refs/changes/49/655049/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-ken","ref":"refs/changes/49/655049/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-ken refs/changes/49/655049/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-ken refs/changes/49/655049/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-ken refs/changes/49/655049/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-ken refs/changes/49/655049/4"}}},"commit":{"parents":[{"commit":"b9e3c6748b0fe110c5538c8820ca38d9baec1c1c","subject":"Dropping the py35 testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-ken/commit/b9e3c6748b0fe110c5538c8820ca38d9baec1c1c"}]}],"author":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-04-23 11:39:27.000000000","tz":480},"committer":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-05-27 05:32:22.000000000","tz":0},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\n1. Replace git.openstack.org URLs with opendev.org URLs\n2. Remove unncessary \"\u003d\u003d\"\n3. Fix the wrong bug url\n4. Update the home page\n\nChange-Id: I0449513aabe3fa5da7774e216a5e854c0c123803\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-ken/commit/c5449307fdbf34e5a23a4f78ce92a625e0ecd00e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-ken/commit/c5449307fdbf34e5a23a4f78ce92a625e0ecd00e"}]},"branch":"refs/heads/master"},"4c50fb3451a7ed0e67d8ee638cc8f298aacf81b4":{"kind":"REWORK","_number":5,"created":"2019-05-29 05:33:11.000000000","uploader":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"ref":"refs/changes/49/655049/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-ken","ref":"refs/changes/49/655049/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-ken refs/changes/49/655049/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-ken refs/changes/49/655049/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-ken refs/changes/49/655049/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-ken refs/changes/49/655049/5"}}},"commit":{"parents":[{"commit":"2ba552d5278e26c3fc32d9acf9fa3255227cc202","subject":"Merge \"Change openstack-dev to openstack-discuss\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-ken/commit/2ba552d5278e26c3fc32d9acf9fa3255227cc202"}]}],"author":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-04-23 11:39:27.000000000","tz":480},"committer":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-05-29 05:33:00.000000000","tz":480},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\n1. Replace git.openstack.org URLs with opendev.org URLs\n2. Remove unncessary \"\u003d\u003d\"\n3. Fix the wrong bug url\n4. Update the home page\n\nChange-Id: I0449513aabe3fa5da7774e216a5e854c0c123803\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-ken/commit/4c50fb3451a7ed0e67d8ee638cc8f298aacf81b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-ken/commit/4c50fb3451a7ed0e67d8ee638cc8f298aacf81b4"}]},"branch":"refs/heads/master"},"6370134f5ed6418aa6579a80afc69c5c8bc5ed90":{"kind":"REWORK","_number":6,"created":"2019-08-05 03:26:55.000000000","uploader":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"ref":"refs/changes/49/655049/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-ken","ref":"refs/changes/49/655049/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-ken refs/changes/49/655049/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-ken refs/changes/49/655049/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-ken refs/changes/49/655049/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-ken refs/changes/49/655049/6"}}},"commit":{"parents":[{"commit":"2ba552d5278e26c3fc32d9acf9fa3255227cc202","subject":"Merge \"Change openstack-dev to openstack-discuss\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-ken/commit/2ba552d5278e26c3fc32d9acf9fa3255227cc202"}]}],"author":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-04-23 11:39:27.000000000","tz":480},"committer":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-08-05 03:26:49.000000000","tz":0},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\n1. Replace git.openstack.org URLs with opendev.org URLs\n2. Remove unncessary \"\u003d\u003d\"\n3. Fix the wrong bug url\n4. Update the home page\n\nChange-Id: I0449513aabe3fa5da7774e216a5e854c0c123803\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-ken/commit/6370134f5ed6418aa6579a80afc69c5c8bc5ed90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-ken/commit/6370134f5ed6418aa6579a80afc69c5c8bc5ed90"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
