)]}'
{"id":"openstack%2Fos-ken~748876","triplet_id":"openstack%2Fos-ken~master~I9658ebaa911f129f567197ff7ad531e093f9752f","project":"openstack/os-ken","branch":"master","hashtags":[],"change_id":"I9658ebaa911f129f567197ff7ad531e093f9752f","subject":"Clean python2 remained stuff","status":"MERGED","created":"2020-08-30 15:12:57.000000000","updated":"2020-09-02 20:05:51.000000000","submitted":"2020-09-02 20:01:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"748876-1599076878235-8a9bf4a0","meta_rev_id":"2b8e119cf8a02213d2a6107eea7d07cbd9673f89","_number":748876,"virtual_id_number":748876,"owner":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-02 20:01:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":2,"date":"2020-08-31 11:39:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2020-09-02 19:49:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2020-09-02 19:49:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-31 11:39:23.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-09-02 19:49:14.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-09-02 20:01:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2d27f5f218721e663b3052556ecf7cf3068d2dc2","author":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"date":"2020-08-30 15:12:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0fadee0b08e5ec8562478627f467a6a902a69b38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-30 15:19:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/131f479c1dc24f9aba0cf899b00ea00d : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de595fa70e7741978c07c615205f2d4d : SUCCESS in 3m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/460f16fbade24d219aa1e496229d6d03 : SUCCESS in 3m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c8ef7d9291b64121a28858f8247217e5 : SUCCESS in 4m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9dcd69312b447fa802a65918be61092 : SUCCESS in 6m 33s","accounts_in_message":[],"_revision_number":1},{"id":"ea02e03de51654c884339a71d6f852b57315b735","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-08-31 11:39:23.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"01d68ca5272b3e3589449e7f11d5fd791ec3ecc6","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-09-02 19:49:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3adeb7528591e92f142354f0612afa1b5531422a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-02 19:49:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3cfe57ed903e1a08661b3e4d598fea5880cab96f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-02 20:01:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e5c86b160d6847c2a3cce86b4d992524 : SUCCESS in 6m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1c4f43d5cbc4484b8b448ed2e9c645f : SUCCESS in 3m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/79abe1ef3e774d80b45af2135fe136a2 : SUCCESS in 5m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5533e3e3adbb4094805e7867d2dd219b : SUCCESS in 7m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a66d259c332245c48624c8f2287d3b44 : SUCCESS in 7m 46s","accounts_in_message":[],"_revision_number":1},{"id":"a507c1064de7e21fd137f478b579f9e578baa150","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-02 20:01:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"2b8e119cf8a02213d2a6107eea7d07cbd9673f89","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-02 20:05:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c6e4a6466234bd3a0b7335c8109a78d : SUCCESS in 2m 12s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"195ec82629425af180309a18e86cacb3bb8e37f5","revisions":{"195ec82629425af180309a18e86cacb3bb8e37f5":{"kind":"REWORK","_number":1,"created":"2020-08-30 15:12:57.000000000","uploader":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"ref":"refs/changes/76/748876/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-ken","ref":"refs/changes/76/748876/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-ken refs/changes/76/748876/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-ken refs/changes/76/748876/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-ken refs/changes/76/748876/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-ken refs/changes/76/748876/1"}}},"commit":{"parents":[{"commit":"9c04dc0c6f86f7984a47eb7250e8250ba322241a","subject":"Update some URLs in README and tox","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-ken/commit/9c04dc0c6f86f7984a47eb7250e8250ba322241a"}]}],"author":{"name":"gugug","email":"gu.jin@99cloud.net","date":"2020-08-30 15:12:19.000000000","tz":480},"committer":{"name":"gugug","email":"gu.jin@99cloud.net","date":"2020-08-30 15:12:55.000000000","tz":480},"subject":"Clean python2 remained stuff","message":"Clean python2 remained stuff\n\nChange-Id: I9658ebaa911f129f567197ff7ad531e093f9752f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-ken/commit/195ec82629425af180309a18e86cacb3bb8e37f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-ken/commit/195ec82629425af180309a18e86cacb3bb8e37f5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
