)]}'
{"id":"openstack%2Fos-ken~749039","triplet_id":"openstack%2Fos-ken~master~I3e5db6b6c2a3147f96f759948be9a0fcba184209","project":"openstack/os-ken","branch":"master","topic":"improve-ci","hashtags":[],"change_id":"I3e5db6b6c2a3147f96f759948be9a0fcba184209","subject":"Add basic tempest-integrated-networking job to os-ken ci","status":"MERGED","created":"2020-08-31 11:45:06.000000000","updated":"2020-09-16 19:20:23.000000000","submitted":"2020-09-16 19:14:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"749039-1600283691345-57e98530","meta_rev_id":"72d33645a74721b44fb98191f84678acb5446d88","_number":749039,"virtual_id_number":749039,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-16 19:14:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"recommended":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":1,"date":"2020-09-16 12:11:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2020-09-16 17:58:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2020-09-16 12:11:00.000000000","_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":2,"date":"2020-09-16 12:11:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2020-09-16 17:58:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-31 12:36:20.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2020-09-03 02:20:51.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-09-15 07:13:59.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2020-09-16 17:58:49.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2020-09-16 19:14:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"453d64bde62eaccb034665da56cf8d68234db66a","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-08-31 11:45:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"39e49386727651ed1c29a16c9316043371f8ebba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-31 12:06:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5e364f1a5884447a94cb885cb297097d : SUCCESS in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af64767b5bf0476f9b74f63bc5b8dcba : SUCCESS in 3m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b0d0d2dded5c49a999da786da377d827 : SUCCESS in 3m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/56e22c33e1a0460794ca1ad6605b0aa8 : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f97df1422774f2fa7074ee949599734 : SUCCESS in 5m 34s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/c179a3fafe114f0cb771dce9b96fd74c : FAILURE in 17m 22s","accounts_in_message":[],"_revision_number":1},{"id":"0c39383b019ffda88470d6991aaf2705ec57da80","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-08-31 12:36:20.000000000","message":"Patch Set 1: Code-Review+1\n\nGood idea\nI suppose the \"keystone-manage: No such file or directory\" failure is setuptools 50","accounts_in_message":[],"_revision_number":1},{"id":"cca62c827f8ebb66e9730a4ee2c07718233a5fa3","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-08-31 12:57:38.000000000","message":"Patch Set 1:\n\n\u003e Good idea\n \u003e I suppose the \"keystone-manage: No such file or directory\" failure\n \u003e is setuptools 50\n\nYes, and that is failing everywhere now","accounts_in_message":[],"_revision_number":1},{"id":"bd226c70d36105770fd7f9031e4d01cee4acdbbd","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-09-02 19:41:13.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0aeab20396da72b978cb6c8c647f1bcab62fb13e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-02 21:21:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c9d9326312ec4895a5ff4d322489ff92 : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35aad84e40334f1eb03256a643ea94a9 : SUCCESS in 4m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bc19ca5e443041de8afbd3ab1a487562 : SUCCESS in 4m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b889b4b21dbc44b682a92f6c770a87ca : SUCCESS in 5m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67a96944bc814c8991df306e06b845a5 : SUCCESS in 9m 52s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/bac2f9cc1df342f8b062b65884d04ad2 : SUCCESS in 1h 16m 52s","accounts_in_message":[],"_revision_number":1},{"id":"b1a4464843220441c89b238af931942430d38809","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-09-03 02:20:51.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ab02fde104447de6cdda7a32e45c537f272976f0","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-09-15 07:13:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"82bb5e53f4087d536b724e75b1283c1dbd03793d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-15 07:14:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"94f2920b1096e2226eb17035b55b93713f19d5cf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-15 08:17:08.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6558085c96a648699aff5a82dc9c3130 : FAILURE in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aeb9c0cdf3f54c5fbcb1b4baa3d9f7df : FAILURE in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b535221bed804a5c9899b3c2866768bc : SUCCESS in 4m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9e6b5083e46c4bc3a39906b559095b44 : SUCCESS in 5m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66f75854ff2a492f92a7314f42ba787a : SUCCESS in 15m 20s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/9a8ee03bd5dd4e5bbaba05454c06b773 : SUCCESS in 1h 00m 08s","accounts_in_message":[],"_revision_number":1},{"id":"b4c40a9b9a022a7d9832cf9d507da65f514e64b3","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-09-16 10:37:23.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e3db8a313fe438e2ec4430d2949d73ef6d6bdb0e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 11:43:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/03427c65d13b4e34ad00ca1650a6fe7c : FAILURE in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ba80e44ea744b7da20facb42c0874a7 : FAILURE in 7m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/47ffb963cd024d289b72e6fdcd73558b : SUCCESS in 5m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2e05fbe567154d9b80948626668af8c3 : SUCCESS in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69dc4ddf9ef64cecb6b881624c8a1f67 : SUCCESS in 16m 21s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/b6fd84e9a895454e8410a47294adea15 : SUCCESS in 1h 03m 02s","accounts_in_message":[],"_revision_number":1},{"id":"08f73ff0c3be7fe7a5f12cc03a3f98e7a8d5bf1f","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-09-16 12:11:00.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"e0350d3f9f860e81307f3aa7a259945d312d1db6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 13:40:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6058ea5d414d41f8bae38d7e937a1900 : SUCCESS in 5m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/692e15cd62fe49c5ad6a0ce20e7ef9d0 : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a1388399c03d4931a7280ae66e4d0300 : SUCCESS in 6m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/41021b4eda4f42afa275f6b1522d846a : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbfa9b79adaf48e4bd320e366b509b5c : SUCCESS in 15m 11s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/d5507282d5ff4c50a382b6a8420fe544 : SUCCESS in 1h 15m 23s","accounts_in_message":[],"_revision_number":2},{"id":"ee14edd362e2db7d60971085a4ee12e940b2e93a","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2020-09-16 17:58:49.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"1fc957bd9eae3e81ab0949591dd790f51841b970","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 17:59:06.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c7d49e7814791c423e00fcbbd824df731f1d992d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 19:14:51.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e99c830f409644bdba109371e9035595 : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a818987ad4c4de7a6752ebab51a0def : SUCCESS in 3m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/917f9e57d31c455aab7fa03c588b6637 : SUCCESS in 6m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cce5f27d6a9f46f59d41457dfbb54170 : SUCCESS in 4m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18d21c1ba4604e4fba3587356ca38c87 : SUCCESS in 14m 26s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/2fff8c6bf12b48a48ebef36e51aa9f41 : SUCCESS in 1h 06m 36s","accounts_in_message":[],"_revision_number":2},{"id":"36fae24ebc44c9d8e11dcbf09ef51a6916a3b0a8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 19:14:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"f156cf1f60843b1c4330e0e0d16c9b9a25eb0469","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 19:20:23.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59b6343b35cd4abb9374421a9d538056 : SUCCESS in 1m 22s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"43e7848b18ec0310ffe5afe5e1f8374a72ce07c9","revisions":{"75b1870c4b525c5e99738521b2c9ac18e9be7638":{"kind":"REWORK","_number":1,"created":"2020-08-31 11:45:06.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/39/749039/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-ken","ref":"refs/changes/39/749039/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-ken refs/changes/39/749039/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-ken refs/changes/39/749039/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-ken refs/changes/39/749039/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-ken refs/changes/39/749039/1"}}},"commit":{"parents":[{"commit":"9c04dc0c6f86f7984a47eb7250e8250ba322241a","subject":"Update some URLs in README and tox","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-ken/commit/9c04dc0c6f86f7984a47eb7250e8250ba322241a"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-08-31 11:42:41.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-08-31 11:44:52.000000000","tz":120},"subject":"Add basic tempest-integrated-networking job to os-ken ci","message":"Add basic tempest-integrated-networking job to os-ken ci\n\nOs-ken is library which us used a lot in the Neutron so we should\nbe sure that new patches in os-ken will not break Neutron.\nTo ensure that this patch adds tempest-integrated-networking job to the\ncheck and gate queues in os-ken.\n\nChange-Id: I3e5db6b6c2a3147f96f759948be9a0fcba184209\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-ken/commit/75b1870c4b525c5e99738521b2c9ac18e9be7638"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-ken/commit/75b1870c4b525c5e99738521b2c9ac18e9be7638"}]},"branch":"refs/heads/master"},"43e7848b18ec0310ffe5afe5e1f8374a72ce07c9":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-09-16 12:11:00.000000000","uploader":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"ref":"refs/changes/39/749039/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-ken","ref":"refs/changes/39/749039/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-ken refs/changes/39/749039/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-ken refs/changes/39/749039/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-ken refs/changes/39/749039/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-ken refs/changes/39/749039/2"}}},"commit":{"parents":[{"commit":"f61da64144a9db5286614f86ad270f30235f4fa5","subject":"Fix lower-constraints and pep8 for Focal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-ken/commit/f61da64144a9db5286614f86ad270f30235f4fa5"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-08-31 11:42:41.000000000","tz":120},"committer":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2020-09-16 12:11:00.000000000","tz":0},"subject":"Add basic tempest-integrated-networking job to os-ken ci","message":"Add basic tempest-integrated-networking job to os-ken ci\n\nOs-ken is library which us used a lot in the Neutron so we should\nbe sure that new patches in os-ken will not break Neutron.\nTo ensure that this patch adds tempest-integrated-networking job to the\ncheck and gate queues in os-ken.\n\nChange-Id: I3e5db6b6c2a3147f96f759948be9a0fcba184209\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-ken/commit/43e7848b18ec0310ffe5afe5e1f8374a72ce07c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-ken/commit/43e7848b18ec0310ffe5afe5e1f8374a72ce07c9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
