)]}'
{"id":"openstack%2Fos-ken~777732","triplet_id":"openstack%2Fos-ken~master~Ic8ab137b724c518f9ea79c33045e6d6c19fad984","project":"openstack/os-ken","branch":"master","topic":"bug/1907089","hashtags":[],"change_id":"Ic8ab137b724c518f9ea79c33045e6d6c19fad984","subject":"Add remove_kill_bfd_session method to BFDLib","status":"ABANDONED","created":"2021-02-26 09:43:34.000000000","updated":"2024-03-26 14:30:45.000000000","total_comment_count":7,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"68b746f120f5b53265857aa90b7ca01cc8a3a6ec","_number":777732,"virtual_id_number":777732,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":30506,"name":"Robert Tomczyk","email":"robert.tomczyk@est.tech","username":"robert.tomczyk"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-03-18 10:11:39.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30506,"name":"Robert Tomczyk","email":"robert.tomczyk@est.tech","username":"robert.tomczyk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":30506,"name":"Robert Tomczyk","email":"robert.tomczyk@est.tech","username":"robert.tomczyk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":30506,"name":"Robert Tomczyk","email":"robert.tomczyk@est.tech","username":"robert.tomczyk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30506,"name":"Robert Tomczyk","email":"robert.tomczyk@est.tech","username":"robert.tomczyk"}],"CC":[{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-26 10:55:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-09 12:12:31.000000000","updated_by":{"_account_id":30506,"name":"Robert Tomczyk","email":"robert.tomczyk@est.tech","username":"robert.tomczyk"},"reviewer":{"_account_id":30506,"name":"Robert Tomczyk","email":"robert.tomczyk@est.tech","username":"robert.tomczyk"},"state":"REVIEWER"},{"updated":"2021-03-09 15:41:46.000000000","updated_by":{"_account_id":30506,"name":"Robert Tomczyk","email":"robert.tomczyk@est.tech","username":"robert.tomczyk"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2021-03-11 12:36:35.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-03-11 12:36:35.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2021-03-11 12:36:35.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2024-03-26 14:30:45.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"state":"CC"}],"messages":[{"id":"de9e27550f1b8c9957aabc55bd745af3db64e5b4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-02-26 09:43:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5593cab0459e20efcf77948490e5503b530534b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 10:55:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/22301cf30913432e9174928de37ec91d : SUCCESS in 5m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90a56d2ca7854508a41dd55058400727 : SUCCESS in 3m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a10ec8b2f677492483df718ca51102f5 : SUCCESS in 5m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/48faa994cf4041ef9b2b96fae8d16883 : SUCCESS in 5m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d8b4a78d55b04739833461a7e0aae6eb : SUCCESS in 5m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4cf6d63b62b4f27903bc84904ab0ab7 : SUCCESS in 7m 00s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/cdf471445cb249ce9c39784accfd7621 : SUCCESS in 1h 07m 38s","accounts_in_message":[],"_revision_number":1},{"id":"8fe97c683bc6fa4b6eaec9205a4af1e9772488ab","author":{"_account_id":30506,"name":"Robert Tomczyk","email":"robert.tomczyk@est.tech","username":"robert.tomczyk"},"date":"2021-03-09 12:33:52.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"e5a168abc51133b319f385d99d2f4e0946d6532f","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-15 11:19:29.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"611e7c8e3753ef87cef16a40d20c178c8d14c4d4","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-03-15 13:04:12.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d2db83d4dc9b1bf394bcb56cc3078c3c4850ce91","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-03-15 13:08:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4e1543bf7f721898670cb6729892b850c75a36e7","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2021-03-15 15:25:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b8398d86ce18725281c12ebc9791c3d8840ce954","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-03-16 10:38:07.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"cb351e1f0d2d2816cf87baddcf73be911d49b052","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-03-16 10:39:16.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"9566682ad0697cac00e8415131fb58336d7877c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-16 11:53:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f788e3d6a0af4fdd908ea59c63e4521a : SUCCESS in 5m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9234fc6ecac24e1089c6cab6d32b3ed6 : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5dd8e37d0a2e49d7850b8efa3c2f0a43 : SUCCESS in 5m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5aa4626230504efaaaf26ea5232fd141 : SUCCESS in 4m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/67269ed9c1bf4480bdb221fa6c75b237 : SUCCESS in 4m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9d5c45ba55f42ab9d4cde514e1906a6 : SUCCESS in 7m 21s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/b344605b0f5a471b94e8d9b7df4fbd2b : SUCCESS in 1h 10m 13s","accounts_in_message":[],"_revision_number":2},{"id":"9c55040e723477e73e1fc3cf7a3e1e25ce028396","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2021-03-17 23:38:05.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"53ecec8fd60029c574c32d414d410056c7b50a0d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-03-18 08:20:54.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"e3712afab682c4c84947e272c489e3ad80980c45","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-03-18 08:22:58.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b0e571d778969f17315bab0c7fad9c27711a36e9","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2021-03-18 09:08:41.000000000","message":"Patch Set 3:\n\nI know os-ken UT coverage is not so good, but isn\u0027t it better to add some UT to cover this change?","accounts_in_message":[],"_revision_number":3},{"id":"c88160c042d65cdf4875e98b475213d895a91078","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 10:11:39.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e20864140363434194f3cc1de1cbca21 : SUCCESS in 5m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5fb3d89fb8df48e38ae84b598c9e1c48 : SUCCESS in 3m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d769cce0bf314922be60cb56e8d9caee : SUCCESS in 4m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/91627e8bc1c34ee48cd5eb2a057d7bfb : SUCCESS in 4m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/39661385269d48aea94d0051be03df06 : SUCCESS in 5m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8e8d4a0c99745b0b6ca7649d01a7c0d : SUCCESS in 8m 18s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/f0fa8526f7774a96b95c505868129f5d : SUCCESS in 1h 14m 33s","accounts_in_message":[],"_revision_number":3},{"id":"f4cc7cadffd428a74697f5cb1336e71f6418b15f","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2021-03-19 01:56:37.000000000","message":"Patch Set 3:\n\nrecheck my devstack patch which consumes os-ken from git repo has been merged. let me check it.","accounts_in_message":[],"_revision_number":3},{"id":"95692f4b231f6d89865f559bdeae558a4571e316","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-19 03:04:31.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0e798a9f6a2946fb9c2bd00316920359 : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57589a0195424e8784e4e060a258cc14 : SUCCESS in 3m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4a6482c61b9643b1b29c977ec5b572e4 : SUCCESS in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9fff6db763d94d019d13add1d19ff021 : SUCCESS in 4m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8a35ce8273b641b2be0062ea443b4cc3 : SUCCESS in 4m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5fcbb33c1b6543c0a67e8c09db258ce5 : SUCCESS in 6m 22s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/b44aae79fd664550864e5299e36fcd99 : SUCCESS in 1h 06m 53s","accounts_in_message":[],"_revision_number":3},{"id":"809e141c865bcec552d00223a021e646ef47ee51","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2021-03-19 03:25:55.000000000","message":"Patch Set 3:\n\nWe can now use os-ken from git in proposed changes :)\nhttps://zuul.opendev.org/t/openstack/build/b44aae79fd664550864e5299e36fcd99/log/controller/logs/pip3-freeze.txt#91","accounts_in_message":[],"_revision_number":3},{"id":"7b5dc685312f12e2947cf627587aef4b85a05ced","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-04-01 11:02:02.000000000","message":"Abandoned\n\nbfdlib in os-ken has several other issues, so forget this path","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"40086807c10216f704c8205e2df76c4ea35b08be","revisions":{"e1ee297d9a55a8e21b34ad229944e0766f2b783c":{"kind":"REWORK","_number":1,"created":"2021-02-26 09:43:34.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/32/777732/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-ken","ref":"refs/changes/32/777732/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-ken refs/changes/32/777732/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-ken refs/changes/32/777732/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-ken refs/changes/32/777732/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-ken refs/changes/32/777732/1"}}},"commit":{"parents":[{"commit":"28b0341e481a8d53e8b6b8a70b8a16a08d027e10","subject":"Fix lower-constraints job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-ken/commit/28b0341e481a8d53e8b6b8a70b8a16a08d027e10"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-02-25 15:51:10.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-02-25 17:17:39.000000000","tz":60},"subject":"Add remove_kill_bfd_session method to BFDLib","message":"Add remove_kill_bfd_session method to BFDLib\n\nadd_bfd_session starts a GreenThread but there\u0027s no option to stop\nthe created bfd_session and remove it from session dict which keeps\ntrack of the created sessions.\nThis patch adds a remove_kill_bfd_session method to kill the created\nthread and remove it from session dict.\n\nRelated-Bug: #1907089\nChange-Id: Ic8ab137b724c518f9ea79c33045e6d6c19fad984\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-ken/commit/e1ee297d9a55a8e21b34ad229944e0766f2b783c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-ken/commit/e1ee297d9a55a8e21b34ad229944e0766f2b783c"}]},"branch":"refs/heads/master"},"d668f4ad8b07b9e2280eaf9a53ebf23617c89bb7":{"kind":"REWORK","_number":2,"created":"2021-03-16 10:38:07.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/32/777732/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-ken","ref":"refs/changes/32/777732/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-ken refs/changes/32/777732/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-ken refs/changes/32/777732/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-ken refs/changes/32/777732/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-ken refs/changes/32/777732/2"}}},"commit":{"parents":[{"commit":"28b0341e481a8d53e8b6b8a70b8a16a08d027e10","subject":"Fix lower-constraints job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-ken/commit/28b0341e481a8d53e8b6b8a70b8a16a08d027e10"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-02-25 15:51:10.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-03-16 10:37:32.000000000","tz":60},"subject":"Add remove_kill_bfd_session method to BFDLib","message":"Add remove_kill_bfd_session method to BFDLib\n\nadd_bfd_session starts a GreenThread but there\u0027s no option to stop\nthe created bfd_session and remove it from session dict which keeps\ntrack of the created sessions.\nThis patch adds a remove_kill_bfd_session method to kill the created\nthread and remove it from session dict.\n\nRelated-Bug: #1907089\nChange-Id: Ic8ab137b724c518f9ea79c33045e6d6c19fad984\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-ken/commit/d668f4ad8b07b9e2280eaf9a53ebf23617c89bb7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-ken/commit/d668f4ad8b07b9e2280eaf9a53ebf23617c89bb7"}]},"branch":"refs/heads/master"},"40086807c10216f704c8205e2df76c4ea35b08be":{"kind":"REWORK","_number":3,"created":"2021-03-18 08:20:54.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/32/777732/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-ken","ref":"refs/changes/32/777732/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-ken refs/changes/32/777732/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-ken refs/changes/32/777732/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-ken refs/changes/32/777732/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-ken refs/changes/32/777732/3"}}},"commit":{"parents":[{"commit":"28b0341e481a8d53e8b6b8a70b8a16a08d027e10","subject":"Fix lower-constraints job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-ken/commit/28b0341e481a8d53e8b6b8a70b8a16a08d027e10"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-02-25 15:51:10.000000000","tz":60},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2021-03-18 08:20:48.000000000","tz":0},"subject":"Add remove_kill_bfd_session method to BFDLib","message":"Add remove_kill_bfd_session method to BFDLib\n\nadd_bfd_session starts a GreenThread but there\u0027s no option to stop\nthe created bfd_session and remove it from session dict which keeps\ntrack of the created sessions.\nThis patch adds a remove_kill_bfd_session method to kill the created\nthread and remove it from session dict.\n\nRelated-Bug: #1907089\nChange-Id: Ic8ab137b724c518f9ea79c33045e6d6c19fad984\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-ken/commit/40086807c10216f704c8205e2df76c4ea35b08be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-ken/commit/40086807c10216f704c8205e2df76c4ea35b08be"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
