)]}'
{"id":"openstack%2Fos-ken~815784","triplet_id":"openstack%2Fos-ken~master~I0e5902c4b13ab11f45f44bb9d9d9540d9630c9bb","project":"openstack/os-ken","branch":"master","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2021-11-02 14:33:30.000000000","reason":"Rodolfo Alonso replied on the change"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2021-11-02 17:50:16.000000000","reason":"Change was submitted"},"33273":{"account":{"_account_id":33273,"name":"Manu B","display_name":"Manu B","email":"manu.b@est.tech","username":"eceghkl"},"last_update":"2021-11-02 17:50:16.000000000","reason":"Change was submitted"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2021-11-02 16:58:57.000000000","reason":"Lajos Katona replied on the change"}},"hashtags":[],"change_id":"I0e5902c4b13ab11f45f44bb9d9d9540d9630c9bb","subject":"Msgpack version upgrade to 1.0.0","status":"MERGED","created":"2021-10-28 08:02:08.000000000","updated":"2021-11-02 17:55:36.000000000","submitted":"2021-11-02 17:50:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"815784","meta_rev_id":"80aa5e587a0b3488fcc3c7893981aad3d829a045","_number":815784,"virtual_id_number":815784,"owner":{"_account_id":33273,"name":"Manu B","display_name":"Manu B","email":"manu.b@est.tech","username":"eceghkl"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-02 17:50:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2021-11-02 14:33:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2021-11-02 16:58:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2021-11-02 16:58:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-28 08:59:57.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-10-28 08:59:57.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2021-10-28 08:59:57.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-10-28 09:05:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-28 11:41:17.000000000","updated_by":{"_account_id":33273,"name":"Manu B","display_name":"Manu B","email":"manu.b@est.tech","username":"eceghkl"},"reviewer":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"state":"CC"}],"messages":[{"id":"ec9bda07df22ba7295cdeb199106d4fa2e90d499","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33273,"name":"Manu B","display_name":"Manu B","email":"manu.b@est.tech","username":"eceghkl"},"date":"2021-10-28 08:02:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d1186060a163950838063c2529aff74ddf721e5e","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":33273,"name":"Manu B","display_name":"Manu B","email":"manu.b@est.tech","username":"eceghkl"},"date":"2021-10-28 08:02:17.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"9de6951d9c77146da282f8711a6dd200405cf12a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-28 09:05:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7467e7a3ba0b4467b960b67d6d901ed9 : SUCCESS in 5m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3fca851d4ecd4739804bdbd455d981fa : SUCCESS in 3m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7902b66b85a642669a8f5e226009d087 : SUCCESS in 5m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/85dbc52ff0e14bee819c69d30115c5c0 : SUCCESS in 5m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7e0aa5ecb120446294d2b04f765af4b0 : SUCCESS in 5m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/573c1499016449ab989c3a009323cf4f : SUCCESS in 7m 17s\n- requirements-check https://zuul.opendev.org/t/openstack/build/50cb47cea057474395d39fec568e1779 : FAILURE in 3m 30s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/505177cb780f4d019c6e32466ddf6c6a : SUCCESS in 1h 01m 34s","accounts_in_message":[],"_revision_number":1},{"id":"866a5c353b0c518147b88bd12dd205b105eb4c7e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":33273,"name":"Manu B","display_name":"Manu B","email":"manu.b@est.tech","username":"eceghkl"},"date":"2021-10-28 09:41:31.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7a38cb6df405a825cc8d300411bca6fa1a59bd8d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-28 11:01:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0ce1b11abbd34f9fb1ea52ef1658b0ee : SUCCESS in 5m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25bb2b38e2494eeba0f18ef8e3622ab5 : SUCCESS in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a170c0f8cfed496490ac250beec29bc4 : SUCCESS in 6m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6f3341a83b3f41cdb6f14bfa8ccc6068 : SUCCESS in 5m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4a4c10a3115245b6ba214439e1114a90 : SUCCESS in 6m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f43232fd5af4127a177cc41312e91dc : SUCCESS in 7m 58s\n- requirements-check https://zuul.opendev.org/t/openstack/build/68bbfa3e1a144f4f8229d02dc33e7239 : SUCCESS in 3m 30s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/63fa7f6d1b904231afa192adde57c714 : SUCCESS in 1h 13m 10s","accounts_in_message":[],"_revision_number":2},{"id":"edd7221ae9720c0d62ff3a7cdaaa131cee7e2a24","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":33273,"name":"Manu B","display_name":"Manu B","email":"manu.b@est.tech","username":"eceghkl"},"date":"2021-10-28 11:40:53.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"f556d276d6795a430fb260c50d2b37a964fb6d84","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-11-02 14:33:30.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"df2b7ac8bbe17024bcbf49a54ffa9791714d3570","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-11-02 16:58:57.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"2c2459da9f6999fbd43238f6e72531fe37ed64f9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-02 17:01:56.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"05c5a0c6bd0846c052f5ab69229cf092f8121a84","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-02 17:50:16.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/87fd2028ef0b45699aa83447a011b466 : SUCCESS in 5m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1628bb5348584a5cb8fefd6ee3df8533 : SUCCESS in 4m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4a0f3fe7fc7f4fe090af76b9a21ae040 : SUCCESS in 5m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/037e3ab6013c4f7aa7088ef45925de47 : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d86a0212a094f15aa89fdcf34172bbd : SUCCESS in 6m 37s\n- requirements-check https://zuul.opendev.org/t/openstack/build/6276c79e366f472db10066230bd8f6e2 : SUCCESS in 2m 47s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/f4d82c7c63a344c698f07718243df8bd : SUCCESS in 45m 47s","accounts_in_message":[],"_revision_number":2},{"id":"3a7207d99e8541acaef58b4a49b73ed4cf8d5e26","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-02 17:50:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"80aa5e587a0b3488fcc3c7893981aad3d829a045","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-02 17:55:36.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94cd9c675dac45bca5b63d0a9c2e30b0 : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4468605ff6d82716d421d4a8d587e861f7afdbb0","revisions":{"f5dcdc9942091dbadb159c1230e46f361bd2a534":{"kind":"REWORK","_number":1,"created":"2021-10-28 08:02:08.000000000","uploader":{"_account_id":33273,"name":"Manu B","display_name":"Manu B","email":"manu.b@est.tech","username":"eceghkl"},"ref":"refs/changes/84/815784/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-ken","ref":"refs/changes/84/815784/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-ken refs/changes/84/815784/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-ken refs/changes/84/815784/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-ken refs/changes/84/815784/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-ken refs/changes/84/815784/1"}}},"commit":{"parents":[{"commit":"dd9a42764f59b79b2252febf558405f313615a6d","subject":"Merge \"updated jsonrpc.Session call to have correct arguments for latest version of ovs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-ken/commit/dd9a42764f59b79b2252febf558405f313615a6d"}]}],"author":{"name":"Manu B","email":"manu.b@est.tech","date":"2021-10-28 07:49:38.000000000","tz":0},"committer":{"name":"Manu B","email":"manu.b@est.tech","date":"2021-10-28 08:01:36.000000000","tz":0},"subject":"Msgpack version upgrade to 1.0.0","message":"Msgpack version upgrade to 1.0.0\n\n* Encoding parameter is removed. UTF-8 is used always. [0]\n* strict_map_key is changed to False to keep the older behaviour.\n  Default is changed to True in 1.0.0\n\nSigned-off-by: Manu B \u003cmanu.b@est.tech\u003e\n\n[0]: https://github.com/msgpack/msgpack-python#major-breaking-changes-in-msgpack-10\n\nChange-Id: I0e5902c4b13ab11f45f44bb9d9d9540d9630c9bb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-ken/commit/f5dcdc9942091dbadb159c1230e46f361bd2a534"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-ken/commit/f5dcdc9942091dbadb159c1230e46f361bd2a534"}]},"branch":"refs/heads/master"},"4468605ff6d82716d421d4a8d587e861f7afdbb0":{"kind":"REWORK","_number":2,"created":"2021-10-28 09:41:31.000000000","uploader":{"_account_id":33273,"name":"Manu B","display_name":"Manu B","email":"manu.b@est.tech","username":"eceghkl"},"ref":"refs/changes/84/815784/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-ken","ref":"refs/changes/84/815784/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-ken refs/changes/84/815784/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-ken refs/changes/84/815784/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-ken refs/changes/84/815784/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-ken refs/changes/84/815784/2"}}},"commit":{"parents":[{"commit":"dd9a42764f59b79b2252febf558405f313615a6d","subject":"Merge \"updated jsonrpc.Session call to have correct arguments for latest version of ovs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-ken/commit/dd9a42764f59b79b2252febf558405f313615a6d"}]}],"author":{"name":"Manu B","email":"manu.b@est.tech","date":"2021-10-28 07:49:38.000000000","tz":0},"committer":{"name":"Manu B","email":"manu.b@est.tech","date":"2021-10-28 09:41:10.000000000","tz":0},"subject":"Msgpack version upgrade to 1.0.0","message":"Msgpack version upgrade to 1.0.0\n\n* Encoding parameter is removed. UTF-8 is used always. [0]\n* strict_map_key is changed to False to keep the older behaviour.\n  Default is changed to True in 1.0.0\n\nSigned-off-by: Manu B \u003cmanu.b@est.tech\u003e\n\n[0]: https://github.com/msgpack/msgpack-python#major-breaking-changes-in-msgpack-10\n\nChange-Id: I0e5902c4b13ab11f45f44bb9d9d9540d9630c9bb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-ken/commit/4468605ff6d82716d421d4a8d587e861f7afdbb0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-ken/commit/4468605ff6d82716d421d4a8d587e861f7afdbb0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
