)]}'
{"id":"openstack%2Fos-ken~953631","triplet_id":"openstack%2Fos-ken~master~Icf346e1aa99f526f416ff1422fa0bd4cfd1ba438","project":"openstack/os-ken","branch":"master","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2025-07-09 14:40:14.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"38146":{"account":{"_account_id":38146,"name":"Anas Jouhdy","display_name":"Amoo","email":"anas.jouhdy@gmail.com","username":"Amoo"},"last_update":"2025-07-09 16:01:36.000000000","reason":"Change was submitted"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2025-07-01 07:13:55.000000000","reason":"\u003cGERRIT_ACCOUNT_8313\u003e replied on the change","reason_account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}}},"hashtags":[],"change_id":"Icf346e1aa99f526f416ff1422fa0bd4cfd1ba438","subject":"Fixing the BGP loop prevention","status":"MERGED","created":"2025-06-27 17:12:50.000000000","updated":"2025-07-09 16:02:32.000000000","submitted":"2025-07-09 16:01:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"953631","meta_rev_id":"97e263dc83d35f31cdb6a3e3361388c9a9b4d034","_number":953631,"virtual_id_number":953631,"owner":{"_account_id":38146,"name":"Anas Jouhdy","display_name":"Amoo","email":"anas.jouhdy@gmail.com","username":"Amoo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-09 16:01:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2025-07-09 14:40:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2025-07-01 07:13:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2025-07-09 14:40:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-27 18:11:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-30 08:01:47.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2025-07-01 07:13:55.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"a5376eaca413898843a6c042feb30a9aa9241bef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38146,"name":"Anas Jouhdy","display_name":"Amoo","email":"anas.jouhdy@gmail.com","username":"Amoo"},"date":"2025-06-27 17:12:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5b77acc9f3df406bca8562cccdb5b314dd7c665b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-27 18:11:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f85697cca3974bc1afe921596fe41f02\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/216f3521f1ad4de0bbd1056cdb5c39b2 : SUCCESS in 2m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6d7d41c0a4f046058c69c1a01b37e9f9 : SUCCESS in 3m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ae92ce7a7e4d472da5233f1206def23c : SUCCESS in 3m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/14a445bab699434d866ac66192b12245 : SUCCESS in 5m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6dc411224251442ab27692a240a3f1ed : SUCCESS in 8m 00s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/7c1ca477a5c24349993b24b5c5b5c910 : SUCCESS in 43m 19s\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/85a87a22ca68411c884ba58dcc8ebb31 : SUCCESS in 54m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c1b67893a281232c54b6aa9f38559baa14bf6a70","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-06-30 08:01:47.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5aa0aea27d2319ecd9592ec8e98548b0835eb1af","author":{"_account_id":38146,"name":"Anas Jouhdy","display_name":"Amoo","email":"anas.jouhdy@gmail.com","username":"Amoo"},"date":"2025-06-30 17:43:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"61f5f4aec1f1ed4e5099ffd6b6e00d0b0dbaf8f7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38146,"name":"Anas Jouhdy","display_name":"Amoo","email":"anas.jouhdy@gmail.com","username":"Amoo"},"date":"2025-06-30 20:50:55.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"86b615df2b3766b4808b0042f4db1e4ff7bd91c2","author":{"_account_id":38146,"name":"Anas Jouhdy","display_name":"Amoo","email":"anas.jouhdy@gmail.com","username":"Amoo"},"date":"2025-06-30 21:01:17.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"97ff3952d102bcf1e15b76e80dfe982e8bfde31b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-30 22:12:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc66578970f0481bb65a54d877e82196\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69fedfe247a243c3bae0204ce71abf36 : SUCCESS in 2m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/64d101f2233c49c38dd70a0a8f66f7c2 : SUCCESS in 6m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1ec82e026d914f82bc160642a613fb6a : SUCCESS in 5m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f7fcb89cf1e3437e9cb600d384fa6c1a : SUCCESS in 10m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb4c7b8879e84799a512698019aa165e : SUCCESS in 9m 44s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/e7e68d59e1c442b38149cb7423f73214 : SUCCESS in 1h 16m 36s\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/5e0cfbc433e44c6bb73883bcadc4e16b : SUCCESS in 54m 01s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7344000cb9426016f442ed669c63f9f6d3cd59b4","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-07-01 07:13:55.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"20946898148b9ab27eb809ec790d2cd5410ec05e","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-07-09 14:40:14.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"6648498fe969df2aa79e52e24a82b5ba284e90b8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-09 14:40:35.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"015b7000679f13ba15d0cfa420f28a34dfa14c56","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-09 16:01:35.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9cf5c810fca3472ba320abf565bef385\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d26fc116ffbb41dcbdba58461164add4 : SUCCESS in 2m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9b1be1a09d2c478399274f3dbf0a89c6 : SUCCESS in 6m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6cd40753a781416287bd466fe72c270f : SUCCESS in 6m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ff7eeb1fceb41db97cdd12dc73c92be : SUCCESS in 8m 11s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/2855727d0ed44afb86faa4f8f85f31c9 : SUCCESS in 1h 16m 17s","accounts_in_message":[],"_revision_number":2},{"id":"24497e71d0beb5edd876bcff93ed6a8970c64f84","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-09 16:01:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"97e263dc83d35f31cdb6a3e3361388c9a9b4d034","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-09 16:02:32.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/62e9ed696ffa42f49e21d7e5219e6235\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1bc5f1e8f4ba4bb4a862b760a3517746 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"bf639392f3dea7909792e257b8b65234dc5210a1","revisions":{"74bc34bc775e213464925da36907ff8ed16079d6":{"kind":"REWORK","_number":1,"created":"2025-06-27 17:12:50.000000000","uploader":{"_account_id":38146,"name":"Anas Jouhdy","display_name":"Amoo","email":"anas.jouhdy@gmail.com","username":"Amoo"},"ref":"refs/changes/31/953631/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-ken","ref":"refs/changes/31/953631/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-ken refs/changes/31/953631/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-ken refs/changes/31/953631/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-ken refs/changes/31/953631/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-ken refs/changes/31/953631/1"}}},"commit":{"parents":[{"commit":"6e296f97ee2845c968edd9a56be80f274b5c367c","subject":"Get rid of deprecated datetime.utcfromtimestamp","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-ken/commit/6e296f97ee2845c968edd9a56be80f274b5c367c"}]}],"author":{"name":"Amoo","email":"anas.jouhdy@gmail.com","date":"2025-06-27 17:11:09.000000000","tz":120},"committer":{"name":"Amoo","email":"anas.jouhdy@gmail.com","date":"2025-06-27 17:11:09.000000000","tz":120},"subject":"Fixing the BGP loop prevention","message":"Fixing the BGP loop prevention\n\nThe method _is_looped_path_attrs did not return True when a BGP loop was detected.\nThis had the effect of not rejecting the new route learn by the peer even though the local AS was in the AS Path.\n\nCloses-Bug: #2115510\nChange-Id: Icf346e1aa99f526f416ff1422fa0bd4cfd1ba438\nSigned-off-by: Amo \u003canas.jouhdy@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-ken/commit/74bc34bc775e213464925da36907ff8ed16079d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-ken/commit/74bc34bc775e213464925da36907ff8ed16079d6"}]},"branch":"refs/heads/master"},"bf639392f3dea7909792e257b8b65234dc5210a1":{"kind":"REWORK","_number":2,"created":"2025-06-30 20:50:55.000000000","uploader":{"_account_id":38146,"name":"Anas Jouhdy","display_name":"Amoo","email":"anas.jouhdy@gmail.com","username":"Amoo"},"ref":"refs/changes/31/953631/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-ken","ref":"refs/changes/31/953631/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-ken refs/changes/31/953631/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-ken refs/changes/31/953631/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-ken refs/changes/31/953631/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-ken refs/changes/31/953631/2"}}},"commit":{"parents":[{"commit":"6e296f97ee2845c968edd9a56be80f274b5c367c","subject":"Get rid of deprecated datetime.utcfromtimestamp","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-ken/commit/6e296f97ee2845c968edd9a56be80f274b5c367c"}]}],"author":{"name":"Amoo","email":"anas.jouhdy@gmail.com","date":"2025-06-27 17:11:09.000000000","tz":120},"committer":{"name":"Amoo","email":"anas.jouhdy@gmail.com","date":"2025-06-30 20:48:42.000000000","tz":120},"subject":"Fixing the BGP loop prevention","message":"Fixing the BGP loop prevention\n\nThe method _is_looped_path_attrs did not return True when a BGP loop was detected.\nThis had the effect of not rejecting the new route learn by the peer even though the local AS was in the AS Path.\n\nCloses-Bug: #2115510\nChange-Id: Icf346e1aa99f526f416ff1422fa0bd4cfd1ba438\nSigned-off-by: Amo \u003canas.jouhdy@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-ken/commit/bf639392f3dea7909792e257b8b65234dc5210a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-ken/commit/bf639392f3dea7909792e257b8b65234dc5210a1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
