)]}'
{"id":"openstack%2Fos-ken~970277","triplet_id":"openstack%2Fos-ken~master~Ie5252ec06a5a0e4fce2131af56046991ee4fbab3","project":"openstack/os-ken","branch":"master","topic":"bug/2126791","attention_set":{},"removed_from_attention_set":{"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2026-08-10 08:38:50.000000000","reason":"Change was abandoned"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2026-08-10 08:38:50.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ie5252ec06a5a0e4fce2131af56046991ee4fbab3","subject":"[eventlet]: Add events for threads to make stopping smoother","status":"ABANDONED","created":"2025-12-09 16:31:03.000000000","updated":"2026-08-10 08:38:50.000000000","total_comment_count":6,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"038ac0cbf67b4fc2e72a306a98d6d8edb2fca8da","_number":970277,"virtual_id_number":970277,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-05-21 19:54:56.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}],"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-09 17:18:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-12-09 17:22:58.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2026-01-08 06:56:24.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2026-08-10 08:38:50.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"}],"messages":[{"id":"df70dd1ab858e46d2ffcd01698bad0aae021c0a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-12-09 16:31:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b3f9b80eb1bcc15135934bb09f96b497dac1258e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-09 17:18:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a5abde9c2cf443e091528b8f69e7714a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2717ace469be49a487dd11d43c00d642 : SUCCESS in 2m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/643c2425630e4057be50f9d21b1b8c97 : SUCCESS in 6m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/baaf334e37b246d496986fbabc97ae88 : SUCCESS in 9m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ccc595ae3ed946288d512061300a6031 : SUCCESS in 4m 10s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/976a11221eb94a4b915076971ded5ca0 : SUCCESS in 44m 30s\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/de5a1a59e7a74c8fa037de94135ab9ca : FAILURE in 41m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e1f06b5e46c48297a4e1493262a25afc6d570d38","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-01-07 08:56:03.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"7475f617ad657c6db180c65c782abda8bdfb3f80","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-07 10:21:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/924712f85e7f41caba5a64928e1b25db\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ff03b7ea2a944b8b38b586d74a2d452 : SUCCESS in 2m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/01ca35dd1acd48d7b3d5477dfe4811ef : SUCCESS in 6m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/276d2b2328404812ba20251fbe569a09 : SUCCESS in 8m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02e7b4121f9f456297c9d4d49b60a61a : SUCCESS in 4m 04s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/229c435eb903407b94e5c4aaf03f30eb : SUCCESS in 1h 18m 42s\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/f42649a57229488b8d8a14c49266d124 : FAILURE in 26m 48s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0038e245866f59442bc1c93fc734781a037baa4d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-01-07 19:02:08.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"59a146b11721b0e765955f22aeba6a0f32a6577a","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-01-07 19:14:18.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"5cdd895d24c8edd94159d77df8e6d8478641163d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-07 20:34:47.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0be44d0657074c67ac3a5112fb7211fd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cfb69f93a0ae4a06ac9f223d957f9b03 : SUCCESS in 3m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f82398ebe8ee469bb9e7fc3d0f17b2d7 : SUCCESS in 8m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6163dc8f5dd543f9a7fab540fba73d81 : SUCCESS in 7m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e66e66facc1e4b1bab7fa42de660e153 : SUCCESS in 6m 57s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/4935d755095e4cd1b844df010281a0e9 : SUCCESS in 1h 20m 57s\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/20ae4ffcffa0442b906108ed31e587ba : FAILURE in 31m 52s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"3303e8001771c5fcc5b65315c19f7d0b8d2f03db","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-01-27 23:38:13.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ef0d2e071f9465ce578f1805ed4664c9e3671378","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 01:00:36.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e5251b0f070f4361a421a3812a0f184c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/215bad635b7c4bf98d821f956038a53f : SUCCESS in 3m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dc240949ae934c8da200fdad9ee6c455 : SUCCESS in 6m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/508078b3a38b42b48e49954d57fca303 : SUCCESS in 3m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b14ae330314a4e54bcc4483fa135701f : SUCCESS in 8m 40s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/d99a12d2157e449fa0536d05a717fc26 : SUCCESS in 1h 16m 47s\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/b6660df733914e82ad48c3e4924954e4 : SUCCESS in 26m 42s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"142b19bb18f97bf77125316010f52f512fcc7407","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-01-28 01:11:05.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"11e0ceee37cb52da72d10c84eac714ed59a04077","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-01-29 11:50:12.000000000","message":"Patch Set 3: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"071c541b1caf7bffb7aa47e7a400052fb471eb1d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-01-30 15:58:52.000000000","message":"Uploaded patch set 4: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":4},{"id":"6e7704ea5f7dac4bb288031739f015382e388513","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-01-30 16:02:57.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"95ad53251c1b6163348f2e1417c9a3b9f88e1343","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-01-30 16:06:36.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"ed2932b3a5043cd29a87ab5bedda23fad5d5a66a","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-01-30 16:07:29.000000000","message":"Patch Set 5: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"927b7911551e527be0e4e2f33643557360f0ec21","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-30 17:20:09.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f2d1388ac29b4304bd7e736b1c60b770\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/389192a51b00444591e71d3a561233cc : SUCCESS in 3m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a574e78e0a9548878b38fa349f5e0872 : SUCCESS in 4m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fe075cb0408f4d948c7ccadccb79b70c : SUCCESS in 2m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/780783f3106a4dca826affe695bb715b : SUCCESS in 8m 08s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/cc986213abc5407a9818ce5e63c09cf0 : SUCCESS in 1h 10m 46s\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/bc70142873cc43d9a755f22b59263c4f : SUCCESS in 48m 08s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"aba2b65e08163a176c1dd68fc959a3bd59c4aa9c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-05-21 16:54:56.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n* Workflow-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"020ee4e0ea33d179a617275c370e374b69a2657a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-21 19:54:56.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/052e54c8a2b1439089212cdc6bbf4b2c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/724950c91dc14364b97e03040c37d057 : SUCCESS in 2m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/55d0ca5146704658bf77edf25592db75 : SUCCESS in 7m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/73586d0946b649878b9d591f8b4d189a : SUCCESS in 5m 27s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/cbc8b7242fe0491cbc685f9c64a41dd1 : SUCCESS in 4m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/460957bf026e4b7da2e840a2d5d61ded : SUCCESS in 8m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/79431bfaf519450c8fa14dd196649815 : SUCCESS in 4m 54s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/50b072dfb0024a3f8f0030690b13640a : FAILURE in 2h 58m 07s\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/f234b231517949bb817e0135690a0fba : SUCCESS in 26m 16s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"038ac0cbf67b4fc2e72a306a98d6d8edb2fca8da","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-08-10 08:38:50.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"de9a2596de5a77151fda4112253d4a95c22daf9b","revisions":{"7fb0ef2e036bb2a30ecb4b86ab902ebf90ce4c26":{"kind":"REWORK","_number":1,"created":"2025-12-09 16:31:03.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/77/970277/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-ken","ref":"refs/changes/77/970277/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-ken refs/changes/77/970277/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-ken refs/changes/77/970277/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-ken refs/changes/77/970277/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-ken refs/changes/77/970277/1"}}},"commit":{"parents":[{"commit":"c89b6b76322f5b2259b6da234f28efc91ab24d2b","subject":"Merge \"hub: use native as default implementation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-ken/commit/c89b6b76322f5b2259b6da234f28efc91ab24d2b"}]}],"author":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2025-12-09 12:19:05.000000000","tz":60},"committer":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2025-12-09 16:28:55.000000000","tz":60},"subject":"[eventlet]: change events to real threading events","message":"[eventlet]: change events to real threading events\n\nTODO: perhaps split the events to threaded and eventlet\nevent classes...\n\nRelated-Bug: #2126791\nChange-Id: Ie5252ec06a5a0e4fce2131af56046991ee4fbab3\nSigned-off-by: lajoskatona \u003clajos.katona@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-ken/commit/7fb0ef2e036bb2a30ecb4b86ab902ebf90ce4c26"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-ken/commit/7fb0ef2e036bb2a30ecb4b86ab902ebf90ce4c26"}]},"branch":"refs/heads/master"},"ff6ac8366d2224e4515ff25743461d9a32c3cc9d":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-01-07 08:56:03.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/77/970277/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-ken","ref":"refs/changes/77/970277/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-ken refs/changes/77/970277/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-ken refs/changes/77/970277/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-ken refs/changes/77/970277/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-ken refs/changes/77/970277/2"}}},"commit":{"parents":[{"commit":"ad36a7682c9f7c8dd549f0852e717c93c52ad849","subject":"Accept empty \"OXM\" fields (tuple, list)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-ken/commit/ad36a7682c9f7c8dd549f0852e717c93c52ad849"}]}],"author":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2025-12-09 12:19:05.000000000","tz":60},"committer":{"name":"Lajos Katona","email":"lajos.katona@est.tech","date":"2026-01-07 08:56:03.000000000","tz":0},"subject":"[eventlet]: change events to real threading events","message":"[eventlet]: change events to real threading events\n\nTODO: perhaps split the events to threaded and eventlet\nevent classes...\n\nRelated-Bug: #2126791\nChange-Id: Ie5252ec06a5a0e4fce2131af56046991ee4fbab3\nSigned-off-by: lajoskatona \u003clajos.katona@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-ken/commit/ff6ac8366d2224e4515ff25743461d9a32c3cc9d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-ken/commit/ff6ac8366d2224e4515ff25743461d9a32c3cc9d"}]},"branch":"refs/heads/master","description":"Rebase"},"a9f79cc5f155578a18a040990d0939b257bc199a":{"kind":"REWORK","_number":3,"created":"2026-01-07 19:02:08.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/77/970277/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-ken","ref":"refs/changes/77/970277/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-ken refs/changes/77/970277/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-ken refs/changes/77/970277/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-ken refs/changes/77/970277/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-ken refs/changes/77/970277/3"}}},"commit":{"parents":[{"commit":"ad36a7682c9f7c8dd549f0852e717c93c52ad849","subject":"Accept empty \"OXM\" fields (tuple, list)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-ken/commit/ad36a7682c9f7c8dd549f0852e717c93c52ad849"}]}],"author":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2025-12-09 12:19:05.000000000","tz":60},"committer":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2026-01-07 19:01:07.000000000","tz":60},"subject":"[eventlet]: change events to real threading events","message":"[eventlet]: change events to real threading events\n\nTODO: perhaps split the events to threaded and eventlet\nevent classes...\n\nRelated-Bug: #2126791\nChange-Id: Ie5252ec06a5a0e4fce2131af56046991ee4fbab3\nSigned-off-by: lajoskatona \u003clajos.katona@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-ken/commit/a9f79cc5f155578a18a040990d0939b257bc199a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-ken/commit/a9f79cc5f155578a18a040990d0939b257bc199a"}]},"branch":"refs/heads/master"},"66e0102aa8f95763ae99e228ff0bf881b24798f6":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2026-01-30 15:58:52.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/77/970277/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-ken","ref":"refs/changes/77/970277/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-ken refs/changes/77/970277/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-ken refs/changes/77/970277/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-ken refs/changes/77/970277/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-ken refs/changes/77/970277/4"}}},"commit":{"parents":[{"commit":"ad36a7682c9f7c8dd549f0852e717c93c52ad849","subject":"Accept empty \"OXM\" fields (tuple, list)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-ken/commit/ad36a7682c9f7c8dd549f0852e717c93c52ad849"}]}],"author":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2025-12-09 12:19:05.000000000","tz":60},"committer":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2026-01-30 10:49:14.000000000","tz":60},"subject":"[eventlet]: Add events for threads to make stopping smoother","message":"[eventlet]: Add events for threads to make stopping smoother\n\nchange events (siblings of EventBase) to real threading events\n\nRelated-Bug: #2126791\nChange-Id: Ie5252ec06a5a0e4fce2131af56046991ee4fbab3\nSigned-off-by: lajoskatona \u003clajos.katona@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-ken/commit/66e0102aa8f95763ae99e228ff0bf881b24798f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-ken/commit/66e0102aa8f95763ae99e228ff0bf881b24798f6"}]},"branch":"refs/heads/master"},"000d5b8c59f88a5413d087588711d2a9f9e42ba4":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2026-01-30 16:06:36.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/77/970277/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-ken","ref":"refs/changes/77/970277/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-ken refs/changes/77/970277/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-ken refs/changes/77/970277/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-ken refs/changes/77/970277/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-ken refs/changes/77/970277/5"}}},"commit":{"parents":[{"commit":"ad36a7682c9f7c8dd549f0852e717c93c52ad849","subject":"Accept empty \"OXM\" fields (tuple, list)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-ken/commit/ad36a7682c9f7c8dd549f0852e717c93c52ad849"}]}],"author":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2025-12-09 12:19:05.000000000","tz":60},"committer":{"name":"Lajos Katona","email":"lajos.katona@est.tech","date":"2026-01-30 16:06:36.000000000","tz":0},"subject":"[eventlet]: Add events for threads to make stopping smoother","message":"[eventlet]: Add events for threads to make stopping smoother\n\nchange events (siblings of EventBase) to real threading events\n\nThe bug was opened to handle the case that ovs-agent is not stopping\nwhen for example the user executes systemctl stop *q-agt*.\nThis patch and the one in Neutron (see [1]) handles the issues with\nos-ken threads hanging.\nFor the whole bug the hanging RPV thread(s) must be handled in a\nseparate patch.\n\n[1]: https://review.opendev.org/c/openstack/neutron/+/967105\nRelated-Bug: #2126791\nChange-Id: Ie5252ec06a5a0e4fce2131af56046991ee4fbab3\nSigned-off-by: lajoskatona \u003clajos.katona@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-ken/commit/000d5b8c59f88a5413d087588711d2a9f9e42ba4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-ken/commit/000d5b8c59f88a5413d087588711d2a9f9e42ba4"}]},"branch":"refs/heads/master","description":"Edit commit message"},"de9a2596de5a77151fda4112253d4a95c22daf9b":{"kind":"REWORK","_number":6,"created":"2026-05-21 16:54:56.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/77/970277/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-ken","ref":"refs/changes/77/970277/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-ken refs/changes/77/970277/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-ken refs/changes/77/970277/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-ken refs/changes/77/970277/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-ken refs/changes/77/970277/6"}}},"commit":{"parents":[{"commit":"c633a9c4c61970ade0a55769798a21fa41ab680b","subject":"Use threading.Timer instead of time.sleep","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-ken/commit/c633a9c4c61970ade0a55769798a21fa41ab680b"}]}],"author":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2025-12-09 12:19:05.000000000","tz":60},"committer":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2026-05-21 16:05:21.000000000","tz":120},"subject":"[eventlet]: Add events for threads to make stopping smoother","message":"[eventlet]: Add events for threads to make stopping smoother\n\nchange events (siblings of EventBase) to real threading events\n\nThe bug was opened to handle the case that ovs-agent is not stopping\nwhen for example the user executes systemctl stop *q-agt*.\nThis patch and the one in Neutron (see [1]) handles the issues with\nos-ken threads hanging.\nFor the whole bug the hanging RPV thread(s) must be handled in a\nseparate patch.\n\n[1]: https://review.opendev.org/c/openstack/neutron/+/967105\nRelated-Bug: #2126791\nChange-Id: Ie5252ec06a5a0e4fce2131af56046991ee4fbab3\nSigned-off-by: lajoskatona \u003clajos.katona@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-ken/commit/de9a2596de5a77151fda4112253d4a95c22daf9b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-ken/commit/de9a2596de5a77151fda4112253d4a95c22daf9b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
