)]}'
{"id":"openstack%2Fos-loganalyze~657652","triplet_id":"openstack%2Fos-loganalyze~master~If5dc57efcfc8768612c4f695c472f3ae2a085700","project":"openstack/os-loganalyze","branch":"master","topic":"watcher-devstack-logging","hashtags":[],"change_id":"If5dc57efcfc8768612c4f695c472f3ae2a085700","subject":"Add watcher to filter list","status":"ABANDONED","created":"2019-05-07 17:26:25.000000000","updated":"2019-11-21 16:51:32.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"fba4f5a37131329111c9fd2049163fd3b8aa64f8","_number":657652,"virtual_id_number":657652,"owner":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-05-07 19:57:26.000000000","_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":1,"date":"2019-05-07 17:40:42.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2019-05-20 23:51:46.000000000","_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-07 17:40:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-20 23:51:46.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"}],"messages":[{"id":"34089df6447675b7bcfc8f4cb5269df627fb031a","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-05-07 17:26:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"78f1c42cfffcb95925a104f828037b3e88ca0876","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-07 17:27:10.000000000","message":"Patch Set 1: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"c51963ad068b23fbb2785a17f0bcbf358f743954","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-05-07 17:27:42.000000000","message":"Patch Set 1:\n\nrecheck dep update","accounts_in_message":[],"_revision_number":1},{"id":"491d2f07c71702d19fc48de6300d96a01a6dfe11","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-07 17:40:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/52/657652/1/check/openstack-tox-pep8/be191fa/ : SUCCESS in 3m 41s\n- openstack-tox-py27 http://logs.openstack.org/52/657652/1/check/openstack-tox-py27/429e580/ : SUCCESS in 3m 51s","accounts_in_message":[],"_revision_number":1},{"id":"4176ec64fb6cfa1f2311e9bd628f8df26cd8ae86","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-05-07 19:57:26.000000000","message":"Patch Set 1:\n\nThis doesn\u0027t seem to be working:\n\nhttps://review.opendev.org/#/c/657661/\n\nI don\u0027t see the severity display levels:\n\nhttp://logs.openstack.org/61/657661/1/check/watcher-tempest-basic_optim/235dcd7/controller/logs/screen-watcher-decision-engine.txt.gz","accounts_in_message":[],"_revision_number":1},{"id":"5526f094f7f4a9d51409f27b85ae14cfc96d7b83","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-05-20 22:57:01.000000000","message":"Patch Set 1:\n\n\u003e This doesn\u0027t seem to be working:\n \u003e \n \u003e https://review.opendev.org/#/c/657661/\n \u003e \n \u003e I don\u0027t see the severity display levels:\n \u003e \n \u003e http://logs.openstack.org/61/657661/1/check/watcher-tempest-basic_optim/235dcd7/controller/logs/screen-watcher-decision-engine.txt.gz\n\nPretty sure you need to replace setup_logging with setup_systemd_logging in devstack for it to work.\nhttps://github.com/openstack/senlin/commit/f06ec494b42667941fecb27e43fb279940a44dc5#diff-a7accb299ab3c3b5058dc9165dc0ee3d","accounts_in_message":[],"_revision_number":1},{"id":"5109774c536b0f00cdc2461e488f65d57dcb8509","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2019-05-20 23:51:46.000000000","message":"Patch Set 1:\n\n\u003e \u003e This doesn\u0027t seem to be working:\n \u003e \u003e\n \u003e \u003e https://review.opendev.org/#/c/657661/\n \u003e \u003e\n \u003e \u003e I don\u0027t see the severity display levels:\n \u003e \u003e\n \u003e \u003e http://logs.openstack.org/61/657661/1/check/watcher-tempest-basic_optim/235dcd7/controller/logs/screen-watcher-decision-engine.txt.gz\n \u003e \n \u003e Pretty sure you need to replace setup_logging with\n \u003e setup_systemd_logging in devstack for it to work.\n \u003e https://github.com/openstack/senlin/commit/f06ec494b42667941fecb27e43fb279940a44dc5#diff-a7accb299ab3c3b5058dc9165dc0ee3d\n\nnvm that wasn\u0027t it","accounts_in_message":[],"_revision_number":1},{"id":"c7a635a53627118cea29a255513451b966efaceb","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-11-21 16:51:32.000000000","message":"Abandoned\n\nDoesn\u0027t seem to be working and no one seems to care much so I\u0027m going to drop this so it\u0027s not in my dashboard anymore.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9cf59c9a4246321ae8e781fedcb53e83feb61556","revisions":{"9cf59c9a4246321ae8e781fedcb53e83feb61556":{"kind":"REWORK","_number":1,"created":"2019-05-07 17:26:25.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/52/657652/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-loganalyze","ref":"refs/changes/52/657652/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-loganalyze refs/changes/52/657652/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-loganalyze refs/changes/52/657652/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-loganalyze refs/changes/52/657652/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-loganalyze refs/changes/52/657652/1"}}},"commit":{"parents":[{"commit":"d80e6a55e050fdfb71ea439810ac69853acaec9f","subject":"Merge \"Add support for Senlin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-loganalyze/commit/d80e6a55e050fdfb71ea439810ac69853acaec9f"}]}],"author":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-05-07 17:25:58.000000000","tz":-240},"committer":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-05-07 17:25:58.000000000","tz":-240},"subject":"Add watcher to filter list","message":"Add watcher to filter list\n\nGive Watcher logs that sweet log analysis.\n\nDepends-On: https://review.opendev.org/657651\nChange-Id: If5dc57efcfc8768612c4f695c472f3ae2a085700\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-loganalyze/commit/9cf59c9a4246321ae8e781fedcb53e83feb61556"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-loganalyze/commit/9cf59c9a4246321ae8e781fedcb53e83feb61556"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
