)]}'
{"id":"openstack%2Fos-loganalyze~734600","triplet_id":"openstack%2Fos-loganalyze~master~Ibaa6d6dcc39734610fa21e953d790217dee4c979","project":"openstack/os-loganalyze","branch":"master","topic":"cleanup","hashtags":[],"change_id":"Ibaa6d6dcc39734610fa21e953d790217dee4c979","subject":"Cleanup repo a bit","status":"ABANDONED","created":"2020-06-09 12:52:39.000000000","updated":"2020-06-09 13:58:54.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"fe8e333c3bb2e8135f3e5004cf8fca7a7d44bcc0","_number":734600,"virtual_id_number":734600,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2020-06-09 13:32:02.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-09 13:32:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"91efab82a51cafabf238ef2e19ba052f78e0e207","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-09 12:52:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c2303a10b7b95c77ab7577f95524aeaa5f5a77af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 12:54:20.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/os-loganalyze on branch master.  The error was:\n\n  expected str for dictionary value @ data[\u0027check\u0027][\u0027jobs\u0027][2][\u0027openstack-tox-py36\u0027]\n\nThe error appears in the following project stanza:\n\n  project:\n      check:\n        jobs:\n          - openstack-tox-pep8\n          - openstack-tox-py27\n          - openstack-tox-py36:\n            voting: false\n      gate:\n        jobs:\n          - openstack-tox-pep8\n          - openstack-tox-py27\n      experimental:\n        jobs:\n          - legacy-dsvm-os-loganalyze\n\n  in \"openstack/os-loganalyze/.zuul.yaml@master\", line 2, column 3","accounts_in_message":[],"_revision_number":1},{"id":"727af7bb18b2529651a0dcb13dea1d91ad243217","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-09 13:04:28.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"98023dc42dc10c332caba12a6d37eabd63589d6d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 13:14:19.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0cf49803505a422b9e9d4928d1ae8bf0 : FAILURE in 5m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/95aff9cf0d894392830913808da507c0 : SUCCESS in 6m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d5f04a9f5f8c415da73284193a4a739d : FAILURE in 6m 43s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0740e7b83ac54e7d684c7f786792c4f3967103f5","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-09 13:19:25.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"00eda658e504f67a5becbbba87bca0dc1cdeba9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 13:32:02.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27f436b9a46c42dabedd1cd55290f704 : SUCCESS in 4m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ef5ac305999d47a1b80434b47fc22cda : SUCCESS in 5m 29s","accounts_in_message":[],"_revision_number":3},{"id":"530e18250a0dd4a43eab6dd7687585ef9fc60e36","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-09 13:58:54.000000000","message":"Abandoned\n\nrepo is not used anymore","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"fe239b16d6ab92390bbf7814dfc6402c35191658","revisions":{"4f3489a0b87a6c218263f1ccad908b08f80a872c":{"kind":"REWORK","_number":1,"created":"2020-06-09 12:52:39.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/00/734600/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-loganalyze","ref":"refs/changes/00/734600/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-loganalyze refs/changes/00/734600/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-loganalyze refs/changes/00/734600/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-loganalyze refs/changes/00/734600/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-loganalyze refs/changes/00/734600/1"}}},"commit":{"parents":[{"commit":"d80e6a55e050fdfb71ea439810ac69853acaec9f","subject":"Merge \"Add support for Senlin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-loganalyze/commit/d80e6a55e050fdfb71ea439810ac69853acaec9f"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-06-09 12:21:48.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-06-09 12:52:21.000000000","tz":120},"subject":"Cleanup repo a bit","message":"Cleanup repo a bit\n\nA couple of cleanups:\n* Add py36 job - non-voting for now since it fails.\n* Cleanup setup.cfg\n* Remove openstack-common.conf and babel.cfg, they are not used.\n* Remove unused Babel dependency, update requirements\n* Update to hacking 3.1.0 and fix problems found\n* Start adding six so that this repo works with both\n  py2 and py3\n\nChange-Id: Ibaa6d6dcc39734610fa21e953d790217dee4c979\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-loganalyze/commit/4f3489a0b87a6c218263f1ccad908b08f80a872c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-loganalyze/commit/4f3489a0b87a6c218263f1ccad908b08f80a872c"}]},"branch":"refs/heads/master"},"e8454dfa364df89eb28e57734ad285b1347f9bb3":{"kind":"REWORK","_number":2,"created":"2020-06-09 13:04:28.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/00/734600/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-loganalyze","ref":"refs/changes/00/734600/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-loganalyze refs/changes/00/734600/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-loganalyze refs/changes/00/734600/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-loganalyze refs/changes/00/734600/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-loganalyze refs/changes/00/734600/2"}}},"commit":{"parents":[{"commit":"d80e6a55e050fdfb71ea439810ac69853acaec9f","subject":"Merge \"Add support for Senlin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-loganalyze/commit/d80e6a55e050fdfb71ea439810ac69853acaec9f"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-06-09 12:21:48.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-06-09 13:04:23.000000000","tz":120},"subject":"Cleanup repo a bit","message":"Cleanup repo a bit\n\nA couple of cleanups:\n* Add py36 job - non-voting for now since it fails.\n* Cleanup setup.cfg\n* Remove openstack-common.conf and babel.cfg, they are not used.\n* Remove unused Babel dependency, update requirements\n* Update to hacking 3.1.0 and fix problems found\n* Start adding six so that this repo works with both\n  py2 and py3\n\nChange-Id: Ibaa6d6dcc39734610fa21e953d790217dee4c979\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-loganalyze/commit/e8454dfa364df89eb28e57734ad285b1347f9bb3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-loganalyze/commit/e8454dfa364df89eb28e57734ad285b1347f9bb3"}]},"branch":"refs/heads/master"},"fe239b16d6ab92390bbf7814dfc6402c35191658":{"kind":"REWORK","_number":3,"created":"2020-06-09 13:19:25.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/00/734600/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-loganalyze","ref":"refs/changes/00/734600/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-loganalyze refs/changes/00/734600/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-loganalyze refs/changes/00/734600/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-loganalyze refs/changes/00/734600/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-loganalyze refs/changes/00/734600/3"}}},"commit":{"parents":[{"commit":"d80e6a55e050fdfb71ea439810ac69853acaec9f","subject":"Merge \"Add support for Senlin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-loganalyze/commit/d80e6a55e050fdfb71ea439810ac69853acaec9f"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-06-09 12:21:48.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-06-09 13:19:18.000000000","tz":120},"subject":"Cleanup repo a bit","message":"Cleanup repo a bit\n\nA couple of cleanups:\n* Cleanup setup.cfg\n* Remove openstack-common.conf and babel.cfg, they are not used.\n* Remove unused Babel dependency, update requirements\n* Update to hacking 2.0.0 and fix problems found\n\nChange-Id: Ibaa6d6dcc39734610fa21e953d790217dee4c979\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-loganalyze/commit/fe239b16d6ab92390bbf7814dfc6402c35191658"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-loganalyze/commit/fe239b16d6ab92390bbf7814dfc6402c35191658"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
