)]}'
{"id":"openstack%2Fos-net-config~836739","triplet_id":"openstack%2Fos-net-config~stable%2Fussuri~I3fc3df3fc8282705253911c3d909a8c51447ad31","project":"openstack/os-net-config","branch":"stable/ussuri","topic":"tripleo_ussuri_teardown","hashtags":[],"change_id":"I3fc3df3fc8282705253911c3d909a8c51447ad31","subject":"Ussuri tear down as EOL","status":"ABANDONED","created":"2022-04-06 07:58:26.000000000","updated":"2022-04-12 14:49:37.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"78232af78acd9558bad4d51ba3cf026cf44bdda7","_number":836739,"virtual_id_number":836739,"owner":{"_account_id":31075,"name":"Pooja Jadhav","email":"pojadhav@redhat.com","username":"pojadhav"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":31075,"name":"Pooja Jadhav","email":"pojadhav@redhat.com","username":"pojadhav"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-04-06 09:31:01.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":31075,"name":"Pooja Jadhav","email":"pojadhav@redhat.com","username":"pojadhav"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":31075,"name":"Pooja Jadhav","email":"pojadhav@redhat.com","username":"pojadhav"},"all":[{"value":-1,"date":"2022-04-06 09:27:01.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":31075,"name":"Pooja Jadhav","email":"pojadhav@redhat.com","username":"pojadhav"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31075,"name":"Pooja Jadhav","email":"pojadhav@redhat.com","username":"pojadhav"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-06 09:31:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"276d91aaebe92217a456b6167b40a0b4de242efb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31075,"name":"Pooja Jadhav","email":"pojadhav@redhat.com","username":"pojadhav"},"date":"2022-04-06 07:58:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8585e566b641195545d60b1c99f07542b1717faa","author":{"_account_id":31075,"name":"Pooja Jadhav","email":"pojadhav@redhat.com","username":"pojadhav"},"date":"2022-04-06 09:27:01.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"ef19e6be639aaf27df6cd091e60aaeaa17173584","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-06 09:31:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eea97a9c74d040af96cfe03f245e39ef : SUCCESS in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/540e8bbb5c91456288d199fea1ab47c0 : SUCCESS in 3m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b4d9cdf84b274c289b1ca300ce604cdf : SUCCESS in 4m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9c9b7e6ad04d443fafaba9e6d980d154 : SUCCESS in 4m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/356512a7d8b24bdf80b53e03bca11329 : SUCCESS in 3m 25s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/e50ef4ea46e94d04aff833e0e7479f44 : SUCCESS in 1h 27m 38s","accounts_in_message":[],"_revision_number":1},{"id":"054199da8f2e6b204b4f73dd6f8349e9fb55a8b6","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":31075,"name":"Pooja Jadhav","email":"pojadhav@redhat.com","username":"pojadhav"},"date":"2022-04-06 14:15:08.000000000","message":"Topic ussuri_teardown removed","accounts_in_message":[],"_revision_number":1},{"id":"62b687059efc99e1467a16619f114275761c9421","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":31075,"name":"Pooja Jadhav","email":"pojadhav@redhat.com","username":"pojadhav"},"date":"2022-04-06 14:16:44.000000000","message":"Topic set to tripleo_ussuri_teardown","accounts_in_message":[],"_revision_number":1},{"id":"78232af78acd9558bad4d51ba3cf026cf44bdda7","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2022-04-12 14:49:37.000000000","message":"Abandoned\n\nabandoning per http://lists.openstack.org/pipermail/openstack-discuss/2022-April/028026.html - so we can move EOL https://review.opendev.org/c/openstack/releases/+/834049","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5b141e4a557cdfdef52856195e43bee0806adac2","revisions":{"5b141e4a557cdfdef52856195e43bee0806adac2":{"kind":"REWORK","_number":1,"created":"2022-04-06 07:58:26.000000000","uploader":{"_account_id":31075,"name":"Pooja Jadhav","email":"pojadhav@redhat.com","username":"pojadhav"},"ref":"refs/changes/39/836739/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-net-config","ref":"refs/changes/39/836739/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-net-config refs/changes/39/836739/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-net-config refs/changes/39/836739/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-net-config refs/changes/39/836739/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-net-config refs/changes/39/836739/1"}}},"commit":{"parents":[{"commit":"3ebf87b434a77e9d37c7d4324dda804c5418e6bc","subject":"For linux bond over VFs, disable spoofcheck by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-net-config/commit/3ebf87b434a77e9d37c7d4324dda804c5418e6bc"}]}],"author":{"name":"Pooja Jadhav","email":"pojadhav@redhat.com","date":"2022-04-06 07:57:35.000000000","tz":330},"committer":{"name":"Pooja Jadhav","email":"pojadhav@redhat.com","date":"2022-04-06 07:58:23.000000000","tz":330},"subject":"Ussuri tear down as EOL","message":"Ussuri tear down as EOL\n\nRef: https://review.opendev.org/c/openstack/releases/+/834049\nChange-Id: I3fc3df3fc8282705253911c3d909a8c51447ad31\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-net-config/commit/5b141e4a557cdfdef52856195e43bee0806adac2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-net-config/commit/5b141e4a557cdfdef52856195e43bee0806adac2"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
