)]}'
{"id":"openstack%2Fos-performance-tools~850740","triplet_id":"openstack%2Fos-performance-tools~master~I2c4c327297b74631d961850e735d5ddc69459e47","project":"openstack/os-performance-tools","branch":"master","attention_set":{},"removed_from_attention_set":{"28743":{"account":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"last_update":"2022-08-06 15:41:42.000000000","reason":"Change was submitted"},"22873":{"account":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"last_update":"2022-08-06 15:41:42.000000000","reason":"Change was submitted"},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2022-08-06 15:28:46.000000000","reason":"\u003cGERRIT_ACCOUNT_8556\u003e replied on the change","reason_account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}}},"hashtags":[],"change_id":"I2c4c327297b74631d961850e735d5ddc69459e47","subject":"Replace deprecated readfp method with read_file","status":"MERGED","created":"2022-07-22 09:21:03.000000000","updated":"2022-08-06 15:42:47.000000000","submitted":"2022-08-06 15:41:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"850740","meta_rev_id":"0422faffcd2e427dc43b2d7cf8f837ced368175b","_number":850740,"virtual_id_number":850740,"owner":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":33493,"name":"chenwei","display_name":"chenwei","email":"jzpcoming@163.com","username":"chenwei"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16523,"name":"ZhangLei","display_name":"ZhangLei","email":"zhang.lei@99cloud.net","username":"zhang.lei"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-06 15:41:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"recommended":{"_account_id":16523,"name":"ZhangLei","display_name":"ZhangLei","email":"zhang.lei@99cloud.net","username":"zhang.lei"},"all":[{"value":1,"date":"2022-08-05 02:18:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":33493,"name":"chenwei","display_name":"chenwei","email":"jzpcoming@163.com","username":"chenwei"},{"value":1,"date":"2022-07-22 14:28:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2022-08-05 01:22:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16523,"name":"ZhangLei","display_name":"ZhangLei","email":"zhang.lei@99cloud.net","username":"zhang.lei"},{"value":2,"date":"2022-08-06 15:28:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"_account_id":33493,"name":"chenwei","display_name":"chenwei","email":"jzpcoming@163.com","username":"chenwei"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16523,"name":"ZhangLei","display_name":"ZhangLei","email":"zhang.lei@99cloud.net","username":"zhang.lei"},{"value":1,"date":"2022-08-06 15:28:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":16523,"name":"ZhangLei","display_name":"ZhangLei","email":"zhang.lei@99cloud.net","username":"zhang.lei"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"_account_id":33493,"name":"chenwei","display_name":"chenwei","email":"jzpcoming@163.com","username":"chenwei"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-22 09:31:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-22 14:28:05.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2022-07-25 05:52:00.000000000","updated_by":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"},{"updated":"2022-07-25 05:52:00.000000000","updated_by":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2022-08-05 01:22:59.000000000","updated_by":{"_account_id":16523,"name":"ZhangLei","display_name":"ZhangLei","email":"zhang.lei@99cloud.net","username":"zhang.lei"},"reviewer":{"_account_id":16523,"name":"ZhangLei","display_name":"ZhangLei","email":"zhang.lei@99cloud.net","username":"zhang.lei"},"state":"REVIEWER"},{"updated":"2022-08-05 02:18:01.000000000","updated_by":{"_account_id":33493,"name":"chenwei","display_name":"chenwei","email":"jzpcoming@163.com","username":"chenwei"},"reviewer":{"_account_id":33493,"name":"chenwei","display_name":"chenwei","email":"jzpcoming@163.com","username":"chenwei"},"state":"REVIEWER"}],"messages":[{"id":"e069c86ae05e220267942b8689e062766c413104","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"date":"2022-07-22 09:21:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d2c2f3c918a574f7adddf0d1e9bcd44f5bce760c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-22 09:31:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/338982aaf1674106a2e7ff59abe6bc54\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/548d84c5500642cf97ffb0c7dbe63fe9 : SUCCESS in 7m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/62f700db85974c289794c93e0c899c39 : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87e87e65d3d349d7ab6133765f4052d2 : SUCCESS in 3m 54s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/7ab0665e4cd64c4a86a46b9a23da22d6 : SUCCESS in 3m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dbca1381b76f41c4a0a8c34a8537ba29 : SUCCESS in 3m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2e112c40c38f40d3a1dd2351b9448ba6 : SUCCESS in 3m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3788227b8de649748003cbbf2d34eb05 : SUCCESS in 2m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/47c491208ef54308914995e27d7fb3f1 : SUCCESS in 3m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ffccc96e06bc468d811727f70d168856 : SUCCESS in 2m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"228569abf9a54adedf4337a8e3f43f967cac6863","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-22 14:28:05.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"a48c953a272b12783d24d4830201a6e39901b25a","author":{"_account_id":16523,"name":"ZhangLei","display_name":"ZhangLei","email":"zhang.lei@99cloud.net","username":"zhang.lei"},"date":"2022-08-05 01:22:59.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"41003310e70514e545aa137cdc4b7fb5ad4715e5","author":{"_account_id":33493,"name":"chenwei","display_name":"chenwei","email":"jzpcoming@163.com","username":"chenwei"},"date":"2022-08-05 02:18:01.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"20b9cea752a6c1b2dd1c8f7912f92ad7e80d582f","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-08-06 15:28:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"da3a6261d8c3fa56407d283ea7b48678be0e8e23","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-06 15:29:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"714454f18a3bff032174c581726f463cfc252d20","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-06 15:41:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8fcad2629e0f47fbb0bd90458e9f3446\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5719d10679844199086cf0af7b01d75 : SUCCESS in 5m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/673ac3d68bdc4da6953821f17d3d6aad : SUCCESS in 3m 34s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/b20187f53d864971a2264b05b1ed14a1 : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e7fe8baf7dce482091103f36ebf7c831 : SUCCESS in 4m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c75927622cd048669628817bc6d9891a : SUCCESS in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7c80e6d2bbdb4f029f4c66849f9cb251 : SUCCESS in 3m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4530e0b9df2e45b89e58563735532b1b : SUCCESS in 3m 37s","accounts_in_message":[],"_revision_number":1},{"id":"bc2b6da97929e3e530469efb61e23589e92366dc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-06 15:41:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"0422faffcd2e427dc43b2d7cf8f837ced368175b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-06 15:42:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3fd46df23b2f455689de4ce08fab8b47\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c13c6b802a44b5699583b8aa386af73 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"609f4af8c8b6e1bf091351df1e10b25cc72ef810","revisions":{"609f4af8c8b6e1bf091351df1e10b25cc72ef810":{"kind":"REWORK","_number":1,"created":"2022-07-22 09:21:03.000000000","uploader":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"ref":"refs/changes/40/850740/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-performance-tools","ref":"refs/changes/40/850740/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-performance-tools refs/changes/40/850740/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-performance-tools refs/changes/40/850740/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-performance-tools refs/changes/40/850740/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-performance-tools refs/changes/40/850740/1"}}},"commit":{"parents":[{"commit":"3386bbed482c9d602e1617da0d7de286de2d5ffb","subject":"Add openstack-tox-py310 job as nv","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-performance-tools/commit/3386bbed482c9d602e1617da0d7de286de2d5ffb"}]}],"author":{"name":"ljhuang","email":"huang.liujie@99cloud.net","date":"2022-07-22 08:17:15.000000000","tz":480},"committer":{"name":"ljhuang","email":"huang.liujie@99cloud.net","date":"2022-07-22 09:20:19.000000000","tz":480},"subject":"Replace deprecated readfp method with read_file","message":"Replace deprecated readfp method with read_file\n\nThe readfp method has been deprecated since version 3.2 [1].\n\n[1] https://docs.python.org/3/library/configparser.html?highlight\u003ddeprecated#configparser.ConfigParser.readfp\n\nChange-Id: I2c4c327297b74631d961850e735d5ddc69459e47\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-performance-tools/commit/609f4af8c8b6e1bf091351df1e10b25cc72ef810"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-performance-tools/commit/609f4af8c8b6e1bf091351df1e10b25cc72ef810"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}}]}],"submit_requirements":[]}
