)]}'
{"id":"openstack%2Fos-performance-tools~966108","triplet_id":"openstack%2Fos-performance-tools~master~I6a50dce11163fd59093c9f5c9deeb859b3f4aadd","project":"openstack/os-performance-tools","branch":"master","topic":"test-python3.13","attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2025-11-04 20:10:33.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I6a50dce11163fd59093c9f5c9deeb859b3f4aadd","subject":"Updating testing as per 2026.1 testing runtime","status":"NEW","created":"2025-11-04 19:56:21.000000000","updated":"2026-04-09 17:20:08.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"32528a3076270a620e25b87bac8cb98e0a7e45a0","_number":966108,"virtual_id_number":966108,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-04-09 17:20:08.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-04-09 17:06:49.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-04 20:10:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-09 17:06:49.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"}],"messages":[{"id":"fbcc8e4bbabab17ffa9b4ac8f4ecef1cd7d504db","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-11-04 19:56:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b10ed0fd7d4bae0e95ebf1b554575bfb7e580f70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-04 20:10:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d9ea384c7bf14b948fb1aebb9972b0c3\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/754b415675714205a5b00f6e0f217de1 : SUCCESS in 2m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79d306421ef04afba49b0eda84db24cd : FAILURE in 3m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5a080151f1394de193bd506d82dccd07 : SUCCESS in 3m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b8a625c1fe094b97935b2d242cddab9e : SUCCESS in 2m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ecbf74f2b89c4c199a6ba2f3e4c0db7c : SUCCESS in 2m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/dd20b86024194714b955ae42bfe28fb8 : SUCCESS in 3m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f2e2ac529534018bf9466758d7f469d : SUCCESS in 6m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fc16480c210d435f98e8b01fddf8e2c0 : SUCCESS in 3m 20s","accounts_in_message":[],"_revision_number":1},{"id":"55de8dfa296150e15aa88fe894e68b79d9d493fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"real_author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-04-09 17:06:41.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_8556\u003e\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"_revision_number":2},{"id":"ca08c037571d01bed8cc0d503480cc0a358b714b","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-04-09 17:06:49.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"32528a3076270a620e25b87bac8cb98e0a7e45a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-09 17:20:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bbc5b46b370c4a738f7700a5284c7f52\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/4b0a91be57194920b5510a995912914f : SUCCESS in 2m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/027ad2ae966b4dc5ad069c90172a30db : SUCCESS in 2m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f701040657614b839fc588fec9864909 : SUCCESS in 2m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/79782da3504c43e1b838a8e518d4a674 : SUCCESS in 2m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9f9476eecc6f42cebe74e0342f275659 : SUCCESS in 2m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a73336d0ca4f4daab0c32f343f3453d8 : SUCCESS in 2m 46s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/45009a0acd6a4d82930e489cac5d4248 : SUCCESS in 5m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ff15cf78ef649f99c8133c0d7fb749b : SUCCESS in 6m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/948ec7c17944494caf808642ac4c47a6 : SUCCESS in 2m 47s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0fe2bb30f7f613168f092d8dcac4d2c31f1509a7","revisions":{"411f2a8fca7a72a14b68f9ab39a29395a10dcbd8":{"kind":"REWORK","_number":1,"created":"2025-11-04 19:56:21.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/08/966108/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-performance-tools","ref":"refs/changes/08/966108/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-performance-tools refs/changes/08/966108/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-performance-tools refs/changes/08/966108/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-performance-tools refs/changes/08/966108/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-performance-tools refs/changes/08/966108/1"}}},"commit":{"parents":[{"commit":"525cd14e18e69f5f0fba2b6c7aa30dc48f1add6e","subject":"Update gate jobs as per the 2025.1 cycle testing runtime","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-performance-tools/commit/525cd14e18e69f5f0fba2b6c7aa30dc48f1add6e"}]}],"author":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2025-11-04 19:56:01.000000000","tz":0},"committer":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2025-11-04 19:56:19.000000000","tz":0},"subject":"Updating testing as per 2026.1 testing runtime","message":"Updating testing as per 2026.1 testing runtime\n\nChange-Id: I6a50dce11163fd59093c9f5c9deeb859b3f4aadd\nSigned-off-by: Ghanshyam Maan \u003cgmaan@ghanshyammann.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-performance-tools/commit/411f2a8fca7a72a14b68f9ab39a29395a10dcbd8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-performance-tools/commit/411f2a8fca7a72a14b68f9ab39a29395a10dcbd8"}]},"branch":"refs/heads/master"},"0fe2bb30f7f613168f092d8dcac4d2c31f1509a7":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-04-09 17:06:41.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"real_uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/08/966108/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-performance-tools","ref":"refs/changes/08/966108/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-performance-tools refs/changes/08/966108/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-performance-tools refs/changes/08/966108/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-performance-tools refs/changes/08/966108/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-performance-tools refs/changes/08/966108/2"}}},"commit":{"parents":[{"commit":"7f5aff1d3ace3de4ca3d3b7ccdd9635bf80df261","subject":"Update hacking to 7.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-performance-tools/commit/7f5aff1d3ace3de4ca3d3b7ccdd9635bf80df261"}]}],"author":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2025-11-04 19:56:01.000000000","tz":0},"committer":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2026-04-09 17:06:41.000000000","tz":0},"subject":"Updating testing as per 2026.1 testing runtime","message":"Updating testing as per 2026.1 testing runtime\n\nChange-Id: I6a50dce11163fd59093c9f5c9deeb859b3f4aadd\nSigned-off-by: Ghanshyam Maan \u003cgmaan@ghanshyammann.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-performance-tools/commit/0fe2bb30f7f613168f092d8dcac4d2c31f1509a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-performance-tools/commit/0fe2bb30f7f613168f092d8dcac4d2c31f1509a7"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
