)]}'
{"id":"openstack%2Fos-resource-classes~791974","triplet_id":"openstack%2Fos-resource-classes~master~I663c05734cfc86a3c2bbc068c3990bab037bbe22","project":"openstack/os-resource-classes","branch":"master","topic":"update-min-tox","attention_set":{},"removed_from_attention_set":{"7634":{"account":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"last_update":"2023-02-08 10:31:48.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I663c05734cfc86a3c2bbc068c3990bab037bbe22","subject":"Change minversion of tox to 3.18.0","status":"MERGED","created":"2021-05-18 13:49:21.000000000","updated":"2023-02-08 10:32:52.000000000","submitted":"2023-02-08 10:31:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"791974-update-min-tox","meta_rev_id":"d1a468077d0dc66ed06ddda6c533a69ab0f79f5b","_number":791974,"virtual_id_number":791974,"owner":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-08 10:31:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2022-06-24 13:47:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-02-08 10:23:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-02-08 10:23:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-18 13:57:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-31 11:30:58.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2023-02-08 10:23:45.000000000","updated_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"}],"messages":[{"id":"ae01f7f0659cd477ba1ac718ec54a50e367b2daa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2021-05-18 13:49:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dd42596d210517a7d76447f43268dc49d2dfe06a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 13:57:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f9d72dd7df934bb6b522897619065eb0 : SUCCESS in 3m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/243f7e88c91a491d95b28c0a2525223c : SUCCESS in 3m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/32cd80401f754ead87d341b60edc2ddb : SUCCESS in 3m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a4612aa759154247bdf15ddc9005d868 : SUCCESS in 3m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/547ff117513e4f7fa9b243f28e8d5018 : SUCCESS in 4m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5fc55408a7343b39ecd06915bed9e4e : SUCCESS in 6m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e96cb105116441d5916fb9fb5b085b16 : SUCCESS in 4m 22s","accounts_in_message":[],"_revision_number":1},{"id":"256783c18a84ac5a7c9802c34600c634813ef5e1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2021-05-18 14:02:55.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"07e105e4c3cbcaddaf4a62d5c7b061c1e04d384a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 14:13:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/66a3cb129e8f4ddaa868488bc4b2d93d : SUCCESS in 3m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09b3b24c53ba4848a1b5fbd5439beca3 : SUCCESS in 3m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/02412a508c074ff99f68417cc018c7ea : SUCCESS in 3m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e8fc24ea43ce40c5b93bdcd46c27b98f : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/848d0d93dc1c457a9ae1893ffc73865d : SUCCESS in 3m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80186ec9c84649ecaeec049fb51a6fcb : SUCCESS in 5m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ad56c23b7e9b48229079769b8443cbec : SUCCESS in 4m 16s","accounts_in_message":[],"_revision_number":2},{"id":"dce82930755479501efd856a3604699ec08d0210","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2021-09-04 02:15:21.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"e3f0b8a924582c74d51244e93a6dd3e68771642e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-04 02:28:56.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/957c847a960c46a7a80ad8b6a32215e7 : SUCCESS in 3m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10abf8c954b74f8288111de26af7bd25 : SUCCESS in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3c5f007a90724586b43e361070c7e61f : SUCCESS in 4m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bafbf36dbb1b4f4aa6367d89f5f83b64 : SUCCESS in 4m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/77fde8e348584811a69cfa25f203cf64 : SUCCESS in 4m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee221e6837c14c658bc105da4cd921f0 : SUCCESS in 6m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6290bcb43851463aaf8f954c11b60ffd : SUCCESS in 4m 11s","accounts_in_message":[],"_revision_number":3},{"id":"236768a03308ee16b149fa18d90c127630dc0d7a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2022-03-04 17:04:25.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"5f5ebc57191d33fffc1f4a3bba3adeb317d330c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-04 17:26:43.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/73d303308dda40c591c4378a78a5ab95 : SUCCESS in 6m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28c377d267a345a6a6a248980f86d147 : SUCCESS in 7m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1c4af23dad7f4fa4abbbb319b5674077 : SUCCESS in 5m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01ddcdf2a6be4d1eb17a4172e1ea9a0e : SUCCESS in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3f7c35240ef847a49b11d6438221bb1a : SUCCESS in 5m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1819bff47a5d4a9e85665851959337d3 : SUCCESS in 5m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ce13a909cb00440fa51aaf1cb7ff891b : SUCCESS in 4m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ff5a5193de454bc8a03c3935845e57f7 : SUCCESS in 4m 38s","accounts_in_message":[],"_revision_number":4},{"id":"fd3df8a7deef67b472df801308c37122845e54af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2022-05-22 12:55:32.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"989c84b9cb876501c76dd73850238ac4ea8de414","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-22 13:08:41.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/64d190b6af334be0b442c14023e232a6 : SUCCESS in 2m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7913a2e60f94181b3110b80e2a39199 : SUCCESS in 6m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3675c278b12d45a5983d888be0c2eb49 : SUCCESS in 3m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/715fcc6a6f1449c0a051c89dd96d8178 : SUCCESS in 2m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a8b07519174d4225a9b725a69529c4e0 : SUCCESS in 3m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0acd78d2c98f46648c21d3febcfc39e6 : SUCCESS in 3m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/073c09d54b614c5389abf9e27cd6bf4a : SUCCESS in 2m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ccd4c007ea7948dbb5a9295d526b9cbc : SUCCESS in 3m 08s","accounts_in_message":[],"_revision_number":5},{"id":"1e5b8626758792c8ef651cafd0d2507688b56ca7","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-05-31 11:30:58.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"364b4b99521608bdc121762b42799b5d136adbe5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-31 11:31:40.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"8b5ce4e882602d199ddc51e14724361355a4b17e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-31 11:45:21.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3b674fecc5a4da9b62410c1f71f54e9 : SUCCESS in 5m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3941622dfdeb45e2b0c6424a14c3bd28 : SUCCESS in 2m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55490a65898e46f29317baf126a73372 : SUCCESS in 2m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2ffa950ef4394ba981075a2f61ce5745 : FAILURE in 2m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4c0b21f5c2414ed8bd92746cc70f65a5 : SUCCESS in 4m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d11123bafb0648c4b3c9d84930af7332 : SUCCESS in 2m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7376a6971491411089cc6dead13d3067 : SUCCESS in 3m 03s","accounts_in_message":[],"_revision_number":5},{"id":"188e6410888d433e434a081355538a0608bb66e7","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2022-06-24 08:40:23.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"aacf89c49f9c417c04e285837b7d883fcd1af389","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-24 08:50:33.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f53273c6f1ca4e39a54d42c722ca6632 : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8faaab8341e34ec99ee0e07682b48d9e : SUCCESS in 5m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/34163e8494ed4b6c91eb252b11aa4dfd : SUCCESS in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2fd956f7e1954721af309cc8615319c7 : SUCCESS in 2m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e0018ec05eab4df69ad6ae909187806a : FAILURE in 2m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c57fdd12a4f24b7ba7369a80c67980e5 : SUCCESS in 3m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5a6780a16a6b41049e61310044218d72 : SUCCESS in 2m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/066a87c4e4b94f76a932366e49d153da : SUCCESS in 3m 20s","accounts_in_message":[],"_revision_number":5},{"id":"cb8061d6d50c55bbc3168eb532fcc527b9f48871","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2022-06-24 13:47:41.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"4261389d6c71759c809d9359a1be85f66f478967","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-24 13:59:09.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/541d129e3b514c3198098d00a9ded2cb : SUCCESS in 2m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ca052eb65834642b5726eaaeeec9dbb : SUCCESS in 3m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b3af16747d59458f950ff88d2f572c92 : SUCCESS in 2m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/06c54cda59a44e81834cb53557353cee : SUCCESS in 3m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b6f9c116f3514c6495d37a297e8c5616 : SUCCESS in 3m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4146eb5172e0420db306f800586ade35 : SUCCESS in 4m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2240a99be0cb46edb514f5560e4c1fcd : SUCCESS in 2m 56s","accounts_in_message":[],"_revision_number":6},{"id":"27be4a546ab48be83d546f7b25a401fc0d05968d","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2022-08-08 07:15:04.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"b3f78fe55f55de632ea4fc6a6688b7cedb899f5b","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2023-02-08 10:23:45.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"6eefb889fe37d048fd56d565a8c6421a50d31b6a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-08 10:24:10.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"5e53d31535348d4c1f73e7b2fe00907514eec86f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-08 10:31:48.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2339f798889e416aa39e46b7281aba9f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c93d3682b284442b05e190a4b38e221 : SUCCESS in 2m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6bfe3d6b12e04da297abb482f5e05669 : SUCCESS in 3m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/59df29371b5f4c85962c2ba6179d20a1 : SUCCESS in 2m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0bbaa64d302b424999ad73bf2f89323e : SUCCESS in 5m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/53c7e545fc824dc49825c07c1b5bacb6 : SUCCESS in 2m 40s","accounts_in_message":[],"_revision_number":6},{"id":"01150c0b91791fd7a935a4b6a72f1daa930c8450","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-08 10:31:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"d1a468077d0dc66ed06ddda6c533a69ab0f79f5b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-08 10:32:52.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/79f6fd9b42cc496b9b89bbf7ee43dce6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/551f641ae8aa4b279872949532f26e0b : SUCCESS in 41s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b8642ed4651447c5acd29fca3c47a120 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"9601b8942a61f6af29f920ea12b74d11da5f19a0","revisions":{"e771e854b72fac8581e1e85378f84acf857bf77a":{"kind":"REWORK","_number":1,"created":"2021-05-18 13:49:21.000000000","uploader":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"ref":"refs/changes/74/791974/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-resource-classes","ref":"refs/changes/74/791974/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-resource-classes refs/changes/74/791974/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-resource-classes refs/changes/74/791974/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-resource-classes refs/changes/74/791974/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-resource-classes refs/changes/74/791974/1"}}},"commit":{"parents":[{"commit":"145ed791d050725394dbd786a5d45600da656812","subject":"Merge \"Fix hacking min version to 3.0.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-resource-classes/commit/145ed791d050725394dbd786a5d45600da656812"}]}],"author":{"name":"Takashi Natsume","email":"takanattie@gmail.com","date":"2021-05-18 13:41:11.000000000","tz":540},"committer":{"name":"Takashi Natsume","email":"takanattie@gmail.com","date":"2021-05-18 13:41:11.000000000","tz":540},"subject":"Changed minversion in tox to 3.18.0","message":"Changed minversion in tox to 3.18.0\n\nThe patch bumps min version of tox to 3.18.0 python in order to\nreplace tox\u0027s whitelist_externals by allowlist_externals option:\nhttps://github.com/tox-dev/tox/blob/master/docs/changelog.rst#v3180-2020-07-23\n\nSigned-off-by: Takashi Natsume \u003ctakanattie@gmail.com\u003e\nChange-Id: I663c05734cfc86a3c2bbc068c3990bab037bbe22\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-resource-classes/commit/e771e854b72fac8581e1e85378f84acf857bf77a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-resource-classes/commit/e771e854b72fac8581e1e85378f84acf857bf77a"}]},"branch":"refs/heads/master"},"c73da979e71302246f6474c7ef146767e680bf9c":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-05-18 14:02:55.000000000","uploader":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"ref":"refs/changes/74/791974/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-resource-classes","ref":"refs/changes/74/791974/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-resource-classes refs/changes/74/791974/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-resource-classes refs/changes/74/791974/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-resource-classes refs/changes/74/791974/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-resource-classes refs/changes/74/791974/2"}}},"commit":{"parents":[{"commit":"145ed791d050725394dbd786a5d45600da656812","subject":"Merge \"Fix hacking min version to 3.0.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-resource-classes/commit/145ed791d050725394dbd786a5d45600da656812"}]}],"author":{"name":"Takashi Natsume","email":"takanattie@gmail.com","date":"2021-05-18 13:41:11.000000000","tz":540},"committer":{"name":"Takashi Natsume","email":"takanattie@gmail.com","date":"2021-05-18 14:02:28.000000000","tz":540},"subject":"Change minversion of tox to 3.18.0","message":"Change minversion of tox to 3.18.0\n\nThe patch bumps min version of tox to 3.18.0 in order to\nreplace whitelist_externals with allowlist_externals option:\nhttps://github.com/tox-dev/tox/blob/master/docs/changelog.rst#v3180-2020-07-23\n\nSigned-off-by: Takashi Natsume \u003ctakanattie@gmail.com\u003e\nChange-Id: I663c05734cfc86a3c2bbc068c3990bab037bbe22\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-resource-classes/commit/c73da979e71302246f6474c7ef146767e680bf9c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-resource-classes/commit/c73da979e71302246f6474c7ef146767e680bf9c"}]},"branch":"refs/heads/master"},"323f15e8b7a6f3d390980da3452cc32e2c747cbb":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-09-04 02:15:21.000000000","uploader":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"ref":"refs/changes/74/791974/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-resource-classes","ref":"refs/changes/74/791974/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-resource-classes refs/changes/74/791974/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-resource-classes refs/changes/74/791974/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-resource-classes refs/changes/74/791974/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-resource-classes refs/changes/74/791974/3"}}},"commit":{"parents":[{"commit":"357bd404d8e84ee8ccf0156f193d6810266d3e72","subject":"Merge \"Add packet rate related resource classes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-resource-classes/commit/357bd404d8e84ee8ccf0156f193d6810266d3e72"}]}],"author":{"name":"Takashi Natsume","email":"takanattie@gmail.com","date":"2021-05-18 13:41:11.000000000","tz":540},"committer":{"name":"Takashi Natsume","email":"takanattie@gmail.com","date":"2021-09-04 02:15:21.000000000","tz":0},"subject":"Change minversion of tox to 3.18.0","message":"Change minversion of tox to 3.18.0\n\nThe patch bumps min version of tox to 3.18.0 in order to\nreplace whitelist_externals with allowlist_externals option:\nhttps://github.com/tox-dev/tox/blob/master/docs/changelog.rst#v3180-2020-07-23\n\nSigned-off-by: Takashi Natsume \u003ctakanattie@gmail.com\u003e\nChange-Id: I663c05734cfc86a3c2bbc068c3990bab037bbe22\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-resource-classes/commit/323f15e8b7a6f3d390980da3452cc32e2c747cbb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-resource-classes/commit/323f15e8b7a6f3d390980da3452cc32e2c747cbb"}]},"branch":"refs/heads/master","description":"Rebase"},"e8308c6a8ee5b0f89146d2068aa92acddee5b85e":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2022-03-04 17:04:25.000000000","uploader":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"ref":"refs/changes/74/791974/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-resource-classes","ref":"refs/changes/74/791974/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-resource-classes refs/changes/74/791974/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-resource-classes refs/changes/74/791974/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-resource-classes refs/changes/74/791974/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-resource-classes refs/changes/74/791974/4"}}},"commit":{"parents":[{"commit":"e532b71313058959976705ea3e18a45799b3cd47","subject":"Test all supported python version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-resource-classes/commit/e532b71313058959976705ea3e18a45799b3cd47"}]}],"author":{"name":"Takashi Natsume","email":"takanattie@gmail.com","date":"2021-05-18 13:41:11.000000000","tz":540},"committer":{"name":"Takashi Natsume","email":"takanattie@gmail.com","date":"2022-03-04 17:04:25.000000000","tz":0},"subject":"Change minversion of tox to 3.18.0","message":"Change minversion of tox to 3.18.0\n\nThe patch bumps min version of tox to 3.18.0 in order to\nreplace whitelist_externals with allowlist_externals option:\nhttps://github.com/tox-dev/tox/blob/master/docs/changelog.rst#v3180-2020-07-23\n\nSigned-off-by: Takashi Natsume \u003ctakanattie@gmail.com\u003e\nChange-Id: I663c05734cfc86a3c2bbc068c3990bab037bbe22\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-resource-classes/commit/e8308c6a8ee5b0f89146d2068aa92acddee5b85e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-resource-classes/commit/e8308c6a8ee5b0f89146d2068aa92acddee5b85e"}]},"branch":"refs/heads/master","description":"Rebase"},"bb4f7bfe80638255b9cddb0326ac23674794782e":{"kind":"REWORK","_number":5,"created":"2022-05-22 12:55:32.000000000","uploader":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"ref":"refs/changes/74/791974/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-resource-classes","ref":"refs/changes/74/791974/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-resource-classes refs/changes/74/791974/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-resource-classes refs/changes/74/791974/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-resource-classes refs/changes/74/791974/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-resource-classes refs/changes/74/791974/5"}}},"commit":{"parents":[{"commit":"5d0738254a3fcb2d1cf43828017e8fcefc87adff","subject":"Merge \"Remove unnecessary unicode prefixes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-resource-classes/commit/5d0738254a3fcb2d1cf43828017e8fcefc87adff"}]}],"author":{"name":"Takashi Natsume","email":"takanattie@gmail.com","date":"2021-05-18 13:41:11.000000000","tz":540},"committer":{"name":"Takashi Natsume","email":"takanattie@gmail.com","date":"2022-05-22 12:54:57.000000000","tz":540},"subject":"Change minversion of tox to 3.18.0","message":"Change minversion of tox to 3.18.0\n\nThe patch bumps min version of tox to 3.18.0 in order to\nreplace whitelist_externals with allowlist_externals option:\nhttps://github.com/tox-dev/tox/blob/master/docs/changelog.rst#v3180-2020-07-23\n\nSigned-off-by: Takashi Natsume \u003ctakanattie@gmail.com\u003e\nChange-Id: I663c05734cfc86a3c2bbc068c3990bab037bbe22\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-resource-classes/commit/bb4f7bfe80638255b9cddb0326ac23674794782e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-resource-classes/commit/bb4f7bfe80638255b9cddb0326ac23674794782e"}]},"branch":"refs/heads/master"},"9601b8942a61f6af29f920ea12b74d11da5f19a0":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2022-06-24 13:47:41.000000000","uploader":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"ref":"refs/changes/74/791974/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-resource-classes","ref":"refs/changes/74/791974/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-resource-classes refs/changes/74/791974/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-resource-classes refs/changes/74/791974/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-resource-classes refs/changes/74/791974/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-resource-classes refs/changes/74/791974/6"}}},"commit":{"parents":[{"commit":"71c89f0079bf53e54bb72a62c4da6f097542c5aa","subject":"Merge \"setup: Replace dashes with underscores, add links\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-resource-classes/commit/71c89f0079bf53e54bb72a62c4da6f097542c5aa"}]}],"author":{"name":"Takashi Natsume","email":"takanattie@gmail.com","date":"2021-05-18 13:41:11.000000000","tz":540},"committer":{"name":"Takashi Natsume","email":"takanattie@gmail.com","date":"2022-06-24 13:47:41.000000000","tz":0},"subject":"Change minversion of tox to 3.18.0","message":"Change minversion of tox to 3.18.0\n\nThe patch bumps min version of tox to 3.18.0 in order to\nreplace whitelist_externals with allowlist_externals option:\nhttps://github.com/tox-dev/tox/blob/master/docs/changelog.rst#v3180-2020-07-23\n\nSigned-off-by: Takashi Natsume \u003ctakanattie@gmail.com\u003e\nChange-Id: I663c05734cfc86a3c2bbc068c3990bab037bbe22\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-resource-classes/commit/9601b8942a61f6af29f920ea12b74d11da5f19a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-resource-classes/commit/9601b8942a61f6af29f920ea12b74d11da5f19a0"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
