)]}'
{"id":"openstack%2Fos-testr~573636","triplet_id":"openstack%2Fos-testr~master~I3a6084db9f86627e3e94abaa4fb4aec52a01126a","project":"openstack/os-testr","branch":"master","topic":"deprecate-ostestr-command","hashtags":[],"change_id":"I3a6084db9f86627e3e94abaa4fb4aec52a01126a","subject":"Deprecate ostestr command","status":"MERGED","created":"2018-06-08 09:12:49.000000000","updated":"2019-06-18 13:44:34.000000000","submitted":"2019-06-18 13:44:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"573636-1560865474912-eb0806c3","meta_rev_id":"4fa83466dac5b030e73b63ce5de689d25807b03b","_number":573636,"virtual_id_number":573636,"owner":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":2,"date":"2019-06-18 13:44:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"recommended":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2019-06-03 11:57:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-06-03 14:33:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-18 12:46:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-06-26 00:11:42.000000000","updated_by":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"reviewer":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"state":"REVIEWER"},{"updated":"2019-06-03 11:57:20.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-06-18 12:46:08.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2019-06-18 13:44:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ca1c64ca4764d6caec722a299ce03371fffd5a15","author":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"date":"2018-06-08 09:12:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a8d9b056d73bc9f59899f080866bbfa053adbaa8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-08 12:31:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/36/573636/1/check/openstack-tox-pep8/767be9e/ : SUCCESS in 4m 39s\n- build-openstack-sphinx-docs http://logs.openstack.org/36/573636/1/check/build-openstack-sphinx-docs/149cbf9/html/ : SUCCESS in 5m 11s","accounts_in_message":[],"_revision_number":1},{"id":"c01cf071d505412311344cfc80fbf86727923a35","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2018-06-26 00:11:10.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\ndeprecation idea looks good. Along with doc update- \n- can we add deprecation warning in command also.\n- add reno for this which can also tell how people can migrate to stestr","accounts_in_message":[],"_revision_number":1},{"id":"154e2f1ffbf3280efa24dea3347001e16b9d9376","author":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"date":"2018-06-26 02:36:53.000000000","message":"Patch Set 1:\n\n(2 comments)\n\n\u003e deprecation idea looks good. Along with doc update-\n \u003e - can we add deprecation warning in command also.\n \u003e - add reno for this which can also tell how people can migrate to\n\nyeah, it would be ideal. But I have a mixed feeling to do that because I don\u0027t want to change the code and behavior as possible for now. And, we need to release a new os-testr for the deprecation. I\u0027m not sure it\u0027s a really worthwhile effort.\n\nBut regarding to this docs, it\u0027ll be updated on the docs site[0] after merging this patch automatically. So, users can see the latest status.\n\n[0] https://docs.openstack.org/os-testr/latest/user/ostestr.html","accounts_in_message":[],"_revision_number":1},{"id":"8b12ce5417bdd416e7e7d764ea803866a6d314e4","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2018-06-26 02:56:01.000000000","message":"Patch Set 1:\n\nactually without release or warning print from CLI, it is difficult for users to know whether it is deprecated or not. Doc are not always read regularly. \n\nTo notify the deprecation, we have to cut a release saying this is end of this pckg and stestr is alternate for this pckg. So that when user update it they get to know the actual state of project clearly.","accounts_in_message":[],"_revision_number":1},{"id":"6357850bc594295e52456fa4ad31311aafe5c5d0","author":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"date":"2019-04-19 05:56:11.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"359cbeb123d31845c08bc562bc69882cc64b2280","author":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"date":"2019-04-19 05:58:52.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b03057bc0cbf5f42c9e9d4578980b699808144a9","author":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"date":"2019-04-19 06:01:19.000000000","message":"Patch Set 3:\n\n@gmann, I\u0027ve added warning message to the command and releasenote, etc.","accounts_in_message":[],"_revision_number":3},{"id":"f893b45013ff9cf0b4d34c17096ebf8406ed5227","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-19 06:12:19.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/36/573636/3/check/test-release-openstack/8e9af30/ : SUCCESS in 3m 06s\n- requirements-check http://logs.openstack.org/36/573636/3/check/requirements-check/13f799a/ : SUCCESS in 3m 27s\n- openstack-tox-pep8 http://logs.openstack.org/36/573636/3/check/openstack-tox-pep8/7c520e0/ : SUCCESS in 4m 07s\n- openstack-tox-py27 http://logs.openstack.org/36/573636/3/check/openstack-tox-py27/0d7034f/ : SUCCESS in 4m 53s\n- openstack-tox-py35 http://logs.openstack.org/36/573636/3/check/openstack-tox-py35/5113ded/ : SUCCESS in 4m 34s\n- openstack-tox-py36 http://logs.openstack.org/36/573636/3/check/openstack-tox-py36/24c0f7e/ : SUCCESS in 4m 37s\n- openstack-tox-py37 http://logs.openstack.org/36/573636/3/check/openstack-tox-py37/cb5ba86/ : SUCCESS in 4m 45s\n- openstack-tox-docs http://logs.openstack.org/36/573636/3/check/openstack-tox-docs/ac7ae4e/html/ : SUCCESS in 4m 46s\n- build-openstack-releasenotes http://logs.openstack.org/36/573636/3/check/build-openstack-releasenotes/779fe39/ : FAILURE in 3m 58s","accounts_in_message":[],"_revision_number":3},{"id":"0dcb3b71b264f37a60a11a29512288a741bac06f","author":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"date":"2019-04-22 09:02:23.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"d9bad596562248bdf69fff4e732d1fafb596b174","author":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"date":"2019-04-22 09:04:42.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"2fd573e272ff07c3c5877825ada244fc6e04cbae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 09:14:32.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/36/573636/5/check/test-release-openstack/d1f7004/ : SUCCESS in 3m 01s\n- requirements-check http://logs.openstack.org/36/573636/5/check/requirements-check/b1a5b99/ : FAILURE in 2m 56s\n- openstack-tox-pep8 http://logs.openstack.org/36/573636/5/check/openstack-tox-pep8/3cd4e64/ : SUCCESS in 4m 19s\n- openstack-tox-py27 http://logs.openstack.org/36/573636/5/check/openstack-tox-py27/e835fa1/ : SUCCESS in 4m 33s\n- openstack-tox-py35 http://logs.openstack.org/36/573636/5/check/openstack-tox-py35/f6c1073/ : SUCCESS in 6m 24s\n- openstack-tox-py36 http://logs.openstack.org/36/573636/5/check/openstack-tox-py36/ecdb1c8/ : SUCCESS in 6m 27s\n- openstack-tox-py37 http://logs.openstack.org/36/573636/5/check/openstack-tox-py37/2508efd/ : SUCCESS in 7m 17s\n- openstack-tox-docs http://logs.openstack.org/36/573636/5/check/openstack-tox-docs/0559bac/html/ : SUCCESS in 4m 07s\n- build-openstack-releasenotes http://logs.openstack.org/36/573636/5/check/build-openstack-releasenotes/da413dc/html/ : SUCCESS in 3m 48s","accounts_in_message":[],"_revision_number":5},{"id":"eef19b9091dc2c50ba46ca84ec43935bc351d690","author":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"date":"2019-05-25 11:16:25.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"bfa71aca6387d07bcaa7c6a8a4598efe787da380","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-25 11:25:01.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/36/573636/5/check/test-release-openstack/bd74498/ : SUCCESS in 4m 15s\n- requirements-check http://logs.openstack.org/36/573636/5/check/requirements-check/6c8abe9/ : FAILURE in 3m 42s\n- openstack-tox-pep8 http://logs.openstack.org/36/573636/5/check/openstack-tox-pep8/57eb5f1/ : SUCCESS in 4m 59s\n- openstack-tox-py27 http://logs.openstack.org/36/573636/5/check/openstack-tox-py27/839e860/ : FAILURE in 4m 41s\n- openstack-tox-py35 http://logs.openstack.org/36/573636/5/check/openstack-tox-py35/bfe8fb2/ : FAILURE in 4m 24s\n- openstack-tox-py36 http://logs.openstack.org/36/573636/5/check/openstack-tox-py36/43541ee/ : FAILURE in 4m 40s\n- openstack-tox-py37 http://logs.openstack.org/36/573636/5/check/openstack-tox-py37/e062f0b/ : FAILURE in 4m 34s\n- openstack-tox-docs http://logs.openstack.org/36/573636/5/check/openstack-tox-docs/282f5e6/html/ : SUCCESS in 4m 18s\n- build-openstack-releasenotes http://logs.openstack.org/36/573636/5/check/build-openstack-releasenotes/2f71117/html/ : SUCCESS in 4m 06s","accounts_in_message":[],"_revision_number":5},{"id":"4f47a29c3fc85acb91ffaf8f6cf938afe3cff190","author":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"date":"2019-05-25 11:34:25.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"0c3be34bbf002ee12bf2c5a2b016024085701dd0","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-05-25 11:40:35.000000000","message":"Patch Set 6:\n\n(2 comments)\n\nThanks!","accounts_in_message":[],"_revision_number":6},{"id":"cc22b53be78e26cea45694edcbc880d3ef70f707","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-25 11:45:02.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/36/573636/6/check/test-release-openstack/f27d991/ : SUCCESS in 4m 09s\n- requirements-check http://logs.openstack.org/36/573636/6/check/requirements-check/e7f9a51/ : FAILURE in 3m 51s\n- openstack-tox-pep8 http://logs.openstack.org/36/573636/6/check/openstack-tox-pep8/3c87fae/ : SUCCESS in 4m 08s\n- openstack-tox-py27 http://logs.openstack.org/36/573636/6/check/openstack-tox-py27/e35c919/ : FAILURE in 4m 06s\n- openstack-tox-py35 http://logs.openstack.org/36/573636/6/check/openstack-tox-py35/fd52116/ : FAILURE in 4m 11s\n- openstack-tox-py36 http://logs.openstack.org/36/573636/6/check/openstack-tox-py36/e258259/ : FAILURE in 3m 30s\n- openstack-tox-py37 http://logs.openstack.org/36/573636/6/check/openstack-tox-py37/5a15ade/ : FAILURE in 4m 21s\n- openstack-tox-docs http://logs.openstack.org/36/573636/6/check/openstack-tox-docs/a7a297e/ : POST_FAILURE in 3m 39s\n- build-openstack-releasenotes http://logs.openstack.org/36/573636/6/check/build-openstack-releasenotes/eb6c895/html/ : SUCCESS in 3m 51s","accounts_in_message":[],"_revision_number":6},{"id":"a255b556956fb3ede6b275a2448a9defd78afd45","author":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"date":"2019-05-25 12:15:09.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"099e906875273d1fcf3fc51516d0871a130442d7","author":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"date":"2019-05-25 12:17:28.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"45db55e1bea7c7a92c8e23e0b664a3f489c40df9","author":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"date":"2019-05-25 12:18:13.000000000","message":"Patch Set 6:\n\n(2 comments)\n\n@Sean, thanks for your comment! Both are fixed.","accounts_in_message":[],"_revision_number":6},{"id":"f30b44fe1221316dbefc8805a2f256de5a0b1d18","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-25 12:26:42.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/36/573636/8/check/test-release-openstack/b7a1795/ : SUCCESS in 3m 04s\n- requirements-check http://logs.openstack.org/36/573636/8/check/requirements-check/adb2d7a/ : SUCCESS in 3m 18s\n- openstack-tox-pep8 http://logs.openstack.org/36/573636/8/check/openstack-tox-pep8/da6b601/ : SUCCESS in 3m 39s\n- openstack-tox-py27 http://logs.openstack.org/36/573636/8/check/openstack-tox-py27/07b4b9f/ : FAILURE in 3m 39s\n- openstack-tox-py35 http://logs.openstack.org/36/573636/8/check/openstack-tox-py35/f7e63e4/ : FAILURE in 3m 45s\n- openstack-tox-py36 http://logs.openstack.org/36/573636/8/check/openstack-tox-py36/2e9fa2e/ : FAILURE in 3m 48s\n- openstack-tox-py37 http://logs.openstack.org/36/573636/8/check/openstack-tox-py37/7f997f1/ : FAILURE in 3m 56s\n- openstack-tox-docs http://logs.openstack.org/36/573636/8/check/openstack-tox-docs/94f0497/ : POST_FAILURE in 3m 40s\n- build-openstack-releasenotes http://logs.openstack.org/36/573636/8/check/build-openstack-releasenotes/7bdaebc/html/ : SUCCESS in 3m 34s","accounts_in_message":[],"_revision_number":8},{"id":"183efe0ea4180e2cbe470154b706b5400cf22177","author":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"date":"2019-05-31 06:07:03.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"5dc96479104588b8ddd86b8ee5814310e76cb658","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-31 06:39:52.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/36/573636/8/check/test-release-openstack/1c22353/ : SUCCESS in 3m 37s\n- requirements-check http://logs.openstack.org/36/573636/8/check/requirements-check/e8f8f0c/ : SUCCESS in 3m 14s\n- openstack-tox-pep8 http://logs.openstack.org/36/573636/8/check/openstack-tox-pep8/6b6a2ce/ : SUCCESS in 3m 59s\n- openstack-tox-py27 http://logs.openstack.org/36/573636/8/check/openstack-tox-py27/b70528c/ : FAILURE in 3m 48s\n- openstack-tox-py35 http://logs.openstack.org/36/573636/8/check/openstack-tox-py35/d822116/ : FAILURE in 3m 52s\n- openstack-tox-py36 http://logs.openstack.org/36/573636/8/check/openstack-tox-py36/727d3d7/ : FAILURE in 3m 59s\n- openstack-tox-py37 http://logs.openstack.org/36/573636/8/check/openstack-tox-py37/9bddf92/ : FAILURE in 4m 10s\n- openstack-tox-docs http://logs.openstack.org/36/573636/8/check/openstack-tox-docs/463552b/ : POST_FAILURE in 4m 13s\n- build-openstack-releasenotes http://logs.openstack.org/36/573636/8/check/build-openstack-releasenotes/6490147/html/ : SUCCESS in 5m 04s","accounts_in_message":[],"_revision_number":8},{"id":"0e68b2c931414754da5a712ba6d0c097343874bf","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-06-02 13:37:37.000000000","message":"Patch Set 8:\n\nthis lgtm. I will +2 once you fix the gate failure.","accounts_in_message":[],"_revision_number":8},{"id":"1bc8fe8d5878dfc423c26a5821070ae4d7a4b409","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-06-02 14:50:58.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"c679625c515b597c3a415da155efab1b918f0aeb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-02 14:58:47.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/36/573636/8/check/test-release-openstack/d8aedad/ : SUCCESS in 3m 51s\n- requirements-check http://logs.openstack.org/36/573636/8/check/requirements-check/023d830/ : SUCCESS in 4m 09s\n- openstack-tox-pep8 http://logs.openstack.org/36/573636/8/check/openstack-tox-pep8/0d458d1/ : SUCCESS in 4m 07s\n- openstack-tox-py27 http://logs.openstack.org/36/573636/8/check/openstack-tox-py27/f31ec65/ : SUCCESS in 4m 20s\n- openstack-tox-py35 http://logs.openstack.org/36/573636/8/check/openstack-tox-py35/b75ff1a/ : SUCCESS in 4m 18s\n- openstack-tox-py36 http://logs.openstack.org/36/573636/8/check/openstack-tox-py36/3e3808b/ : SUCCESS in 4m 13s\n- openstack-tox-py37 http://logs.openstack.org/36/573636/8/check/openstack-tox-py37/24924d4/ : SUCCESS in 4m 35s\n- openstack-tox-docs http://logs.openstack.org/36/573636/8/check/openstack-tox-docs/0e87e0d/ : POST_FAILURE in 4m 02s\n- build-openstack-releasenotes http://logs.openstack.org/36/573636/8/check/build-openstack-releasenotes/394c957/html/ : SUCCESS in 3m 54s","accounts_in_message":[],"_revision_number":8},{"id":"e3ef6255a427a6a527586f2cce9d4b5c968bcce7","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-06-02 15:13:22.000000000","message":"Patch Set 8: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"d220528acc01b80187a78c01e810aa05d667c8bc","author":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"date":"2019-06-03 08:32:51.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"6bf76b95a845de94be95df93ba3e491419c2e07d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-03 08:53:08.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/36/573636/9/check/test-release-openstack/9b20604/ : SUCCESS in 3m 03s\n- requirements-check http://logs.openstack.org/36/573636/9/check/requirements-check/240add4/ : SUCCESS in 3m 34s\n- openstack-tox-pep8 http://logs.openstack.org/36/573636/9/check/openstack-tox-pep8/eb31610/ : SUCCESS in 4m 04s\n- openstack-tox-py27 http://logs.openstack.org/36/573636/9/check/openstack-tox-py27/68e8c61/ : SUCCESS in 4m 00s\n- openstack-tox-py36 http://logs.openstack.org/36/573636/9/check/openstack-tox-py36/0be841f/ : SUCCESS in 4m 15s\n- openstack-tox-py37 http://logs.openstack.org/36/573636/9/check/openstack-tox-py37/a76bf4c/ : SUCCESS in 4m 00s\n- openstack-tox-docs http://logs.openstack.org/36/573636/9/check/openstack-tox-docs/385568c/ : FAILURE in 3m 46s\n- build-openstack-releasenotes http://logs.openstack.org/36/573636/9/check/build-openstack-releasenotes/cf8b0d8/html/ : SUCCESS in 3m 13s","accounts_in_message":[],"_revision_number":9},{"id":"8d1a1f5556622211d044e99902aded9206a83c94","author":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"date":"2019-06-03 09:44:33.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"f293cb4b75139dac59134d7e1d37d80c02b62ff3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-03 10:04:50.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/36/573636/10/check/test-release-openstack/70a9945/ : SUCCESS in 3m 06s\n- requirements-check http://logs.openstack.org/36/573636/10/check/requirements-check/59ba524/ : SUCCESS in 3m 10s\n- openstack-tox-pep8 http://logs.openstack.org/36/573636/10/check/openstack-tox-pep8/2919b24/ : SUCCESS in 3m 57s\n- openstack-tox-py27 http://logs.openstack.org/36/573636/10/check/openstack-tox-py27/0723290/ : SUCCESS in 4m 17s\n- openstack-tox-py36 http://logs.openstack.org/36/573636/10/check/openstack-tox-py36/02a0282/ : SUCCESS in 3m 36s\n- openstack-tox-py37 http://logs.openstack.org/36/573636/10/check/openstack-tox-py37/1199913/ : SUCCESS in 4m 59s\n- openstack-tox-docs http://logs.openstack.org/36/573636/10/check/openstack-tox-docs/7af6dc9/html/ : SUCCESS in 3m 53s\n- build-openstack-releasenotes http://logs.openstack.org/36/573636/10/check/build-openstack-releasenotes/a2f55dc/html/ : SUCCESS in 4m 12s","accounts_in_message":[],"_revision_number":10},{"id":"60aaec841829e90fd39db3d3bb38ba0d3ff88f9c","author":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"date":"2019-06-03 10:11:11.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"d56afe8052bf0c2628e0ca520c2133643e58e1e6","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-06-03 11:57:20.000000000","message":"Patch Set 10: Code-Review+1\n\n\u003e (1 comment)\n\nI would say just get rid of it, though my argument for that probably equally applies to just keeping it like the latest patchset.\n\nSince we are deprecating ostestr, it doesn\u0027t really matter too much. But perhaps you are right and it\u0027s better to just keep that in and not update to follow the current PTI since it soon won\u0027t matter anymore and then we will at least have left behind those docs.\n\nIt can all be pulled from git, but given the current state of this project, the path of least resistance is probably best.","accounts_in_message":[],"_revision_number":10},{"id":"5ff565d9dc69118b73fb916b422542c7c7449f56","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-06-03 14:33:16.000000000","message":"Patch Set 10: Code-Review+2\n\nlgtm, thanks.","accounts_in_message":[],"_revision_number":10},{"id":"0a87b4c3743c01610b084b2cda77b73493bb8984","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-06-18 12:46:08.000000000","message":"Patch Set 10: Workflow+1\n\nthere are not much reviewer here so approving it.","accounts_in_message":[],"_revision_number":10},{"id":"328764d5e8b26b085ff0c13f3d3206a391d67b66","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 12:46:21.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"df6ac8acde08cfd8eb62a274b409da467ec30413","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 13:44:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":10},{"id":"da1b44f4c6b6a11d545f1422ef8557b1f7e131d9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 13:44:34.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack http://logs.openstack.org/36/573636/10/gate/test-release-openstack/8c95cbc/ : SUCCESS in 4m 25s\n- requirements-check http://logs.openstack.org/36/573636/10/gate/requirements-check/5348c96/ : SUCCESS in 3m 31s\n- openstack-tox-pep8 http://logs.openstack.org/36/573636/10/gate/openstack-tox-pep8/604a80a/ : SUCCESS in 4m 18s\n- openstack-tox-py27 http://logs.openstack.org/36/573636/10/gate/openstack-tox-py27/145d4c2/ : SUCCESS in 4m 27s\n- openstack-tox-py36 http://logs.openstack.org/36/573636/10/gate/openstack-tox-py36/8180137/ : SUCCESS in 6m 35s\n- openstack-tox-py37 http://logs.openstack.org/36/573636/10/gate/openstack-tox-py37/43605e0/ : SUCCESS in 10m 44s\n- openstack-tox-docs http://logs.openstack.org/36/573636/10/gate/openstack-tox-docs/93b3ce7/html/ : SUCCESS in 5m 13s\n- build-openstack-releasenotes http://logs.openstack.org/36/573636/10/gate/build-openstack-releasenotes/77ec2da/html/ : SUCCESS in 4m 02s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"a5bc990fdc6f6e8d93f41114fface8d64ae7a274","revisions":{"d780e7d25ba913f6eb59d05526378ebf22049582":{"kind":"REWORK","_number":1,"created":"2018-06-08 09:12:49.000000000","uploader":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"ref":"refs/changes/36/573636/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-testr","ref":"refs/changes/36/573636/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/36/573636/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/36/573636/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/36/573636/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-testr refs/changes/36/573636/1"}}},"commit":{"parents":[{"commit":"c224acc6646d8f9f209e4714e834daf7461215aa","subject":"Merge \"Avoid tox_install.sh for constraints support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/c224acc6646d8f9f209e4714e834daf7461215aa"}]}],"author":{"name":"Masayuki Igawa","email":"masayuki@igawa.io","date":"2018-06-08 09:09:02.000000000","tz":540},"committer":{"name":"Masayuki Igawa","email":"masayuki@igawa.io","date":"2018-06-08 09:09:02.000000000","tz":540},"subject":"Deprecate ostestr command","message":"Deprecate ostestr command\n\nThis commit adds deprecation comments to README and the user\ndoc. We already swiched to use `stestr` command instead of\n`ostestr`. And we haven\u0027t maintained this command recently. So, it\nwould be good to make it clearer for users.\n\nChange-Id: I3a6084db9f86627e3e94abaa4fb4aec52a01126a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/d780e7d25ba913f6eb59d05526378ebf22049582"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/d780e7d25ba913f6eb59d05526378ebf22049582"}]},"branch":"refs/heads/master"},"2e774ae96c652393f0063c61bd34661b1169d555":{"kind":"REWORK","_number":2,"created":"2019-04-19 05:56:11.000000000","uploader":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"ref":"refs/changes/36/573636/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-testr","ref":"refs/changes/36/573636/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/36/573636/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/36/573636/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/36/573636/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-testr refs/changes/36/573636/2"}}},"commit":{"parents":[{"commit":"a36d10a3e3eba34649f4a51100bd0b6f174dc1b0","subject":"add python 3.7 unit test job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/a36d10a3e3eba34649f4a51100bd0b6f174dc1b0"}]}],"author":{"name":"Masayuki Igawa","email":"masayuki@igawa.io","date":"2018-06-08 09:09:02.000000000","tz":540},"committer":{"name":"Masayuki Igawa","email":"masayuki@igawa.io","date":"2019-04-19 05:52:04.000000000","tz":540},"subject":"Deprecate ostestr command","message":"Deprecate ostestr command\n\nThis commit adds deprecation comments to README, the user doc and\nwarning message to in the command. We already switched to use\n`stestr` command instead of `ostestr`. And we haven\u0027t maintained\nthis command recently. So, it would be good to make it clearer for\nusers.\n\nChange-Id: I3a6084db9f86627e3e94abaa4fb4aec52a01126a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/2e774ae96c652393f0063c61bd34661b1169d555"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/2e774ae96c652393f0063c61bd34661b1169d555"}]},"branch":"refs/heads/master"},"6ffd129cfbe5f54ed28bd7132ba2d2b475ab0871":{"kind":"REWORK","_number":3,"created":"2019-04-19 05:58:52.000000000","uploader":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"ref":"refs/changes/36/573636/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-testr","ref":"refs/changes/36/573636/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/36/573636/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/36/573636/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/36/573636/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-testr refs/changes/36/573636/3"}}},"commit":{"parents":[{"commit":"a36d10a3e3eba34649f4a51100bd0b6f174dc1b0","subject":"add python 3.7 unit test job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/a36d10a3e3eba34649f4a51100bd0b6f174dc1b0"}]}],"author":{"name":"Masayuki Igawa","email":"masayuki@igawa.io","date":"2018-06-08 09:09:02.000000000","tz":540},"committer":{"name":"Masayuki Igawa","email":"masayuki@igawa.io","date":"2019-04-19 05:58:13.000000000","tz":540},"subject":"Deprecate ostestr command","message":"Deprecate ostestr command\n\nThis commit adds deprecation comments to README, the user doc and\nwarning message to in the command. We already switched to use\n`stestr` command instead of `ostestr`. And we haven\u0027t maintained\nthis command recently. So, it would be good to make it clearer for\nusers.\n\nChange-Id: I3a6084db9f86627e3e94abaa4fb4aec52a01126a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/6ffd129cfbe5f54ed28bd7132ba2d2b475ab0871"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/6ffd129cfbe5f54ed28bd7132ba2d2b475ab0871"}]},"branch":"refs/heads/master"},"c8f119ef7227bdb1d12091ad8155044e8ea41526":{"kind":"REWORK","_number":4,"created":"2019-04-22 09:02:23.000000000","uploader":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"ref":"refs/changes/36/573636/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-testr","ref":"refs/changes/36/573636/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/36/573636/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/36/573636/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/36/573636/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-testr refs/changes/36/573636/4"}}},"commit":{"parents":[{"commit":"a36d10a3e3eba34649f4a51100bd0b6f174dc1b0","subject":"add python 3.7 unit test job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/a36d10a3e3eba34649f4a51100bd0b6f174dc1b0"}]}],"author":{"name":"Masayuki Igawa","email":"masayuki@igawa.io","date":"2018-06-08 09:09:02.000000000","tz":540},"committer":{"name":"Masayuki Igawa","email":"masayuki@igawa.io","date":"2019-04-22 09:02:21.000000000","tz":540},"subject":"Deprecate ostestr command","message":"Deprecate ostestr command\n\nThis commit adds deprecation comments to README, the user doc and\nwarning message to in the command. We already switched to use\n`stestr` command instead of `ostestr`. And we haven\u0027t maintained\nthis command recently. So, it would be good to make it clearer for\nusers.\n\nChange-Id: I3a6084db9f86627e3e94abaa4fb4aec52a01126a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/c8f119ef7227bdb1d12091ad8155044e8ea41526"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/c8f119ef7227bdb1d12091ad8155044e8ea41526"}]},"branch":"refs/heads/master"},"94cd83eaef7d918dacc37ba05389ee989a93d881":{"kind":"REWORK","_number":5,"created":"2019-04-22 09:04:42.000000000","uploader":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"ref":"refs/changes/36/573636/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-testr","ref":"refs/changes/36/573636/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/36/573636/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/36/573636/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/36/573636/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-testr refs/changes/36/573636/5"}}},"commit":{"parents":[{"commit":"a36d10a3e3eba34649f4a51100bd0b6f174dc1b0","subject":"add python 3.7 unit test job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/a36d10a3e3eba34649f4a51100bd0b6f174dc1b0"}]}],"author":{"name":"Masayuki Igawa","email":"masayuki@igawa.io","date":"2018-06-08 09:09:02.000000000","tz":540},"committer":{"name":"Masayuki Igawa","email":"masayuki@igawa.io","date":"2019-04-22 09:04:33.000000000","tz":540},"subject":"Deprecate ostestr command","message":"Deprecate ostestr command\n\nThis commit adds deprecation comments to README, the user doc and\nwarning message to in the command. We already switched to use\n`stestr` command instead of `ostestr`. And we haven\u0027t maintained\nthis command recently. So, it would be good to make it clearer for\nusers.\n\nChange-Id: I3a6084db9f86627e3e94abaa4fb4aec52a01126a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/94cd83eaef7d918dacc37ba05389ee989a93d881"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/94cd83eaef7d918dacc37ba05389ee989a93d881"}]},"branch":"refs/heads/master"},"611c4d71fc070ccc1cbe3202f0c5392b94144dff":{"kind":"REWORK","_number":6,"created":"2019-05-25 11:34:25.000000000","uploader":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"ref":"refs/changes/36/573636/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-testr","ref":"refs/changes/36/573636/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/36/573636/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/36/573636/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/36/573636/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-testr refs/changes/36/573636/6"}}},"commit":{"parents":[{"commit":"a36d10a3e3eba34649f4a51100bd0b6f174dc1b0","subject":"add python 3.7 unit test job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/a36d10a3e3eba34649f4a51100bd0b6f174dc1b0"}]}],"author":{"name":"Masayuki Igawa","email":"masayuki@igawa.io","date":"2018-06-08 09:09:02.000000000","tz":540},"committer":{"name":"Masayuki Igawa","email":"masayuki@igawa.io","date":"2019-05-25 11:33:57.000000000","tz":120},"subject":"Deprecate ostestr command","message":"Deprecate ostestr command\n\nThis commit adds deprecation comments to README, the user doc and\nwarning message to in the command. We already switched to use\n`stestr` command instead of `ostestr`. And we haven\u0027t maintained\nthis command recently. So, it would be good to make it clearer for\nusers.\n\nChange-Id: I3a6084db9f86627e3e94abaa4fb4aec52a01126a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/611c4d71fc070ccc1cbe3202f0c5392b94144dff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/611c4d71fc070ccc1cbe3202f0c5392b94144dff"}]},"branch":"refs/heads/master"},"bf471e7bfe71f4ae0f34d94a5cd59c0fa9370156":{"kind":"REWORK","_number":7,"created":"2019-05-25 12:15:09.000000000","uploader":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"ref":"refs/changes/36/573636/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-testr","ref":"refs/changes/36/573636/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/36/573636/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/36/573636/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/36/573636/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-testr refs/changes/36/573636/7"}}},"commit":{"parents":[{"commit":"a36d10a3e3eba34649f4a51100bd0b6f174dc1b0","subject":"add python 3.7 unit test job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/a36d10a3e3eba34649f4a51100bd0b6f174dc1b0"}]}],"author":{"name":"Masayuki Igawa","email":"masayuki@igawa.io","date":"2018-06-08 09:09:02.000000000","tz":540},"committer":{"name":"Masayuki Igawa","email":"masayuki@igawa.io","date":"2019-05-25 12:14:57.000000000","tz":120},"subject":"Deprecate ostestr command","message":"Deprecate ostestr command\n\nThis commit adds deprecation comments to README, the user doc and\nwarning message to in the command. We already switched to use\n`stestr` command instead of `ostestr`. And we haven\u0027t maintained\nthis command recently. So, it would be good to make it clearer for\nusers.\n\nChange-Id: I3a6084db9f86627e3e94abaa4fb4aec52a01126a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/bf471e7bfe71f4ae0f34d94a5cd59c0fa9370156"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/bf471e7bfe71f4ae0f34d94a5cd59c0fa9370156"}]},"branch":"refs/heads/master"},"454e86319049093e4a9aa0011930144126e1682a":{"kind":"REWORK","_number":8,"created":"2019-05-25 12:17:28.000000000","uploader":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"ref":"refs/changes/36/573636/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-testr","ref":"refs/changes/36/573636/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/36/573636/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/36/573636/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/36/573636/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-testr refs/changes/36/573636/8"}}},"commit":{"parents":[{"commit":"a36d10a3e3eba34649f4a51100bd0b6f174dc1b0","subject":"add python 3.7 unit test job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/a36d10a3e3eba34649f4a51100bd0b6f174dc1b0"}]}],"author":{"name":"Masayuki Igawa","email":"masayuki@igawa.io","date":"2018-06-08 09:09:02.000000000","tz":540},"committer":{"name":"Masayuki Igawa","email":"masayuki@igawa.io","date":"2019-05-25 12:17:08.000000000","tz":120},"subject":"Deprecate ostestr command","message":"Deprecate ostestr command\n\nThis commit adds deprecation comments to README, the user doc and\nwarning message to in the command. We already switched to use\n`stestr` command instead of `ostestr`. And we haven\u0027t maintained\nthis command recently. So, it would be good to make it clearer for\nusers.\n\nChange-Id: I3a6084db9f86627e3e94abaa4fb4aec52a01126a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/454e86319049093e4a9aa0011930144126e1682a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/454e86319049093e4a9aa0011930144126e1682a"}]},"branch":"refs/heads/master"},"b8269824c3ece1a2b3545e828ee7704d58a8a1fe":{"kind":"REWORK","_number":9,"created":"2019-06-03 08:32:51.000000000","uploader":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"ref":"refs/changes/36/573636/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-testr","ref":"refs/changes/36/573636/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/36/573636/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/36/573636/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/36/573636/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-testr refs/changes/36/573636/9"}}},"commit":{"parents":[{"commit":"a36d10a3e3eba34649f4a51100bd0b6f174dc1b0","subject":"add python 3.7 unit test job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/a36d10a3e3eba34649f4a51100bd0b6f174dc1b0"}]}],"author":{"name":"Masayuki Igawa","email":"masayuki@igawa.io","date":"2018-06-08 09:09:02.000000000","tz":540},"committer":{"name":"Masayuki Igawa","email":"masayuki@igawa.io","date":"2019-06-03 08:32:30.000000000","tz":540},"subject":"Deprecate ostestr command","message":"Deprecate ostestr command\n\nThis commit adds deprecation comments to README, the user doc and\nwarning message to in the command. We already switched to use\n`stestr` command instead of `ostestr`. And we haven\u0027t maintained\nthis command recently. So, it would be good to make it clearer for\nusers.\n\nChange-Id: I3a6084db9f86627e3e94abaa4fb4aec52a01126a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/b8269824c3ece1a2b3545e828ee7704d58a8a1fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/b8269824c3ece1a2b3545e828ee7704d58a8a1fe"}]},"branch":"refs/heads/master"},"a5bc990fdc6f6e8d93f41114fface8d64ae7a274":{"kind":"REWORK","_number":10,"created":"2019-06-03 09:44:33.000000000","uploader":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"ref":"refs/changes/36/573636/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-testr","ref":"refs/changes/36/573636/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/36/573636/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/36/573636/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/36/573636/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-testr refs/changes/36/573636/10"}}},"commit":{"parents":[{"commit":"a36d10a3e3eba34649f4a51100bd0b6f174dc1b0","subject":"add python 3.7 unit test job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/a36d10a3e3eba34649f4a51100bd0b6f174dc1b0"}]}],"author":{"name":"Masayuki Igawa","email":"masayuki@igawa.io","date":"2018-06-08 09:09:02.000000000","tz":540},"committer":{"name":"Masayuki Igawa","email":"masayuki@igawa.io","date":"2019-06-03 09:44:07.000000000","tz":540},"subject":"Deprecate ostestr command","message":"Deprecate ostestr command\n\nThis commit adds deprecation comments to README, the user doc and\nwarning message to in the command. We already switched to use\n`stestr` command instead of `ostestr`. And we haven\u0027t maintained\nthis command recently. So, it would be good to make it clearer for\nusers.\n\nChange-Id: I3a6084db9f86627e3e94abaa4fb4aec52a01126a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/a5bc990fdc6f6e8d93f41114fface8d64ae7a274"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/a5bc990fdc6f6e8d93f41114fface8d64ae7a274"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
