)]}'
{"id":"openstack%2Fos-testr~651275","triplet_id":"openstack%2Fos-testr~master~I70a3ed6c3edaafc8bb6269b1ae07bd46b06f4db4","project":"openstack/os-testr","branch":"master","hashtags":[],"change_id":"I70a3ed6c3edaafc8bb6269b1ae07bd46b06f4db4","subject":"Default concurrency to 0 to avoid TypeError in stestr","status":"MERGED","created":"2019-04-09 15:28:46.000000000","updated":"2019-06-02 13:50:53.000000000","submitted":"2019-06-02 13:50:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"651275-1559483453664-4f1731f4","meta_rev_id":"3294788be05105797f6c0dd26674480848957edf","_number":651275,"virtual_id_number":651275,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},{"value":0,"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},{"value":0,"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":2,"date":"2019-06-02 13:50:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-05-04 14:46:08.000000000","_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":0,"_account_id":17880,"name":"Shachar Snapiri","email":"shachar.snapiri@toganetworks.com","username":"snapiri"},{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":4190,"name":"lifeless","email":"robertc@robertcollins.net","username":"lifeless"},{"value":0,"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"recommended":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2019-04-15 15:49:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},{"value":0,"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},{"value":0,"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},{"value":2,"date":"2019-06-02 13:42:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"value":1,"date":"2019-05-06 10:00:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2019-05-29 21:01:05.000000000","_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":1,"date":"2019-05-12 18:07:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17880,"name":"Shachar Snapiri","email":"shachar.snapiri@toganetworks.com","username":"snapiri"},{"value":1,"date":"2019-05-25 11:29:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":1,"date":"2019-05-06 10:30:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":4190,"name":"lifeless","email":"robertc@robertcollins.net","username":"lifeless"},{"value":0,"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2019-05-24 20:10:36.000000000","_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},{"value":0,"date":"2019-05-24 20:10:36.000000000","_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},{"value":0,"date":"2019-05-24 20:10:36.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},{"value":1,"date":"2019-06-02 13:42:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"value":0,"date":"2019-05-24 20:10:36.000000000","_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":0,"_account_id":17880,"name":"Shachar Snapiri","email":"shachar.snapiri@toganetworks.com","username":"snapiri"},{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"date":"2019-05-24 20:10:36.000000000","_account_id":4190,"name":"lifeless","email":"robertc@robertcollins.net","username":"lifeless"},{"value":0,"date":"2019-05-24 20:10:36.000000000","_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":4190,"name":"lifeless","email":"robertc@robertcollins.net","username":"lifeless"},{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":17880,"name":"Shachar Snapiri","email":"shachar.snapiri@toganetworks.com","username":"snapiri"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-15 15:49:08.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-05-06 10:00:50.000000000","updated_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"},{"updated":"2019-05-06 10:30:16.000000000","updated_by":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"reviewer":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"state":"REVIEWER"},{"updated":"2019-05-12 18:07:32.000000000","updated_by":{"_account_id":17880,"name":"Shachar Snapiri","email":"shachar.snapiri@toganetworks.com","username":"snapiri"},"reviewer":{"_account_id":17880,"name":"Shachar Snapiri","email":"shachar.snapiri@toganetworks.com","username":"snapiri"},"state":"REVIEWER"},{"updated":"2019-05-24 20:10:36.000000000","updated_by":{"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},"reviewer":{"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},"state":"REVIEWER"},{"updated":"2019-05-24 20:10:36.000000000","updated_by":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"reviewer":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"state":"REVIEWER"},{"updated":"2019-05-24 20:10:36.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2019-05-24 20:10:36.000000000","updated_by":{"_account_id":4190,"name":"lifeless","email":"robertc@robertcollins.net","username":"lifeless"},"reviewer":{"_account_id":4190,"name":"lifeless","email":"robertc@robertcollins.net","username":"lifeless"},"state":"REVIEWER"},{"updated":"2019-05-24 20:10:36.000000000","updated_by":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"reviewer":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"state":"REVIEWER"},{"updated":"2019-05-24 20:10:36.000000000","updated_by":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"reviewer":{"_account_id":5638,"name":"Davanum Srinivas","email":"davanum@gmail.com","username":"dims-v"},"state":"REVIEWER"},{"updated":"2019-05-25 11:29:17.000000000","updated_by":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"reviewer":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"state":"REVIEWER"},{"updated":"2019-05-29 21:01:05.000000000","updated_by":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"reviewer":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"state":"REVIEWER"},{"updated":"2019-06-02 13:42:39.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2019-06-02 13:50:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"072a9a7fe9c22b23e6d7407510d5a45dea67fd67","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-04-09 15:28:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b9e98cead1ef6b923d7532ff5e93b251d7aacb5c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-09 15:46:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/75/651275/1/check/openstack-tox-pep8/897f5dd/ : SUCCESS in 6m 10s\n- openstack-tox-py27 http://logs.openstack.org/75/651275/1/check/openstack-tox-py27/79545a2/ : SUCCESS in 7m 03s\n- openstack-tox-py35 http://logs.openstack.org/75/651275/1/check/openstack-tox-py35/3361550/ : SUCCESS in 6m 32s\n- openstack-tox-py36 http://logs.openstack.org/75/651275/1/check/openstack-tox-py36/ea3d229/ : SUCCESS in 5m 01s\n- openstack-tox-py37 http://logs.openstack.org/75/651275/1/check/openstack-tox-py37/8ee4bec/ : SUCCESS in 7m 41s\n- openstack-tox-docs http://logs.openstack.org/75/651275/1/check/openstack-tox-docs/752cd72/html/ : SUCCESS in 5m 52s","accounts_in_message":[],"_revision_number":1},{"id":"f5d788497a1dfc112bd0d0da1cd29fe5384e8801","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-04-15 15:49:08.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"a715d27530b180d76bfc52df817a7f4df09ab717","author":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"date":"2019-04-18 08:03:23.000000000","message":"Patch Set 1: Code-Review+1\n\nyes, it should be set.","accounts_in_message":[],"_revision_number":1},{"id":"18b4f2244c717dd2d1cdf127912daef5dba46871","author":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"date":"2019-04-18 08:30:47.000000000","message":"Patch Set 1: -Code-Review\n\nhmm, I have mixed feeling here. We encourage to use stestr directly instead of ostestr command now..","accounts_in_message":[],"_revision_number":1},{"id":"1ce22368c4e2fbe54f784c72823711ee4bb5bb4e","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-05-04 14:46:08.000000000","message":"Patch Set 1:\n\nI feel like there\u0027s a pretty big difference between saying \"we recommend you use another tool\" and \"this tool is unmaintained; if you run into issues, you *must* use another tool\".\n\nFWIW, requirements recently updated stestr to 2.3.1, which broke all of swift\u0027s func tests. We\u0027ve got https://review.opendev.org/#/c/651277/ working its way through the gate now to allow 9 other already-approved patches to land; I\u0027ll look at what needs to change to use stestr directly instead next.\n\nEven then, though, having this bad interaction negatively impacts my ability to\n\n- bring up swift services from master,\n- have a separate checkout from an older tag, and\n- run the old functional tests against the newer service\n\nas the old version will still want to invoke ostestr with --serial.","accounts_in_message":[],"_revision_number":1},{"id":"34efbecfa2be3d7484836b5a9f3de067f3a72f90","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2019-05-06 10:00:50.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"5cbbf49f9fbeccfec932c1f208bc2dae3011db98","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2019-05-06 10:07:42.000000000","message":"Patch Set 1:\n\n\u003e hmm, I have mixed feeling here. We encourage to use stestr directly\n \u003e instead of ostestr command now..\n\nDo you have any links where this is recommended? I found only Matthew\u0027s message on the ML that states \"Moving forward I\u0027d like to get everything switched over to using stestr, either\ndirectly or indirectly via ostestr.\".\n\nIf it\u0027s true to call stestr directly, it might be good to issue a warning message when running ostestr to increase visibility of the transition.","accounts_in_message":[],"_revision_number":1},{"id":"8a4230b5e21b09a45b9ac74f249a64d6ee2f7587","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2019-05-06 10:30:16.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"1ee5244fd5f60343767d7ae83f7610b99a637f4e","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2019-05-06 10:32:02.000000000","message":"Patch Set 1:\n\n+1 to Jakub\u0027s suggestion of logging a Warning","accounts_in_message":[],"_revision_number":1},{"id":"f3573c05b75fc6384b4eb1206b16baab9ac5340a","author":{"_account_id":17880,"name":"Shachar Snapiri","email":"shachar.snapiri@toganetworks.com","username":"snapiri"},"date":"2019-05-12 18:07:32.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"ef648de71ff95097111b112b5003703bebddae01","author":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"date":"2019-05-25 11:29:17.000000000","message":"Patch Set 1: Code-Review+1\n\nThanks for your comments and sorry for delayed reply. Yeah, I think it\u0027s better to be fixed in os-testr because some unit tests fail because of this issue.\n\nI think/hope a newer version(2.4.0) stestr will be released soon but it will take some time.","accounts_in_message":[],"_revision_number":1},{"id":"6af66931268ae8440cade0da0e08a03958c9a95f","author":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"date":"2019-05-25 11:29:29.000000000","message":"Patch Set 1:\n\nrecheck\n\njust in case","accounts_in_message":[],"_revision_number":1},{"id":"127986af20b88b9f989fdb92865a00dcb7c2c810","author":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"date":"2019-05-25 11:37:08.000000000","message":"Patch Set 1:\n\n\u003e \u003e hmm, I have mixed feeling here. We encourage to use stestr\n \u003e directly\n \u003e \u003e instead of ostestr command now..\n \u003e \n \u003e Do you have any links where this is recommended? I found only\n \u003e Matthew\u0027s message on the ML that states \"Moving forward I\u0027d like to\n \u003e get everything switched over to using stestr, either\n \u003e directly or indirectly via ostestr.\".\n \u003e \n \u003e If it\u0027s true to call stestr directly, it might be good to issue a\n \u003e warning message when running ostestr to increase visibility of the\n \u003e transition.\n\nyeah, true. We actually didn\u0027t advertise it explicitly. So, that\u0027s why I pushed this patch https://review.opendev.org/#/c/573636/","accounts_in_message":[],"_revision_number":1},{"id":"b2d0831525902e9da4647db2a0cc79c7c2f09c49","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-25 11:48:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/75/651275/1/check/openstack-tox-pep8/ae28bcb/ : SUCCESS in 5m 39s\n- openstack-tox-py27 http://logs.openstack.org/75/651275/1/check/openstack-tox-py27/9ac5e74/ : SUCCESS in 6m 20s\n- openstack-tox-py35 http://logs.openstack.org/75/651275/1/check/openstack-tox-py35/854a74b/ : SUCCESS in 5m 02s\n- openstack-tox-py36 http://logs.openstack.org/75/651275/1/check/openstack-tox-py36/ff15040/ : SUCCESS in 6m 19s\n- openstack-tox-py37 http://logs.openstack.org/75/651275/1/check/openstack-tox-py37/d38401f/ : SUCCESS in 6m 22s\n- openstack-tox-docs http://logs.openstack.org/75/651275/1/check/openstack-tox-docs/ad52cce/html/ : SUCCESS in 6m 08s","accounts_in_message":[],"_revision_number":1},{"id":"f852dfd104eaf43f184f49c56edfee78ae134635","author":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"date":"2019-05-29 21:01:05.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"a565cd7ba4b5541988f6da4ce2c4fe80c235227f","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-06-02 13:42:24.000000000","message":"Patch Set 1:\n\n\u003e \u003e \u003e hmm, I have mixed feeling here. We encourage to use stestr\n \u003e \u003e directly\n \u003e \u003e \u003e instead of ostestr command now..\n \u003e \u003e\n \u003e \u003e Do you have any links where this is recommended? I found only\n \u003e \u003e Matthew\u0027s message on the ML that states \"Moving forward I\u0027d like\n \u003e to\n \u003e \u003e get everything switched over to using stestr, either\n \u003e \u003e directly or indirectly via ostestr.\".\n \u003e \u003e\n \u003e \u003e If it\u0027s true to call stestr directly, it might be good to issue a\n \u003e \u003e warning message when running ostestr to increase visibility of\n \u003e the\n \u003e \u003e transition.\n \u003e \n \u003e yeah, true. We actually didn\u0027t advertise it explicitly. So, that\u0027s\n \u003e why I pushed this patch https://review.opendev.org/#/c/573636/\n\nYeah, we have not released it or marked as deprecated yet. Let\u0027s get this fix in and then we will mark ostestr deprecated with new version release.","accounts_in_message":[],"_revision_number":1},{"id":"01294c4d9785fffee5c5720c2a37cae2d1fe5a39","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-06-02 13:42:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7011a5e8846e7ff8e3b0dc89ce42a3d966db42ce","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-02 13:42:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0493b0f874ce623bfe672f5f02a1a226345175f1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-02 13:50:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"3294788be05105797f6c0dd26674480848957edf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-02 13:50:53.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/75/651275/1/gate/openstack-tox-pep8/cf301b0/ : SUCCESS in 5m 42s\n- openstack-tox-py27 http://logs.openstack.org/75/651275/1/gate/openstack-tox-py27/268d78d/ : SUCCESS in 5m 57s\n- openstack-tox-py35 http://logs.openstack.org/75/651275/1/gate/openstack-tox-py35/a41a91a/ : SUCCESS in 6m 30s\n- openstack-tox-py36 http://logs.openstack.org/75/651275/1/gate/openstack-tox-py36/bfc6ac7/ : SUCCESS in 5m 59s\n- openstack-tox-py37 http://logs.openstack.org/75/651275/1/gate/openstack-tox-py37/421e079/ : SUCCESS in 5m 58s\n- openstack-tox-docs http://logs.openstack.org/75/651275/1/gate/openstack-tox-docs/5be3838/html/ : SUCCESS in 5m 33s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"595c06c5bc662ec945e728933df6ac5d6ab116c6","revisions":{"595c06c5bc662ec945e728933df6ac5d6ab116c6":{"kind":"REWORK","_number":1,"created":"2019-04-09 15:28:46.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/75/651275/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-testr","ref":"refs/changes/75/651275/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/75/651275/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/75/651275/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/75/651275/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-testr refs/changes/75/651275/1"}}},"commit":{"parents":[{"commit":"a36d10a3e3eba34649f4a51100bd0b6f174dc1b0","subject":"add python 3.7 unit test job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/a36d10a3e3eba34649f4a51100bd0b6f174dc1b0"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-04-09 15:26:16.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-04-09 15:28:45.000000000","tz":-420},"subject":"Default concurrency to 0 to avoid TypeError in stestr","message":"Default concurrency to 0 to avoid TypeError in stestr\n\nThe traceback would look like\n\nTraceback (most recent call last):\n  File \".../bin/ostestr\", line 10, in \u003cmodule\u003e\n    sys.exit(main())\n  File \".../os_testr/ostestr.py\", line 263, in main\n    exit(ostestr(sys.argv[1:]))\n  File \".../os_testr/ostestr.py\", line 259, in ostestr\n    return _select_and_call_runner(opts, regex, others)\n  File \".../os_testr/ostestr.py\", line 212, in _select_and_call_runner\n    black_regex\u003dopts.black_regex)\n  File \".../os_testr/ostestr.py\", line 167, in call_testr\n    black_regex\u003dblack_regex)\n  File \".../stestr/commands/run.py\", line 334, in run_command\n    concurrency \u003d _to_int(concurrency)\n  File \".../stestr/commands/run.py\", line 39, in _to_int\n    i \u003d int(possible)\nTypeError: int() argument must be a string or a number, not \u0027NoneType\u0027\n\nNew enough stestr will tolerate it, but still write a warning to stderr.\nSee https://github.com/mtreinish/stestr/commit/ac8dac0\n\nChange-Id: I70a3ed6c3edaafc8bb6269b1ae07bd46b06f4db4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/595c06c5bc662ec945e728933df6ac5d6ab116c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/595c06c5bc662ec945e728933df6ac5d6ab116c6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
