)]}'
{"id":"openstack%2Fos-testr~654290","triplet_id":"openstack%2Fos-testr~master~I05cd5327a5ab883701fa183a1b4264f91219147c","project":"openstack/os-testr","branch":"master","topic":"dropping-py35-testing","hashtags":[],"change_id":"I05cd5327a5ab883701fa183a1b4264f91219147c","subject":"Dropping the py35 testing","status":"MERGED","created":"2019-04-22 03:55:02.000000000","updated":"2019-06-02 15:04:45.000000000","submitted":"2019-06-02 15:04:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"654290-1559487885878-fc91e94c","meta_rev_id":"b81a71bb28f274676ab221a3f0edb28e7888ede8","_number":654290,"virtual_id_number":654290,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":2,"date":"2019-06-02 15:04:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"recommended":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"all":[{"value":1,"date":"2019-04-22 08:09:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":2,"date":"2019-06-02 14:54:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":1,"date":"2019-06-02 14:54:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-22 08:09:31.000000000","updated_by":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"reviewer":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"state":"REVIEWER"},{"updated":"2019-06-02 15:04:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0e53708e4da5c0a8650613295aa6d22c177ea9cc","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-04-22 03:55:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e9f77c10fd61ec62a5566ee9abd92670f890921f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 04:04:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/90/654290/1/check/test-release-openstack/10201e6/ : SUCCESS in 3m 22s\n- openstack-tox-pep8 http://logs.openstack.org/90/654290/1/check/openstack-tox-pep8/21a1e7a/ : SUCCESS in 3m 59s\n- openstack-tox-py27 http://logs.openstack.org/90/654290/1/check/openstack-tox-py27/1c55f20/ : SUCCESS in 4m 34s\n- openstack-tox-py36 http://logs.openstack.org/90/654290/1/check/openstack-tox-py36/5f871b2/ : SUCCESS in 4m 17s\n- openstack-tox-py37 http://logs.openstack.org/90/654290/1/check/openstack-tox-py37/3ee774f/ : SUCCESS in 4m 23s\n- openstack-tox-docs http://logs.openstack.org/90/654290/1/check/openstack-tox-docs/6cd0679/html/ : SUCCESS in 3m 58s","accounts_in_message":[],"_revision_number":1},{"id":"ebe87285fee495c29464e4eb015b64188daf63e0","author":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"date":"2019-04-22 08:09:31.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"fe76d0a2465bcbbb9e0d0a9e62537af0c5209791","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-06-02 14:54:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6190c99870ae045fef3233287b4227457877c06f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-02 14:55:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9fe8325639b6db4547ef4651f2d65fdafb4ee405","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-02 15:04:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b7c13d3d48f2d4622d9b8916a4e3bc2fa4621d3a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-02 15:04:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack http://logs.openstack.org/90/654290/1/gate/test-release-openstack/807df1b/ : SUCCESS in 3m 02s\n- openstack-tox-pep8 http://logs.openstack.org/90/654290/1/gate/openstack-tox-pep8/d128ccd/ : SUCCESS in 4m 24s\n- openstack-tox-py27 http://logs.openstack.org/90/654290/1/gate/openstack-tox-py27/7cceefc/ : SUCCESS in 4m 20s\n- openstack-tox-py36 http://logs.openstack.org/90/654290/1/gate/openstack-tox-py36/a0c67d8/ : SUCCESS in 3m 57s\n- openstack-tox-py37 http://logs.openstack.org/90/654290/1/gate/openstack-tox-py37/337a428/ : SUCCESS in 4m 10s\n- openstack-tox-docs http://logs.openstack.org/90/654290/1/gate/openstack-tox-docs/8c8d077/html/ : SUCCESS in 3m 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"77002ea67d70873a93d9b1bc1301bcc4f7bbb736","revisions":{"77002ea67d70873a93d9b1bc1301bcc4f7bbb736":{"kind":"REWORK","_number":1,"created":"2019-04-22 03:55:02.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/90/654290/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-testr","ref":"refs/changes/90/654290/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/90/654290/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/90/654290/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/90/654290/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-testr refs/changes/90/654290/1"}}},"commit":{"parents":[{"commit":"5674bb0f7c1f0869a9996ff6f3e9b3cc1aff96e8","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/5674bb0f7c1f0869a9996ff6f3e9b3cc1aff96e8"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-04-22 03:54:37.000000000","tz":0},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2019-04-22 03:55:00.000000000","tz":0},"subject":"Dropping the py35 testing","message":"Dropping the py35 testing\n\nAll the integration testing has been moved to\nBionic now[1] and py3.5 is not tested runtime for\nTrain or stable/stein[2].\n\nAs per below ML thread, we are good to drop the py35\ntesting now:\nhttp://lists.openstack.org/pipermail/openstack-discuss/2019-April/005097.html\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-April/004647.html\n[2]\nhttps://governance.openstack.org/tc/reference/runtimes/stein.html\nhttps://governance.openstack.org/tc/reference/runtimes/train.html\n\nChange-Id: I05cd5327a5ab883701fa183a1b4264f91219147c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/77002ea67d70873a93d9b1bc1301bcc4f7bbb736"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/77002ea67d70873a93d9b1bc1301bcc4f7bbb736"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
