)]}'
{"id":"openstack%2Fos-testr~655062","triplet_id":"openstack%2Fos-testr~master~Ifcb01ece24fb28d4dc28daa0a638787280fe8eb5","project":"openstack/os-testr","branch":"master","topic":"bug/1826529","hashtags":[],"change_id":"Ifcb01ece24fb28d4dc28daa0a638787280fe8eb5","subject":"Replace git.openstack.org URLs with opendev.org URLs","status":"MERGED","created":"2019-04-23 11:39:54.000000000","updated":"2019-06-30 05:33:35.000000000","submitted":"2019-06-30 05:33:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"655062-1561872815459-a45a0009","meta_rev_id":"f84aa7f269dabb399b70b38829c4ebb774fac4b7","_number":655062,"virtual_id_number":655062,"owner":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":2,"date":"2019-06-30 05:33:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":30413,"name":"chaiguanglong","email":"chaigl@awcloud.com","username":"chaiguanglong"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"recommended":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"all":[{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":1,"date":"2019-06-25 09:33:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":1,"date":"2019-06-25 09:33:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":1,"date":"2019-06-25 09:33:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-06-30 05:29:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":1,"date":"2019-06-25 09:33:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30413,"name":"chaiguanglong","email":"chaigl@awcloud.com","username":"chaiguanglong"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-30 05:29:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":30413,"name":"chaiguanglong","email":"chaigl@awcloud.com","username":"chaiguanglong"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"_account_id":30413,"name":"chaiguanglong","email":"chaigl@awcloud.com","username":"chaiguanglong"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-30 03:07:26.000000000","updated_by":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"reviewer":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"state":"REVIEWER"},{"updated":"2019-05-03 03:23:54.000000000","updated_by":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"reviewer":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"state":"REVIEWER"},{"updated":"2019-06-05 06:44:07.000000000","updated_by":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"reviewer":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"state":"REVIEWER"},{"updated":"2019-06-05 06:46:32.000000000","updated_by":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"reviewer":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"state":"REVIEWER"},{"updated":"2019-06-05 10:42:01.000000000","updated_by":{"_account_id":30413,"name":"chaiguanglong","email":"chaigl@awcloud.com","username":"chaiguanglong"},"reviewer":{"_account_id":30413,"name":"chaiguanglong","email":"chaigl@awcloud.com","username":"chaiguanglong"},"state":"REVIEWER"},{"updated":"2019-06-09 16:36:14.000000000","updated_by":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"reviewer":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"state":"REVIEWER"},{"updated":"2019-06-30 05:29:25.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2019-06-30 05:33:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0ed5899e0b315fd41007053d0277c94ee0765024","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-04-23 11:39:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"92708a4c471408e4bc275956cc5266e78e254b6a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 12:14:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/62/655062/1/check/test-release-openstack/f4047c6/ : SUCCESS in 3m 30s\n- openstack-tox-pep8 http://logs.openstack.org/62/655062/1/check/openstack-tox-pep8/66b8853/ : SUCCESS in 4m 20s\n- openstack-tox-py27 http://logs.openstack.org/62/655062/1/check/openstack-tox-py27/76e7aae/ : SUCCESS in 4m 24s\n- openstack-tox-py35 http://logs.openstack.org/62/655062/1/check/openstack-tox-py35/2a279e5/ : SUCCESS in 4m 27s\n- openstack-tox-py36 http://logs.openstack.org/62/655062/1/check/openstack-tox-py36/f7fb2c0/ : SUCCESS in 4m 33s\n- openstack-tox-py37 http://logs.openstack.org/62/655062/1/check/openstack-tox-py37/4156705/ : SUCCESS in 5m 55s\n- openstack-tox-docs http://logs.openstack.org/62/655062/1/check/openstack-tox-docs/d76a25c/html/ : SUCCESS in 6m 22s","accounts_in_message":[],"_revision_number":1},{"id":"d3934caa9db02ff58327844895b78bb33e966dad","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-04-24 03:06:31.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"13b32cab9f66ee4879b33933699905ef486a74c1","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-04-26 10:11:53.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"304b84224340c3ff3c7c279f9bfa4fa8dd7b22d9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 10:21:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/62/655062/2/check/test-release-openstack/b5bf6de/ : SUCCESS in 2m 56s\n- openstack-tox-pep8 http://logs.openstack.org/62/655062/2/check/openstack-tox-pep8/5edc179/ : SUCCESS in 4m 32s\n- openstack-tox-py27 http://logs.openstack.org/62/655062/2/check/openstack-tox-py27/503dba3/ : SUCCESS in 4m 32s\n- openstack-tox-py35 http://logs.openstack.org/62/655062/2/check/openstack-tox-py35/64093be/ : SUCCESS in 4m 34s\n- openstack-tox-py36 http://logs.openstack.org/62/655062/2/check/openstack-tox-py36/071188a/ : SUCCESS in 6m 55s\n- openstack-tox-py37 http://logs.openstack.org/62/655062/2/check/openstack-tox-py37/c2046c8/ : SUCCESS in 4m 18s\n- openstack-tox-docs http://logs.openstack.org/62/655062/2/check/openstack-tox-docs/8d0c2d5/html/ : SUCCESS in 3m 57s","accounts_in_message":[],"_revision_number":2},{"id":"88e0b1c039dc5c1f4ed48a89d8c24fe74de55341","author":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"date":"2019-04-30 03:07:26.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"71197b067e81c4c755aac266aa61555caf804817","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2019-04-30 11:30:31.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"3aaa555638e6727b2ba64d18a84fcb2d6f0bfa7f","author":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"date":"2019-05-03 03:23:54.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"a7fa68e787f87f46fffaf4b5280f8eeab814ac84","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2019-05-23 09:00:22.000000000","message":"Removed Code-Review+1 by jacky06 \u003czhang.min@99cloud.net\u003e\n","accounts_in_message":[],"_revision_number":2},{"id":"6ffe74a9617b99b73df0715503cf575597b6d4f6","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2019-05-23 09:00:26.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"8b5912631fcd231e44647788140223cf6908537b","author":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"date":"2019-06-04 06:01:21.000000000","message":"Patch Set 2: Code-Review+1\n\nLGTM. I grepped the source code with \u0027openstack.org\u0027 and the other openstack.org domains are for the documentation sites and mailing list.\n\n\n$ ag openstack.org\nHACKING.rst\n4:Read the OpenStack Style Commandments https://docs.openstack.org/hacking/latest/\n\ndoc/source/index.rst\n15:* Documentation: https://docs.openstack.org/os-testr/\n16:* Source: https://git.openstack.org/cgit/openstack/os-testr\n\nCONTRIBUTING.rst\n4:- http://docs.openstack.org/infra/manual/developers.html\n10:- http://docs.openstack.org/infra/manual/developers.html#development-workflow\n\nos_testr/tests/files/setup.cfg\n6:author-email \u003d openstack-discuss@lists.openstack.org\n7:home-page \u003d http://www.openstack.org/\n\nREADME.rst\n16:* Documentation: https://docs.openstack.org/os-testr/\n17:* Source: https://git.openstack.org/cgit/openstack/os-testr\n\ntox.ini\n16:       -c{env:UPPER_CONSTRAINTS_FILE:https://git.openstack.org/cgit/openstack/requirements/plain/upper-constraints.txt}\n\nsetup.cfg\n7:author-email \u003d openstack-discuss@lists.openstack.org\n8:home-page \u003d  https://docs.openstack.org/os-testr/latest/","accounts_in_message":[],"_revision_number":2},{"id":"ade41439686fd7492cf5c911689d8c828ae93ce1","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2019-06-05 02:06:19.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"d8ea4376e9f7625bed8773346ba56ed6e01bcf59","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-06-05 02:18:14.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"4fd7f356d273de62f4c137abc7d4395ac9a786a6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 02:28:53.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/62/655062/4/check/test-release-openstack/4f48927/ : SUCCESS in 2m 59s\n- openstack-tox-pep8 http://logs.openstack.org/62/655062/4/check/openstack-tox-pep8/779f98e/ : SUCCESS in 4m 04s\n- openstack-tox-py27 http://logs.openstack.org/62/655062/4/check/openstack-tox-py27/fdbbb3a/ : SUCCESS in 3m 44s\n- openstack-tox-py36 http://logs.openstack.org/62/655062/4/check/openstack-tox-py36/c224c6a/ : SUCCESS in 4m 07s\n- openstack-tox-py37 http://logs.openstack.org/62/655062/4/check/openstack-tox-py37/9f5e441/ : SUCCESS in 4m 41s\n- openstack-tox-docs http://logs.openstack.org/62/655062/4/check/openstack-tox-docs/a214981/html/ : SUCCESS in 4m 04s","accounts_in_message":[],"_revision_number":4},{"id":"106e847e8f17396bb6c04dbe8f1ce2fde123f7b7","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2019-06-05 02:31:35.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"74dbb8dd9af6e34d7bc5f1e16928c6deb8a27d18","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-06-05 04:29:40.000000000","message":"Patch Set 4: Code-Review+2\n\nthanks","accounts_in_message":[],"_revision_number":4},{"id":"a50e2ffa7dafd27ff36f1d14d786e9bbaf50bb5e","author":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"date":"2019-06-05 05:15:45.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)\n\nOverall, LGTM. But one comment for the upper constraints URL.","accounts_in_message":[],"_revision_number":4},{"id":"31be7d34c009f72cc95026e80574534e402ce4fc","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2019-06-05 06:43:01.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"71734763f5bd15e3218dcaa04071b23bceba6d90","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2019-06-05 06:43:34.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"f9a9e3f440af2138f3df7eb69ecd3b31437ab7b9","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2019-06-05 06:44:07.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"d3917865c225fc32f5bc488037af928fd3f9b725","author":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"date":"2019-06-05 06:46:32.000000000","message":"Patch Set 6: Code-Review+1\n\nThanks. LGTM","accounts_in_message":[],"_revision_number":6},{"id":"b40441b228a78f8035ddab25bf8b76c8d734c666","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 06:53:15.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/62/655062/6/check/test-release-openstack/8cda03d/ : SUCCESS in 4m 08s\n- openstack-tox-pep8 http://logs.openstack.org/62/655062/6/check/openstack-tox-pep8/ae335ed/ : SUCCESS in 3m 41s\n- openstack-tox-py27 http://logs.openstack.org/62/655062/6/check/openstack-tox-py27/6ded8c1/ : SUCCESS in 4m 12s\n- openstack-tox-py36 http://logs.openstack.org/62/655062/6/check/openstack-tox-py36/bb980be/ : SUCCESS in 5m 16s\n- openstack-tox-py37 http://logs.openstack.org/62/655062/6/check/openstack-tox-py37/1b7f9cc/ : SUCCESS in 4m 24s\n- openstack-tox-docs http://logs.openstack.org/62/655062/6/check/openstack-tox-docs/6ab5e65/html/ : SUCCESS in 3m 48s","accounts_in_message":[],"_revision_number":6},{"id":"eea3260d78119aa589c08da94cd8da5a6ed4c446","author":{"_account_id":30413,"name":"chaiguanglong","email":"chaigl@awcloud.com","username":"chaiguanglong"},"date":"2019-06-05 10:42:01.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"350040c5d2fafc8bc90f30182f579708b1f5f0d5","author":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"date":"2019-06-09 16:36:14.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"46245e15f38c175c55c2840e6f0c04239ad0cea5","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2019-06-25 09:33:30.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"465ac002911d24e47d1dcd43b233271701afab43","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-25 09:56:37.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/62/655062/7/check/test-release-openstack/8d06db1/ : SUCCESS in 3m 18s\n- openstack-tox-pep8 http://logs.openstack.org/62/655062/7/check/openstack-tox-pep8/18ff312/ : SUCCESS in 2m 50s\n- openstack-tox-py27 http://logs.openstack.org/62/655062/7/check/openstack-tox-py27/b4ceab3/ : SUCCESS in 2m 26s\n- openstack-tox-py36 http://logs.openstack.org/62/655062/7/check/openstack-tox-py36/be498b0/ : SUCCESS in 3m 45s\n- openstack-tox-py37 http://logs.openstack.org/62/655062/7/check/openstack-tox-py37/6bd53df/ : SUCCESS in 3m 11s\n- openstack-tox-docs http://logs.openstack.org/62/655062/7/check/openstack-tox-docs/dba3cd6/html/ : SUCCESS in 2m 37s","accounts_in_message":[],"_revision_number":7},{"id":"5b76d63bd91853e28bb7a3f712621ce452a6fe3f","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-06-30 05:29:25.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"e6ed0c317de77bd61ab4ea89332daef08f4e8128","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-30 05:29:36.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"b33526ef1bb6be5929d08ce1fa9b2c1d406a0c40","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-30 05:33:35.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack http://logs.openstack.org/62/655062/7/gate/test-release-openstack/2137842/ : SUCCESS in 1m 58s\n- openstack-tox-pep8 http://logs.openstack.org/62/655062/7/gate/openstack-tox-pep8/e0c1068/ : SUCCESS in 2m 34s\n- openstack-tox-py27 http://logs.openstack.org/62/655062/7/gate/openstack-tox-py27/deb9d4f/ : SUCCESS in 2m 42s\n- openstack-tox-py36 http://logs.openstack.org/62/655062/7/gate/openstack-tox-py36/9a65b00/ : SUCCESS in 2m 37s\n- openstack-tox-py37 http://logs.openstack.org/62/655062/7/gate/openstack-tox-py37/77f791e/ : SUCCESS in 3m 01s\n- openstack-tox-docs http://logs.openstack.org/62/655062/7/gate/openstack-tox-docs/cc2b3d1/html/ : SUCCESS in 2m 23s","accounts_in_message":[],"_revision_number":7},{"id":"7ef8e6f8efe687498b889d98b2d79fddb56e8c3a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-30 05:33:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"ad00e6d41b441e10f0b4ed81511e66660a9eeaa0","revisions":{"35db7dcfb4661563a2ef04718e7b725f6f03af3f":{"kind":"REWORK","_number":1,"created":"2019-04-23 11:39:54.000000000","uploader":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"ref":"refs/changes/62/655062/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-testr","ref":"refs/changes/62/655062/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/62/655062/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/62/655062/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/62/655062/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-testr refs/changes/62/655062/1"}}},"commit":{"parents":[{"commit":"5674bb0f7c1f0869a9996ff6f3e9b3cc1aff96e8","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/5674bb0f7c1f0869a9996ff6f3e9b3cc1aff96e8"}]}],"author":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-04-23 11:39:37.000000000","tz":480},"committer":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-04-23 11:39:37.000000000","tz":480},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nChange-Id: Ifcb01ece24fb28d4dc28daa0a638787280fe8eb5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/35db7dcfb4661563a2ef04718e7b725f6f03af3f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/35db7dcfb4661563a2ef04718e7b725f6f03af3f"}]},"branch":"refs/heads/master"},"fdf3ab79df3074b4502c8c0860f3748accc60bbb":{"kind":"REWORK","_number":2,"created":"2019-04-26 10:11:53.000000000","uploader":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"ref":"refs/changes/62/655062/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-testr","ref":"refs/changes/62/655062/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/62/655062/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/62/655062/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/62/655062/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-testr refs/changes/62/655062/2"}}},"commit":{"parents":[{"commit":"5674bb0f7c1f0869a9996ff6f3e9b3cc1aff96e8","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/5674bb0f7c1f0869a9996ff6f3e9b3cc1aff96e8"}]}],"author":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-04-23 11:39:37.000000000","tz":480},"committer":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-04-26 10:11:49.000000000","tz":0},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nChange-Id: Ifcb01ece24fb28d4dc28daa0a638787280fe8eb5\nCloses-Bug: #1826529\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/fdf3ab79df3074b4502c8c0860f3748accc60bbb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/fdf3ab79df3074b4502c8c0860f3748accc60bbb"}]},"branch":"refs/heads/master"},"73b76f3459905e82e06f047c3263175c79f6e7fb":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-06-05 02:06:19.000000000","uploader":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"ref":"refs/changes/62/655062/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-testr","ref":"refs/changes/62/655062/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/62/655062/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/62/655062/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/62/655062/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-testr refs/changes/62/655062/3"}}},"commit":{"parents":[{"commit":"f16df2636cd1c0ee3e3c62ecf2698ab7de781b8d","subject":"Merge \"Fix warning message with double \"to\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/f16df2636cd1c0ee3e3c62ecf2698ab7de781b8d"}]}],"author":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-04-23 11:39:37.000000000","tz":480},"committer":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2019-06-05 02:06:19.000000000","tz":0},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nChange-Id: Ifcb01ece24fb28d4dc28daa0a638787280fe8eb5\nCloses-Bug: #1826529\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/73b76f3459905e82e06f047c3263175c79f6e7fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/73b76f3459905e82e06f047c3263175c79f6e7fb"}]},"branch":"refs/heads/master"},"6d4aecf256f9a02131582928480f9fc99c51b6a1":{"kind":"REWORK","_number":4,"created":"2019-06-05 02:18:14.000000000","uploader":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"ref":"refs/changes/62/655062/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-testr","ref":"refs/changes/62/655062/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/62/655062/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/62/655062/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/62/655062/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-testr refs/changes/62/655062/4"}}},"commit":{"parents":[{"commit":"f16df2636cd1c0ee3e3c62ecf2698ab7de781b8d","subject":"Merge \"Fix warning message with double \"to\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/f16df2636cd1c0ee3e3c62ecf2698ab7de781b8d"}]}],"author":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-04-23 11:39:37.000000000","tz":480},"committer":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-06-05 02:17:28.000000000","tz":480},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\n1. Replace git.openstack.org URLs with opendev.org URLs\n2. Update some URLs to latest\n\nChange-Id: Ifcb01ece24fb28d4dc28daa0a638787280fe8eb5\nCloses-Bug: #1826529\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/6d4aecf256f9a02131582928480f9fc99c51b6a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/6d4aecf256f9a02131582928480f9fc99c51b6a1"}]},"branch":"refs/heads/master"},"96e504b0b0f0dc9a4576e27c918da6acc90bb3fb":{"kind":"REWORK","_number":5,"created":"2019-06-05 06:43:01.000000000","uploader":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"ref":"refs/changes/62/655062/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-testr","ref":"refs/changes/62/655062/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/62/655062/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/62/655062/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/62/655062/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-testr refs/changes/62/655062/5"}}},"commit":{"parents":[{"commit":"f16df2636cd1c0ee3e3c62ecf2698ab7de781b8d","subject":"Merge \"Fix warning message with double \"to\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/f16df2636cd1c0ee3e3c62ecf2698ab7de781b8d"}]}],"author":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-04-23 11:39:37.000000000","tz":480},"committer":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2019-06-05 06:42:47.000000000","tz":0},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\n1. Replace git.openstack.org URLs with opendev.org URLs\n2. Update some URLs to latest\n\nChange-Id: Ifcb01ece24fb28d4dc28daa0a638787280fe8eb5\nCloses-Bug: #1826529\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/96e504b0b0f0dc9a4576e27c918da6acc90bb3fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/96e504b0b0f0dc9a4576e27c918da6acc90bb3fb"}]},"branch":"refs/heads/master"},"2769e8dadfe2e76b185b9f07ac502b72964a3043":{"kind":"REWORK","_number":6,"created":"2019-06-05 06:43:34.000000000","uploader":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"ref":"refs/changes/62/655062/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-testr","ref":"refs/changes/62/655062/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/62/655062/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/62/655062/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/62/655062/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-testr refs/changes/62/655062/6"}}},"commit":{"parents":[{"commit":"f16df2636cd1c0ee3e3c62ecf2698ab7de781b8d","subject":"Merge \"Fix warning message with double \"to\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/f16df2636cd1c0ee3e3c62ecf2698ab7de781b8d"}]}],"author":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-04-23 11:39:37.000000000","tz":480},"committer":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2019-06-05 06:43:27.000000000","tz":0},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\n1. Replace git.openstack.org URLs with opendev.org URLs\n2. Update some URLs to latest\n\nChange-Id: Ifcb01ece24fb28d4dc28daa0a638787280fe8eb5\nCloses-Bug: #1826529\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/2769e8dadfe2e76b185b9f07ac502b72964a3043"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/2769e8dadfe2e76b185b9f07ac502b72964a3043"}]},"branch":"refs/heads/master"},"ad00e6d41b441e10f0b4ed81511e66660a9eeaa0":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2019-06-25 09:33:30.000000000","uploader":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"ref":"refs/changes/62/655062/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-testr","ref":"refs/changes/62/655062/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/62/655062/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/62/655062/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/62/655062/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-testr refs/changes/62/655062/7"}}},"commit":{"parents":[{"commit":"bced00648aed9e99abcbb8b94fcd49d55417cfef","subject":"Merge \"Deprecate ostestr command\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/bced00648aed9e99abcbb8b94fcd49d55417cfef"}]}],"author":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-04-23 11:39:37.000000000","tz":480},"committer":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2019-06-25 09:33:30.000000000","tz":0},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\n1. Replace git.openstack.org URLs with opendev.org URLs\n2. Update some URLs to latest\n\nChange-Id: Ifcb01ece24fb28d4dc28daa0a638787280fe8eb5\nCloses-Bug: #1826529\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/ad00e6d41b441e10f0b4ed81511e66660a9eeaa0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/ad00e6d41b441e10f0b4ed81511e66660a9eeaa0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
