)]}'
{"id":"openstack%2Fos-testr~661384","triplet_id":"openstack%2Fos-testr~master~I69b1f9f30ac51a5d34eec78ceda8fdd9405a6ff1","project":"openstack/os-testr","branch":"master","topic":"double-to","hashtags":[],"change_id":"I69b1f9f30ac51a5d34eec78ceda8fdd9405a6ff1","subject":"Fix warning message with double \"to\"","status":"MERGED","created":"2019-05-24 19:56:21.000000000","updated":"2019-06-02 15:04:46.000000000","submitted":"2019-06-02 15:04:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"661384-1559487886510-33fda56b","meta_rev_id":"013ee5e088184393c6d3471c37ab451152f55e01","_number":661384,"virtual_id_number":661384,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":2,"date":"2019-06-02 15:04:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-06-02 14:50:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-02 14:50:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-02 14:50:41.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2019-06-02 15:04:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f98bc07dbe1bf4b65d38dd36a8db418bde95102a","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-05-24 19:56:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b30f9f75325a435eaa4c0ae2ff9ca0e141fdc63c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-24 20:06:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/84/661384/1/check/openstack-tox-pep8/5517f22/ : SUCCESS in 4m 06s\n- openstack-tox-py27 http://logs.openstack.org/84/661384/1/check/openstack-tox-py27/ee6f1aa/ : FAILURE in 3m 43s\n- openstack-tox-py35 http://logs.openstack.org/84/661384/1/check/openstack-tox-py35/4f578ad/ : FAILURE in 7m 32s\n- openstack-tox-py36 http://logs.openstack.org/84/661384/1/check/openstack-tox-py36/8256a37/ : FAILURE in 6m 15s\n- openstack-tox-py37 http://logs.openstack.org/84/661384/1/check/openstack-tox-py37/31ff96e/ : FAILURE in 5m 01s\n- openstack-tox-docs http://logs.openstack.org/84/661384/1/check/openstack-tox-docs/c0fa3e7/html/ : SUCCESS in 7m 16s","accounts_in_message":[],"_revision_number":1},{"id":"3cfd36eafdf345b9c230436670675e0e9029b34c","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-05-24 20:10:16.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"3d4cf165b914ecd524cb1588e351d57a5e6f65f8","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-05-24 20:11:47.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"0bde36ef5be4a8bafcfa222f01e31f4d98a68d86","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-24 20:22:07.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/84/661384/3/check/openstack-tox-pep8/290e01a/ : SUCCESS in 5m 35s\n- openstack-tox-py27 http://logs.openstack.org/84/661384/3/check/openstack-tox-py27/81fe944/ : FAILURE in 4m 09s\n- openstack-tox-py35 http://logs.openstack.org/84/661384/3/check/openstack-tox-py35/3943b70/ : FAILURE in 4m 03s\n- openstack-tox-py36 http://logs.openstack.org/84/661384/3/check/openstack-tox-py36/35800f5/ : FAILURE in 6m 19s\n- openstack-tox-py37 http://logs.openstack.org/84/661384/3/check/openstack-tox-py37/eb91918/ : FAILURE in 5m 23s\n- openstack-tox-docs http://logs.openstack.org/84/661384/3/check/openstack-tox-docs/a7c2911/html/ : SUCCESS in 4m 11s","accounts_in_message":[],"_revision_number":3},{"id":"399486d3eb169360b4cdff008f5cb01bf526a24d","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-06-02 14:50:35.000000000","message":"Patch Set 3:\n\nthis is doc only change. let\u0027s get this in before we mark ostestr deprecated.","accounts_in_message":[],"_revision_number":3},{"id":"363273b9259f8fd39ac0fd423595576315164a9a","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-06-02 14:50:41.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"6ae733797599353d5d81792b0b04890f7501a8e5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-02 14:55:54.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/84/661384/3/check/openstack-tox-pep8/b82b501/ : SUCCESS in 3m 45s\n- openstack-tox-py27 http://logs.openstack.org/84/661384/3/check/openstack-tox-py27/5887c35/ : SUCCESS in 4m 07s\n- openstack-tox-py35 http://logs.openstack.org/84/661384/3/check/openstack-tox-py35/b75c25a/ : SUCCESS in 4m 12s\n- openstack-tox-py36 http://logs.openstack.org/84/661384/3/check/openstack-tox-py36/ba1a0e6/ : SUCCESS in 4m 25s\n- openstack-tox-py37 http://logs.openstack.org/84/661384/3/check/openstack-tox-py37/79c3abb/ : SUCCESS in 4m 10s\n- openstack-tox-docs http://logs.openstack.org/84/661384/3/check/openstack-tox-docs/263f6c0/html/ : SUCCESS in 4m 07s","accounts_in_message":[],"_revision_number":3},{"id":"dfc5a6ec4c90f0e14dcc78e3f73efa92a3399794","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-02 14:56:06.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"4fdd591d5b597bea65559d2cc5dc0af377214b88","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-02 15:04:46.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/84/661384/3/gate/openstack-tox-pep8/7cd8bae/ : SUCCESS in 5m 33s\n- openstack-tox-py27 http://logs.openstack.org/84/661384/3/gate/openstack-tox-py27/6a7ac3e/ : SUCCESS in 3m 44s\n- openstack-tox-py36 http://logs.openstack.org/84/661384/3/gate/openstack-tox-py36/3d32c37/ : SUCCESS in 5m 40s\n- openstack-tox-py37 http://logs.openstack.org/84/661384/3/gate/openstack-tox-py37/b85ef8a/ : SUCCESS in 5m 23s\n- openstack-tox-docs http://logs.openstack.org/84/661384/3/gate/openstack-tox-docs/19a585e/html/ : SUCCESS in 3m 37s","accounts_in_message":[],"_revision_number":3},{"id":"a63bc9f388ee79a41a35cc77e891a55477d87234","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-02 15:04:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"70f183c0529fb4abf6afdb7269c24505af873b1a","revisions":{"366b7ca390d2288af7d8462e61e1e51d849701d7":{"kind":"REWORK","_number":1,"created":"2019-05-24 19:56:21.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/84/661384/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-testr","ref":"refs/changes/84/661384/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/84/661384/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/84/661384/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/84/661384/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-testr refs/changes/84/661384/1"}}},"commit":{"parents":[{"commit":"5674bb0f7c1f0869a9996ff6f3e9b3cc1aff96e8","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/5674bb0f7c1f0869a9996ff6f3e9b3cc1aff96e8"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-05-24 19:54:41.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-05-24 19:54:41.000000000","tz":120},"subject":"Fix warning message with double \"to\"","message":"Fix warning message with double \"to\"\n\nFix warning message:\nOne \"to\" is enough, also improve sentence with adding \"file\".\n\nChange-Id: I69b1f9f30ac51a5d34eec78ceda8fdd9405a6ff1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/366b7ca390d2288af7d8462e61e1e51d849701d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/366b7ca390d2288af7d8462e61e1e51d849701d7"}]},"branch":"refs/heads/master"},"633210f8401e100771fc5e3b7fca8c3d406a8cec":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-05-24 20:10:16.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/84/661384/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-testr","ref":"refs/changes/84/661384/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/84/661384/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/84/661384/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/84/661384/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-testr refs/changes/84/661384/2"}}},"commit":{"parents":[{"commit":"595c06c5bc662ec945e728933df6ac5d6ab116c6","subject":"Default concurrency to 0 to avoid TypeError in stestr","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/595c06c5bc662ec945e728933df6ac5d6ab116c6"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-05-24 19:54:41.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","date":"2019-05-24 20:10:16.000000000","tz":0},"subject":"Fix warning message with double \"to\"","message":"Fix warning message with double \"to\"\n\nFix warning message:\nOne \"to\" is enough, also improve sentence with adding \"file\".\n\nChange-Id: I69b1f9f30ac51a5d34eec78ceda8fdd9405a6ff1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/633210f8401e100771fc5e3b7fca8c3d406a8cec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/633210f8401e100771fc5e3b7fca8c3d406a8cec"}]},"branch":"refs/heads/master"},"70f183c0529fb4abf6afdb7269c24505af873b1a":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-05-24 20:11:47.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/84/661384/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-testr","ref":"refs/changes/84/661384/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/84/661384/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/84/661384/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/84/661384/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-testr refs/changes/84/661384/3"}}},"commit":{"parents":[{"commit":"5674bb0f7c1f0869a9996ff6f3e9b3cc1aff96e8","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/5674bb0f7c1f0869a9996ff6f3e9b3cc1aff96e8"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-05-24 19:54:41.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","date":"2019-05-24 20:11:47.000000000","tz":0},"subject":"Fix warning message with double \"to\"","message":"Fix warning message with double \"to\"\n\nFix warning message:\nOne \"to\" is enough, also improve sentence with adding \"file\".\n\nChange-Id: I69b1f9f30ac51a5d34eec78ceda8fdd9405a6ff1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/70f183c0529fb4abf6afdb7269c24505af873b1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/70f183c0529fb4abf6afdb7269c24505af873b1a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
