)]}'
{"id":"openstack%2Fos-testr~700778","triplet_id":"openstack%2Fos-testr~master~I17e5aa6f45133d9d12aa38cec9cc42694913d4bd","project":"openstack/os-testr","branch":"master","topic":"bug/1857901","hashtags":[],"change_id":"I17e5aa6f45133d9d12aa38cec9cc42694913d4bd","subject":"subunit2html dealing with incorrect data format","status":"MERGED","created":"2019-12-30 15:36:22.000000000","updated":"2020-02-19 00:53:13.000000000","submitted":"2020-02-19 00:51:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"700778-1582073489362-0cc33de8","meta_rev_id":"4431b97c6df1759869685dfe57a9abab6f68bb6a","_number":700778,"virtual_id_number":700778,"owner":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":26106,"name":"Tom Stappaerts","email":"tom.stappaerts@nokia.com","username":"TomStappaerts"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":23003,"name":"Gal Amado","email":"gamado@redhat.com","username":"gamado"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-19 00:51:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"date":"2020-01-22 14:23:13.000000000","_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},{"value":0,"_account_id":18142,"name":"David Sariel","email":"dsariel@redhat.com","username":"dsariel"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},{"value":2,"date":"2020-01-31 00:45:50.000000000","_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":1,"date":"2020-02-10 13:24:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":1,"date":"2020-01-30 10:28:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26106,"name":"Tom Stappaerts","email":"tom.stappaerts@nokia.com","username":"TomStappaerts"},{"value":1,"date":"2020-01-22 14:27:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":1,"date":"2020-02-05 12:53:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23003,"name":"Gal Amado","email":"gamado@redhat.com","username":"gamado"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},{"value":2,"date":"2020-02-19 00:41:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},{"value":1,"date":"2020-02-06 12:25:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":18142,"name":"David Sariel","email":"dsariel@redhat.com","username":"dsariel"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"date":"2020-02-09 14:19:54.000000000","_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":26106,"name":"Tom Stappaerts","email":"tom.stappaerts@nokia.com","username":"TomStappaerts"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":23003,"name":"Gal Amado","email":"gamado@redhat.com","username":"gamado"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},{"value":1,"date":"2020-02-19 00:41:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},{"value":0,"_account_id":18142,"name":"David Sariel","email":"dsariel@redhat.com","username":"dsariel"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":18142,"name":"David Sariel","email":"dsariel@redhat.com","username":"dsariel"},{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"_account_id":23003,"name":"Gal Amado","email":"gamado@redhat.com","username":"gamado"},{"_account_id":26106,"name":"Tom Stappaerts","email":"tom.stappaerts@nokia.com","username":"TomStappaerts"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-07 11:45:31.000000000","updated_by":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"reviewer":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"state":"REVIEWER"},{"updated":"2020-01-21 07:05:44.000000000","updated_by":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"reviewer":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"state":"REVIEWER"},{"updated":"2020-01-22 14:27:54.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"REVIEWER"},{"updated":"2020-01-30 10:28:25.000000000","updated_by":{"_account_id":26106,"name":"Tom Stappaerts","email":"tom.stappaerts@nokia.com","username":"TomStappaerts"},"reviewer":{"_account_id":26106,"name":"Tom Stappaerts","email":"tom.stappaerts@nokia.com","username":"TomStappaerts"},"state":"REVIEWER"},{"updated":"2020-01-31 00:45:50.000000000","updated_by":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"reviewer":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"state":"REVIEWER"},{"updated":"2020-02-05 12:53:37.000000000","updated_by":{"_account_id":23003,"name":"Gal Amado","email":"gamado@redhat.com","username":"gamado"},"reviewer":{"_account_id":23003,"name":"Gal Amado","email":"gamado@redhat.com","username":"gamado"},"state":"REVIEWER"},{"updated":"2020-02-06 12:25:20.000000000","updated_by":{"_account_id":18142,"name":"David Sariel","email":"dsariel@redhat.com","username":"dsariel"},"reviewer":{"_account_id":18142,"name":"David Sariel","email":"dsariel@redhat.com","username":"dsariel"},"state":"REVIEWER"},{"updated":"2020-02-09 14:19:54.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2020-02-10 13:24:41.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"},{"updated":"2020-02-19 00:41:09.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2020-02-19 00:51:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b2b6a44481c6fd7be8e63e38cc017fd57ca02d1c","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2019-12-30 15:36:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fff8e65e33201ac01d94c99e118c0887922e5990","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-30 15:41:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4b6b7a8c4c245c18b98cf8b5ff41d22 : SUCCESS in 2m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8129b891e3d442f8abc58f14d8296f09 : SUCCESS in 3m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7d297b82c739457689193bbf162d3258 : SUCCESS in 3m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/132fe865bd22444cbf2749314e0f08c8 : SUCCESS in 3m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8abc62b7ba344278ff05f241eb5a8c7 : SUCCESS in 3m 12s","accounts_in_message":[],"_revision_number":1},{"id":"a4dcbd3d99bf355c3c20d1bed83cc8b4e85f74d2","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2019-12-31 08:05:12.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"e79c04c6f9000042c99b5af9fadd7330f9db0333","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-31 08:14:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4437cce820b243f3a6771ff203e9611f : SUCCESS in 3m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2a82d31e9ad344ccaf3060fde7ef461b : SUCCESS in 3m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ac3bc15d6fd0490f866e98a60779dd15 : SUCCESS in 4m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5a1cded30aad4670976f0eb3166562a4 : SUCCESS in 4m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/729fb8ab17664b2dbd29071eb5e1dbc2 : SUCCESS in 4m 11s","accounts_in_message":[],"_revision_number":2},{"id":"8dc050b1c447997d2cf4b95cee52faed0144bb0b","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2019-12-31 08:28:27.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"521c3031c4e04f5f3dfcd4865a6eb2098c9ebbb2","author":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"date":"2020-01-08 08:09:22.000000000","message":"Patch Set 2: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":2},{"id":"806d1d426941fdf6c5fafe1edfbd131f63503965","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2020-01-08 12:53:33.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"002ae49038247fad1a9a20ed3825e340147f047c","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2020-01-08 12:54:21.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"09f3cace7f6e1b59d30473c535e29319ec1e6502","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2020-01-08 12:54:59.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"a269a34a7591829348d522e6687456be14eafc06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-08 13:05:26.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac1dbe61c33143b993981a7eae6d1315 : SUCCESS in 3m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/dab02fac78d24bfb8d8008b380d928fc : SUCCESS in 4m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1c03c941ca0040a39668d11b39f4b053 : SUCCESS in 3m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bd96adb15d6841ad9c92c2e261a99334 : SUCCESS in 5m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17d9181921274deab2ccd6ad95014643 : SUCCESS in 3m 29s","accounts_in_message":[],"_revision_number":4},{"id":"1722227d732c7b2b3779196573d62b54bf6e7eb9","author":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"date":"2020-01-09 05:18:12.000000000","message":"Patch Set 4:\n\nThanks for updating! But could you add unit tests for this change?","accounts_in_message":[],"_revision_number":4},{"id":"2a444bc473377939fe05798068a2a47a282f1971","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2020-01-12 10:31:07.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"f3c6241b1acf03e58b75f1301242850c80a7acd3","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2020-01-12 10:33:17.000000000","message":"Patch Set 5:\n\n\u003e Thanks for updating! But could you add unit tests for this change?\n\nSorry for the late response.\nSure, already done.","accounts_in_message":[],"_revision_number":5},{"id":"eb069fbc607bbc0475844181698e47b0165f1f71","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-12 10:35:02.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d71512daecc41c3947db044d71ccf4e : FAILURE in 2m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7d26809a5b8940e2af21de2e41bc9f30 : SUCCESS in 3m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ead9269423184cdebe193a981c92be6b : SUCCESS in 2m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0ae2155ac3d74bf49f526421d8c03267 : SUCCESS in 2m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dde9b18cbca244d4a4b56495ae75ada0 : SUCCESS in 2m 58s","accounts_in_message":[],"_revision_number":5},{"id":"ceffd8de534fb2c80eb4035a287043ea7f313b57","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2020-01-12 11:28:55.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"f2421156db61bf6f12b13d62751ef4537c66192f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-12 11:35:48.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9765410a0384f318bc11a1d802fd295 : SUCCESS in 3m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b9f2665852b445e8bd09ffadab0a58f4 : SUCCESS in 3m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/98b8088d68e74ec8b47a66d117819176 : SUCCESS in 2m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/607122db31614b8aa30c52a9f915dc32 : SUCCESS in 2m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/668f3387d506452bad5fe9bed92ab827 : SUCCESS in 2m 48s","accounts_in_message":[],"_revision_number":6},{"id":"b6c165085bf304931f71ac341f748403a9e17c9f","author":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"date":"2020-01-21 00:20:43.000000000","message":"Patch Set 6:\n\n(1 comment)\n\nthanks for updating! but I\u0027m not sure about \"\\xe2\\x82\\xa5\".","accounts_in_message":[],"_revision_number":6},{"id":"f8e22ef1dd10da69ed4ef82c1d4c3081b5843a27","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2020-01-21 06:37:26.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"059a50e78563621159db279bbf82e8ae74af42b8","author":{"_account_id":4727,"name":"Itzik Brown","email":"itzikb@redhat.com","username":"itzikb1"},"date":"2020-01-21 07:05:44.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"fd659cefc24d37614d76a4d45bec32a2c87f3ad3","author":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"date":"2020-01-22 01:54:42.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"c723cefe50051652888fa3ee2f835aec6b7a3039","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2020-01-22 08:55:44.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"8f2a731db77e7e56dfe42bef183f851a4f47e08a","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2020-01-22 08:58:22.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"a72c25ce4506c1924e38a282fdd2607429ca44c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 09:02:51.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37b4c6a7f24d4913907e7c93ae5b9759 : SUCCESS in 3m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/67c4edc2b2eb4ea0a89599d2e480008b : SUCCESS in 3m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bed6545c8e644705a68a165496021227 : SUCCESS in 3m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/86ced97c09d042f2855f5325398d8471 : SUCCESS in 5m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67842a643fbe4d0387129a8e900c4b26 : SUCCESS in 4m 48s","accounts_in_message":[],"_revision_number":7},{"id":"7b3fe53537a35d4c41bb49491e8a7b888ea33fbc","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2020-01-22 14:20:19.000000000","message":"Patch Set 7:\n\nThere is a bug on subunit2html.\nI have upload a patch for this https://review.opendev.org/#/c/700778/","accounts_in_message":[],"_revision_number":7},{"id":"1d46d06fd9c31256cf4ba36502e7ba874d596c97","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2020-01-22 14:23:13.000000000","message":"Patch Set 7:\n\nIgnore the last comment","accounts_in_message":[],"_revision_number":7},{"id":"7b23cdddc9875034a11911785b9aeb398fc72420","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-01-22 14:27:54.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"732b8b0f9706d9d4fd2317ffd436400b112492ac","author":{"_account_id":26106,"name":"Tom Stappaerts","email":"tom.stappaerts@nokia.com","username":"TomStappaerts"},"date":"2020-01-30 10:28:25.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"e8b129882dc1ee429867816f74545e12f46bdedd","author":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"date":"2020-01-31 00:45:50.000000000","message":"Patch Set 7: Code-Review+2\n\nThanks for updating","accounts_in_message":[],"_revision_number":7},{"id":"69263b7926306d8ecf89ddfd947501460f7f2c72","author":{"_account_id":23003,"name":"Gal Amado","email":"gamado@redhat.com","username":"gamado"},"date":"2020-02-05 12:53:37.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"fc727a40b3a410c108c2b1b1261ee3586b56b400","author":{"_account_id":18142,"name":"David Sariel","email":"dsariel@redhat.com","username":"dsariel"},"date":"2020-02-06 12:25:20.000000000","message":"Patch Set 7: Code-Review+1\n\ncan this be moved to the workflow? Thanks","accounts_in_message":[],"_revision_number":7},{"id":"c1b3fef68cd79bf577bed6aff6d6aa4c4f909fe1","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2020-02-10 13:24:41.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"37851f31b4deab73faff6da36ffde909ae011066","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-02-19 00:41:09.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"3b2f6fb1bc476114fc9ec2a97b9a8f5637070861","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-19 00:41:20.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"f607db2948c71f1627651e466ddc8a6c0c203b88","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-19 00:51:29.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed76946dd2ca46a68ca1802d551e4450 : SUCCESS in 5m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4c8be7100f574857af6ad16677800f7c : SUCCESS in 6m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0df9baeccce245c3ae64bbe1c58a6227 : SUCCESS in 6m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0e3d7d1cde6d4113b6efcae89f56033a : SUCCESS in 6m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7bec167c2f9d433b93d54cba0a7b603b : SUCCESS in 4m 53s","accounts_in_message":[],"_revision_number":7},{"id":"c377be7de6f3aaddba251476ef11dfa10b812be9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-19 00:51:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"ed9e1259b5fdd0b7d5b983a543b0c1ea2dd84ec6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-19 00:53:13.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33b06e5f323e4b5e8ce9a63c943622c8 : SUCCESS in 1m 24s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"2c07d4e638122f845b3d0440f4864df8356d2a52","revisions":{"b8bcf243ee7c42e7aca1a6e00e61f9656ff56d69":{"kind":"REWORK","_number":1,"created":"2019-12-30 15:36:22.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/78/700778/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-testr","ref":"refs/changes/78/700778/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/78/700778/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/78/700778/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/78/700778/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-testr refs/changes/78/700778/1"}}},"commit":{"parents":[{"commit":"88594493596e6f3ac4c18dffc1cb4258453cf0ed","subject":"Add Python 3 Train unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/88594493596e6f3ac4c18dffc1cb4258453cf0ed"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2019-12-30 14:52:41.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2019-12-30 15:10:26.000000000","tz":120},"subject":"subunit2html dealing with incorrect data format","message":"subunit2html dealing with incorrect data format\n\nThe patch encodes a data string which for some reason\nwas not coded to utf-8.\n\nChange-Id: I17e5aa6f45133d9d12aa38cec9cc42694913d4bd\nCloses-Bug: #1857901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/b8bcf243ee7c42e7aca1a6e00e61f9656ff56d69"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/b8bcf243ee7c42e7aca1a6e00e61f9656ff56d69"}]},"branch":"refs/heads/master"},"486a3cfe36f05bed487ab0330f4961f35859a4dc":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-12-31 08:05:12.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/78/700778/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-testr","ref":"refs/changes/78/700778/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/78/700778/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/78/700778/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/78/700778/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-testr refs/changes/78/700778/2"}}},"commit":{"parents":[{"commit":"88594493596e6f3ac4c18dffc1cb4258453cf0ed","subject":"Add Python 3 Train unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/88594493596e6f3ac4c18dffc1cb4258453cf0ed"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2019-12-30 14:52:41.000000000","tz":120},"committer":{"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","date":"2019-12-31 08:04:37.000000000","tz":0},"subject":"subunit2html dealing with incorrect data format","message":"subunit2html dealing with incorrect data format\n\nThe patch encodes a data string which contains \nnon-ASCII characters to unicode type.\n\nIn python2 there are two text types str and unicode.\nWhen we add two different text types, for example:\n\u003e\u003e\u003e \"Hello \" + u\"World\"\nu\u0027Hello World\u0027\n\nThe string on the left is decoded by using the default \nsystem encoding into a Unicode string.\nIf it would contain non-ASCII characters(as in our case) \nthis normally blow up with an UnicodeDecodeError because \nthe default encoding is set to ASCII.\n\nChange-Id: I17e5aa6f45133d9d12aa38cec9cc42694913d4bd\nCloses-Bug: #1857901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/486a3cfe36f05bed487ab0330f4961f35859a4dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/486a3cfe36f05bed487ab0330f4961f35859a4dc"}]},"branch":"refs/heads/master"},"33881e816e4161e37fd4c2d80440576f6361d537":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-01-08 12:53:33.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/78/700778/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-testr","ref":"refs/changes/78/700778/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/78/700778/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/78/700778/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/78/700778/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-testr refs/changes/78/700778/3"}}},"commit":{"parents":[{"commit":"88594493596e6f3ac4c18dffc1cb4258453cf0ed","subject":"Add Python 3 Train unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/88594493596e6f3ac4c18dffc1cb4258453cf0ed"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2019-12-30 14:52:41.000000000","tz":120},"committer":{"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","date":"2020-01-08 12:50:49.000000000","tz":0},"subject":"subunit2html dealing with incorrect data format","message":"subunit2html dealing with incorrect data format\n\nThe patch encodes a data string which contains\nnon-ASCII characters to unicode type.\n\nIn python2 there are two text types str and unicode.\nWhen we add two different text types, for example:\n\u003e\u003e\u003e \"Hello \" + u\"World\"\nu\u0027Hello World\u0027\n\nThe string on the left is decoded by using the default\nsystem encoding into a Unicode string.\nIf it would contain non-ASCII characters(as in our case)\nthis normally blow up with an UnicodeDecodeError because \nthe default encoding is set to ASCII.\n\nChange-Id: I17e5aa6f45133d9d12aa38cec9cc42694913d4bd\nCloses-Bug: #1857901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/33881e816e4161e37fd4c2d80440576f6361d537"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/33881e816e4161e37fd4c2d80440576f6361d537"}]},"branch":"refs/heads/master"},"cf775454f7f762cd16ad587de111875a17bd9194":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2020-01-08 12:54:59.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/78/700778/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-testr","ref":"refs/changes/78/700778/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/78/700778/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/78/700778/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/78/700778/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-testr refs/changes/78/700778/4"}}},"commit":{"parents":[{"commit":"88594493596e6f3ac4c18dffc1cb4258453cf0ed","subject":"Add Python 3 Train unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/88594493596e6f3ac4c18dffc1cb4258453cf0ed"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2019-12-30 14:52:41.000000000","tz":120},"committer":{"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","date":"2020-01-08 12:54:08.000000000","tz":0},"subject":"subunit2html dealing with incorrect data format","message":"subunit2html dealing with incorrect data format\n\nThe patch encodes a data string which contains\nnon-ASCII characters to unicode type.\n\nIn python2 there are two text types str and unicode.\nWhen we add two different text types, for example:\n\u003e\u003e\u003e \"Hello \" + u\"World\"\nu\u0027Hello World\u0027\n\nThe string on the left is decoded by using the default\nsystem encoding into a Unicode string.\nIf it would contain non-ASCII characters(as in our case)\nthis normally blow up with an UnicodeDecodeError because\nthe default encoding is set to ASCII.\n\nChange-Id: I17e5aa6f45133d9d12aa38cec9cc42694913d4bd\nCloses-Bug: #1857901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/cf775454f7f762cd16ad587de111875a17bd9194"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/cf775454f7f762cd16ad587de111875a17bd9194"}]},"branch":"refs/heads/master"},"9c42506de513e65e951f5ce2377b995efffdd897":{"kind":"REWORK","_number":5,"created":"2020-01-12 10:31:07.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/78/700778/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-testr","ref":"refs/changes/78/700778/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/78/700778/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/78/700778/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/78/700778/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-testr refs/changes/78/700778/5"}}},"commit":{"parents":[{"commit":"88594493596e6f3ac4c18dffc1cb4258453cf0ed","subject":"Add Python 3 Train unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/88594493596e6f3ac4c18dffc1cb4258453cf0ed"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2019-12-30 14:52:41.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2020-01-12 10:31:04.000000000","tz":120},"subject":"subunit2html dealing with incorrect data format","message":"subunit2html dealing with incorrect data format\n\nThe patch encodes a data string which contains\nnon-ASCII characters to unicode type.\n\nIn python2 there are two text types str and unicode.\nWhen we add two different text types, for example:\n\u003e\u003e\u003e \"Hello \" + u\"World\"\nu\u0027Hello World\u0027\n\nThe string on the left is decoded by using the default\nsystem encoding into a Unicode string.\nIf it would contain non-ASCII characters(as in our case)\nthis normally blow up with an UnicodeDecodeError because\nthe default encoding is set to ASCII.\n\nChange-Id: I17e5aa6f45133d9d12aa38cec9cc42694913d4bd\nCloses-Bug: #1857901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/9c42506de513e65e951f5ce2377b995efffdd897"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/9c42506de513e65e951f5ce2377b995efffdd897"}]},"branch":"refs/heads/master"},"c507992c1733f094458b009d240fa7cf3febf40a":{"kind":"REWORK","_number":6,"created":"2020-01-12 11:28:55.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/78/700778/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-testr","ref":"refs/changes/78/700778/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/78/700778/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/78/700778/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/78/700778/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-testr refs/changes/78/700778/6"}}},"commit":{"parents":[{"commit":"88594493596e6f3ac4c18dffc1cb4258453cf0ed","subject":"Add Python 3 Train unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/88594493596e6f3ac4c18dffc1cb4258453cf0ed"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2019-12-30 14:52:41.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2020-01-12 11:28:42.000000000","tz":120},"subject":"subunit2html dealing with incorrect data format","message":"subunit2html dealing with incorrect data format\n\nThe patch encodes a data string which contains\nnon-ASCII characters to unicode type.\n\nIn python2 there are two text types str and unicode.\nWhen we add two different text types, for example:\n\u003e\u003e\u003e \"Hello \" + u\"World\"\nu\u0027Hello World\u0027\n\nThe string on the left is decoded by using the default\nsystem encoding into a Unicode string.\nIf it would contain non-ASCII characters(as in our case)\nthis normally blow up with an UnicodeDecodeError because\nthe default encoding is set to ASCII.\n\nChange-Id: I17e5aa6f45133d9d12aa38cec9cc42694913d4bd\nCloses-Bug: #1857901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/c507992c1733f094458b009d240fa7cf3febf40a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/c507992c1733f094458b009d240fa7cf3febf40a"}]},"branch":"refs/heads/master"},"2c07d4e638122f845b3d0440f4864df8356d2a52":{"kind":"REWORK","_number":7,"created":"2020-01-22 08:55:44.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/78/700778/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-testr","ref":"refs/changes/78/700778/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/78/700778/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/78/700778/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/78/700778/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-testr refs/changes/78/700778/7"}}},"commit":{"parents":[{"commit":"88594493596e6f3ac4c18dffc1cb4258453cf0ed","subject":"Add Python 3 Train unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/88594493596e6f3ac4c18dffc1cb4258453cf0ed"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2019-12-30 14:52:41.000000000","tz":120},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2020-01-22 08:55:32.000000000","tz":120},"subject":"subunit2html dealing with incorrect data format","message":"subunit2html dealing with incorrect data format\n\nThe patch encodes a data string which contains\nnon-ASCII characters to unicode type.\n\nIn python2 there are two text types str and unicode.\nWhen we add two different text types, for example:\n\u003e\u003e\u003e \"Hello \" + u\"World\"\nu\u0027Hello World\u0027\n\nThe string on the left is decoded by using the default\nsystem encoding into a Unicode string.\nIf it would contain non-ASCII characters(as in our case)\nthis normally blow up with an UnicodeDecodeError because\nthe default encoding is set to ASCII.\n\nChange-Id: I17e5aa6f45133d9d12aa38cec9cc42694913d4bd\nCloses-Bug: #1857901\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/2c07d4e638122f845b3d0440f4864df8356d2a52"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/2c07d4e638122f845b3d0440f4864df8356d2a52"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
