)]}'
{"id":"openstack%2Fos-testr~730189","triplet_id":"openstack%2Fos-testr~master~I9ec9dd3b8b1936e6b5768b29a87ddec348bb842c","project":"openstack/os-testr","branch":"master","hashtags":[],"change_id":"I9ec9dd3b8b1936e6b5768b29a87ddec348bb842c","subject":"Fix hacking min version to 3.0.1","status":"ABANDONED","created":"2020-05-22 03:54:31.000000000","updated":"2020-07-21 02:54:36.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e8245b5d7e286079ef5212cca8d5f98534f7a45c","_number":730189,"virtual_id_number":730189,"owner":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-05-22 04:02:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":-2,"date":"2020-07-20 16:20:14.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-22 04:02:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-20 16:20:14.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"}],"messages":[{"id":"3dd6b5d45b41a0ecfe3df3010ffb7bf02a7415c1","author":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"date":"2020-05-22 03:54:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"26b98913f1e2eabdc33b35d301f41dd2a563d7fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-22 04:02:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/a413ed41f3ac4781b8718b0de5492ac2 : SUCCESS in 2m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/955061a7ec844fd3b9f44419b225b929 : SUCCESS in 2m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/45ba1b61157244a79c76623af12d4028 : SUCCESS in 3m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/67f8c69180ab47c28c514d169ec7081a : SUCCESS in 3m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/225fdd7bfab24ec3a40b6f09042e8780 : SUCCESS in 4m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4bcea7d4a2f4a78986be981ac557461 : SUCCESS in 4m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/523744ac837348d89934acd26b745a7c : SUCCESS in 3m 39s","accounts_in_message":[],"_revision_number":1},{"id":"00a01f158d044a49b6dda098806680c9dad16f74","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-07-20 16:20:14.000000000","message":"Patch Set 1: Code-Review-2\n\nalready done in https://review.opendev.org/#/c/727618/","accounts_in_message":[],"_revision_number":1},{"id":"e8245b5d7e286079ef5212cca8d5f98534f7a45c","author":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"date":"2020-07-21 02:54:36.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b243822fc6adc39f5eb389cfe76d494c47da1dbd","revisions":{"b243822fc6adc39f5eb389cfe76d494c47da1dbd":{"kind":"REWORK","_number":1,"created":"2020-05-22 03:54:31.000000000","uploader":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"ref":"refs/changes/89/730189/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-testr","ref":"refs/changes/89/730189/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/89/730189/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/89/730189/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/89/730189/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-testr refs/changes/89/730189/1"}}},"commit":{"parents":[{"commit":"ce1af65b21bd854420e4444bc354c6cd245e3f1d","subject":"[ussuri][goal] Update contributor documentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/ce1af65b21bd854420e4444bc354c6cd245e3f1d"}]}],"author":{"name":"zhangboye","email":"zhangboye@inspur.com","date":"2020-05-22 03:54:09.000000000","tz":480},"committer":{"name":"zhangboye","email":"zhangboye@inspur.com","date":"2020-05-22 03:54:24.000000000","tz":480},"subject":"Fix hacking min version to 3.0.1","message":"Fix hacking min version to 3.0.1\n\nChange-Id: I9ec9dd3b8b1936e6b5768b29a87ddec348bb842c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/b243822fc6adc39f5eb389cfe76d494c47da1dbd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/b243822fc6adc39f5eb389cfe76d494c47da1dbd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
