)]}'
{"id":"openstack%2Fos-testr~740367","triplet_id":"openstack%2Fos-testr~master~Ie17d56ca4949b9555712eb10950dfb6c545d1459","project":"openstack/os-testr","branch":"master","topic":"migrate-to-focal","hashtags":[],"change_id":"Ie17d56ca4949b9555712eb10950dfb6c545d1459","subject":"DNM: testing focal","status":"ABANDONED","created":"2020-07-09 20:34:35.000000000","updated":"2020-08-01 19:37:48.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d999bacbb53f3de4db740af6cf4df45db96d245d","_number":740367,"virtual_id_number":740367,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2020-07-31 23:35:59.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-31 23:35:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a905a683895f2ab511a0e78ab58e557cc8251438","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-07-09 20:34:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"be01efa027c333fe50605e16a2ee08efda92cdac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-09 21:10:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59f84442749940368f5e920d585ea9be : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9a19667da0ef4063b3d13b3726731e90 : FAILURE in 4m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c7cbc80894b94f40abb3c3c60b272d03 : RETRY_LIMIT in 2m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/970aabf7af09425b82f6672502fc60b8 : FAILURE in 5m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d97e47adb0fc4a99a65df75c65a1c649 : SUCCESS in 5m 07s","accounts_in_message":[],"_revision_number":1},{"id":"3d3a9ea526eb1ba7e71e52c46a7d33267fc2b2c3","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-07-26 17:33:43.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b73b9900fe3963c19880ca6ba6a00fa7367b7a25","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-26 18:01:51.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc8d6adcc23f4f8994b05d867f453c72 : SUCCESS in 2m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e9df98a570684b368659cab8dc2cae70 : FAILURE in 3m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3f15ec35722444c088562014533cd5fd : RETRY_LIMIT in 2m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f66361929a04479c9980b8b03cda753d : FAILURE in 5m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35884d05327a46fdba48a56772ebf310 : SUCCESS in 3m 34s","accounts_in_message":[],"_revision_number":1},{"id":"6fd8a5848fb26000c41a038b32cbc919fa2f35eb","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-07-31 19:09:08.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"2a244c6efc964e79e41c36d589053c9eaa28ec29","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-31 19:36:24.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e3ebda210394246b792e81590a4df3d : FAILURE in 3m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ce2e7f50f5f04599b81ae92a1086a3da : SUCCESS in 3m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0480b60bb013424985453be6ba5e8372 : SUCCESS in 3m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/528c0f95f4c9431f8d72e3b776b13916 : SUCCESS in 3m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5f8d2dd1ce94aa399bc416e04f2de88 : SUCCESS in 3m 39s\n\nWarning:\n  Comments left for invalid file os_testr/tests/testlist_builder.py","accounts_in_message":[],"_revision_number":2},{"id":"ef7a066690938f2308ae177f55dec914df3d1b4f","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-07-31 23:22:36.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"0f686687e13263ed559c78a02be38542d9215c30","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-31 23:35:59.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ef135ba3f124317adca71d9e221bb92 : SUCCESS in 3m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d6b522ff4acd48639e08c1b2f610cd76 : SUCCESS in 4m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/13899d80170b45a2bf74cf7782d5c472 : SUCCESS in 3m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/903bc7a6db4846cea78d78cd86a343bb : SUCCESS in 4m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8bd55579dcbf417d986c342de1c57fe3 : SUCCESS in 4m 10s","accounts_in_message":[],"_revision_number":3},{"id":"3a17739f7e0a489ef3c24c6e2817304b6eba5e95","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-08-01 19:37:48.000000000","message":"Abandoned\n\nall good here no updates needed for focal","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"aa31820aeb591d5dd0a00eb9ce3320c1619a1436","revisions":{"9840683cdf944e31b0bd119a2a943cfca3a69f60":{"kind":"REWORK","_number":1,"created":"2020-07-09 20:34:35.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/67/740367/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-testr","ref":"refs/changes/67/740367/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/67/740367/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/67/740367/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/67/740367/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-testr refs/changes/67/740367/1"}}},"commit":{"parents":[{"commit":"ce1af65b21bd854420e4444bc354c6cd245e3f1d","subject":"[ussuri][goal] Update contributor documentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/ce1af65b21bd854420e4444bc354c6cd245e3f1d"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-07-09 20:34:16.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-07-09 20:34:16.000000000","tz":-300},"subject":"DNM: testing focal","message":"DNM: testing focal\n\nDepends-On: https://review.opendev.org/#/c/734700\n\nChange-Id: Ie17d56ca4949b9555712eb10950dfb6c545d1459\nStory: #2007865\nTask: #40212\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/9840683cdf944e31b0bd119a2a943cfca3a69f60"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/9840683cdf944e31b0bd119a2a943cfca3a69f60"}]},"branch":"refs/heads/master"},"8a098d5d9df357a1dfd1c8e68dbdac3404fea04c":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-07-31 19:09:08.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/67/740367/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-testr","ref":"refs/changes/67/740367/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/67/740367/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/67/740367/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/67/740367/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-testr refs/changes/67/740367/2"}}},"commit":{"parents":[{"commit":"e33e5e2ad568c016aa92ea5741e591c6ddd8ed35","subject":"Use unittest.mock instead of third party mock","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/e33e5e2ad568c016aa92ea5741e591c6ddd8ed35"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-07-09 20:34:16.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-07-31 19:09:08.000000000","tz":0},"subject":"DNM: testing focal","message":"DNM: testing focal\n\nDepends-On: https://review.opendev.org/#/c/734700\n\nChange-Id: Ie17d56ca4949b9555712eb10950dfb6c545d1459\nStory: #2007865\nTask: #40212\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/8a098d5d9df357a1dfd1c8e68dbdac3404fea04c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/8a098d5d9df357a1dfd1c8e68dbdac3404fea04c"}]},"branch":"refs/heads/master"},"aa31820aeb591d5dd0a00eb9ce3320c1619a1436":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-07-31 23:22:36.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/67/740367/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-testr","ref":"refs/changes/67/740367/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/67/740367/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/67/740367/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/67/740367/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-testr refs/changes/67/740367/3"}}},"commit":{"parents":[{"commit":"eabcb9475635bad8d4ab908509a4c080e06f58e5","subject":"Use unittest.mock instead of third party mock","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/eabcb9475635bad8d4ab908509a4c080e06f58e5"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-07-09 20:34:16.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-07-31 23:22:36.000000000","tz":0},"subject":"DNM: testing focal","message":"DNM: testing focal\n\nDepends-On: https://review.opendev.org/#/c/734700\n\nChange-Id: Ie17d56ca4949b9555712eb10950dfb6c545d1459\nStory: #2007865\nTask: #40212\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/aa31820aeb591d5dd0a00eb9ce3320c1619a1436"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/aa31820aeb591d5dd0a00eb9ce3320c1619a1436"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
