)]}'
{"id":"openstack%2Fos-testr~793541","triplet_id":"openstack%2Fos-testr~master~Ib3c11dee2182d4cd9da816cee85bf9422e726722","project":"openstack/os-testr","branch":"master","hashtags":[],"change_id":"Ib3c11dee2182d4cd9da816cee85bf9422e726722","subject":"setup.cfg: Replace dashes with underscores","status":"MERGED","created":"2021-05-28 06:11:02.000000000","updated":"2021-06-02 09:35:04.000000000","submitted":"2021-06-02 09:33:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"793541","meta_rev_id":"4da9f595b7d0e6cdfc873725f883f24f2f5a7697","_number":793541,"virtual_id_number":793541,"owner":{"_account_id":33379,"name":"Anand Bhat","email":"anandgvbhat@gmail.com","username":"anandgvb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-02 09:33:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"all":[{"value":2,"date":"2021-06-02 09:19:11.000000000","_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"all":[{"value":1,"date":"2021-06-02 09:19:11.000000000","_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-28 06:27:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-28 06:58:52.000000000","updated_by":{"_account_id":33379,"name":"Anand Bhat","email":"anandgvbhat@gmail.com","username":"anandgvb"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"}],"messages":[{"id":"c5deb803a5799490dfbc869fe3766641ab302561","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33379,"name":"Anand Bhat","email":"anandgvbhat@gmail.com","username":"anandgvb"},"date":"2021-05-28 06:11:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b3563c67bbe3c6de3ffec822abe9ee5dc98ee5d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-28 06:27:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d134c2ddf9ea4a3b9ad6130db0ddf3e7 : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d57309077e134b7293c550e747221c39 : SUCCESS in 5m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/721b5a1a6d7348ffa2b41357441ddbd9 : FAILURE in 5m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/539ed9d4d94e4cdcb619874c59fe6e0b : FAILURE in 6m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/957ec9109d4b4c97a15f2848c0f55bf7 : FAILURE in 5m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7a0e13177584746a497be1f598a058c : SUCCESS in 9m 15s","accounts_in_message":[],"_revision_number":1},{"id":"b5cca02604597aa049f0477603e46182649dc04c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33379,"name":"Anand Bhat","email":"anandgvbhat@gmail.com","username":"anandgvb"},"date":"2021-05-29 19:38:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"270d74f028956e54c7f44dfde600ae8a790faa13","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-29 19:46:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/f6a9cd1b9d714c30833b00c9657b4e1b : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/919b866f06184d7e9cc8015ca404538d : SUCCESS in 3m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/19963e6ff87a46209a125f76ee15bdb0 : SUCCESS in 3m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6eb95054199746d8a1a0d166d523ab2d : SUCCESS in 3m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4dffdb5851004315ac3867654fe8ab0b : SUCCESS in 4m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87a1b34e85734c509d2bd67533fba178 : SUCCESS in 6m 39s","accounts_in_message":[],"_revision_number":2},{"id":"9528205c644c4bd0d6457b47481bd3fe380c9da5","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2021-06-02 09:19:11.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nlgtm, the user warning could have been seen in pep8 job, f.e.: https://storage.gra.cloud.ovh.net/v1/AUTH_dcaab5e32b234d56b626f72581e3644c/zuul_opendev_logs_0e2/793852/1/gate/openstack-tox-pep8/0e2a5c1/job-output.txt","accounts_in_message":[],"_revision_number":2},{"id":"315ce9b0f8edced011ae852f61b6fd8fd1e23ede","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-02 09:19:25.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"df7fc028e2931972dd48e6d4668e13a8965709af","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-02 09:33:46.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/5af6dc4bd4a344c585f7a22bc30bd6ef : SUCCESS in 2m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2406c68b93964bd88ff16d48a168e2df : SUCCESS in 4m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/db04e4420f274b46afeaa2e2ec80893c : SUCCESS in 5m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e8cc22e6eff64698b5cf7d01cf342f5d : SUCCESS in 4m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0513286de7c24399a7c9b2c8859be2b1 : SUCCESS in 7m 35s","accounts_in_message":[],"_revision_number":2},{"id":"27c8d2d98c95adfc7f5aa1e12f683a2fa9242084","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-02 09:33:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"4da9f595b7d0e6cdfc873725f883f24f2f5a7697","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-02 09:35:04.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/662f13d1b5e54705af5e2dce9190748e : SUCCESS in 58s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e620c09f11cf45489c30bd47fdc910058122204c","revisions":{"a4d4ce5afe068c329c41e099d52be345c0531bda":{"kind":"REWORK","_number":1,"created":"2021-05-28 06:11:02.000000000","uploader":{"_account_id":33379,"name":"Anand Bhat","email":"anandgvbhat@gmail.com","username":"anandgvb"},"ref":"refs/changes/41/793541/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-testr","ref":"refs/changes/41/793541/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/41/793541/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/41/793541/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/41/793541/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-testr refs/changes/41/793541/1"}}},"commit":{"parents":[{"commit":"804fa2cf4e79ce703e49ec7ee46b503ff1a8d1f9","subject":"Merge \" Switch testing to Xena testing runtime\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/804fa2cf4e79ce703e49ec7ee46b503ff1a8d1f9"}]}],"author":{"name":"Anand Bhat","email":"anandgvbhat@gmail.com","date":"2021-05-28 06:10:06.000000000","tz":330},"committer":{"name":"Anand Bhat","email":"anandgvbhat@gmail.com","date":"2021-05-28 06:10:06.000000000","tz":330},"subject":"setup.cfg: Replace dashes with underscores","message":"setup.cfg: Replace dashes with underscores\n\nSetuptools v54.1.0 introduces a warning that the use of dash-separated\noptions in \u0027setup.cfg\u0027 will not be supported in a future version [1].\nGet ahead of the issue by replacing the dashes with underscores. Without\nthis, we see \u0027UserWarning\u0027 messages like the following on new enough\nversions of setuptools:\n\n  UserWarning: Usage of dash-separated \u0027description-file\u0027 will not be\n  supported in future versions. Please use the underscore name\n  \u0027description_file\u0027 instead\n\n[1] https://github.com/pypa/setuptools/commit/a2e9ae4cb\n\nChange-Id: Ib3c11dee2182d4cd9da816cee85bf9422e726722\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/a4d4ce5afe068c329c41e099d52be345c0531bda"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/a4d4ce5afe068c329c41e099d52be345c0531bda"}]},"branch":"refs/heads/master"},"e620c09f11cf45489c30bd47fdc910058122204c":{"kind":"REWORK","_number":2,"created":"2021-05-29 19:38:04.000000000","uploader":{"_account_id":33379,"name":"Anand Bhat","email":"anandgvbhat@gmail.com","username":"anandgvb"},"ref":"refs/changes/41/793541/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/os-testr","ref":"refs/changes/41/793541/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/41/793541/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/41/793541/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/os-testr refs/changes/41/793541/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/os-testr refs/changes/41/793541/2"}}},"commit":{"parents":[{"commit":"804fa2cf4e79ce703e49ec7ee46b503ff1a8d1f9","subject":"Merge \" Switch testing to Xena testing runtime\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/804fa2cf4e79ce703e49ec7ee46b503ff1a8d1f9"}]}],"author":{"name":"Anand Bhat","email":"anandgvbhat@gmail.com","date":"2021-05-29 19:35:38.000000000","tz":330},"committer":{"name":"Anand Bhat","email":"anandgvbhat@gmail.com","date":"2021-05-29 19:37:07.000000000","tz":330},"subject":"setup.cfg: Replace dashes with underscores","message":"setup.cfg: Replace dashes with underscores\n\nSetuptools v54.1.0 introduces a warning that the use of dash-separated\noptions in \u0027setup.cfg\u0027 will not be supported in a future version [1].\nGet ahead of the issue by replacing the dashes with underscores. Without\nthis, we see \u0027UserWarning\u0027 messages like the following on new enough\nversions of setuptools:\n\n  UserWarning: Usage of dash-separated \u0027description-file\u0027 will not be\n  supported in future versions. Please use the underscore name\n  \u0027description_file\u0027 instead\n\n[1] https://github.com/pypa/setuptools/commit/a2e9ae4cb\n\nChange-Id: Ib3c11dee2182d4cd9da816cee85bf9422e726722\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/e620c09f11cf45489c30bd47fdc910058122204c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/os-testr/commit/e620c09f11cf45489c30bd47fdc910058122204c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
